)]}'
{"id":"openstack%2Fmistral~695332","triplet_id":"openstack%2Fmistral~stable%2Fstein~I793fbfec03032d9ff7137c11becb6d1c18ec54bc","project":"openstack/mistral","branch":"stable/stein","topic":"bug/1843175-stable/stein","hashtags":[],"change_id":"I793fbfec03032d9ff7137c11becb6d1c18ec54bc","subject":"Fix error validate token when run cron trigger","status":"MERGED","created":"2019-11-21 01:06:51.000000000","updated":"2019-11-21 07:26:26.000000000","submitted":"2019-11-21 07:24:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695332-1574321095981-29596582","meta_rev_id":"b1848321895b2e6975400c3bdedf3493fdcd1d6e","_number":695332,"virtual_id_number":695332,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 07:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":2,"date":"2019-11-21 06:26:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-11-21 01:06:51.000000000","_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},{"value":1,"date":"2019-11-21 06:26:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-21 01:06:51.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"REVIEWER"},{"updated":"2019-11-21 06:26:39.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-21 07:24:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f9b4d187266e631b9cc3b77a531aa42b54369f5","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-11-21 01:06:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2b1d00d415ced38aee4e227c7423104a8cb1b822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 03:21:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53329772a73a4d8995f2823d8184fedb : SUCCESS in 7m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4aebdfcb491649618e96344b3bf8682a : SUCCESS in 7m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1b8f479ca2994eaa84a4018e33e05f34 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4e0d3e90a8e46cfba23800e17dd60c2 : SUCCESS in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52f4f8462f8443fe90ef6dbfa8638e44 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0ad1f4aba147d3829cc2c671d4109b : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7cae222221745d186d35000bf5df74c : SUCCESS in 8m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/90f648f8092f411589c3fc807f3f50e4 : SUCCESS in 41m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cd8d3ede168945559d1cc0067bd02a45 : SUCCESS in 43m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e89a9996b8fd492697431736227e05ec : SUCCESS in 42m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bd5dd6c9944e425e97b9206121eac02e : SUCCESS in 26m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/253dc1c3c6114c3fa6032c374e615d0c : SUCCESS in 25m 49s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/78f1bd49434747719de81bc8925db715 : SUCCESS in 1h 39m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/163c501f813e41d38d45a874817ad647 : SUCCESS in 2h 11m 48s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/cee7ee2b5950423da2802d0a7b98b657 : FAILURE in 1h 16m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e04b9a6d308483f8be6fa51b36e9315 : SUCCESS in 9m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ae76b8cc10e42d694f1c6f532d68cecbe2a48e06","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-21 06:26:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5ff259c9f96152721cdffa5c9d52335e5e31f02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 06:26:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c467a6cdf49df1bf6cc2422c65fc2d25514af00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 07:24:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd55946beae540d4add0c724f2cf011e : SUCCESS in 4m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/872632fb31c54e0c8915eb47aad38db2 : SUCCESS in 8m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a8f363443cab46b8a6dd8d443ce108aa : SUCCESS in 9m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7288079fdeb548cd9ef88c2baedf0400 : SUCCESS in 10m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1cdf7f13d51400fa99bae795f83debe : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/633123d597b045bfab28c2361685ee1a : SUCCESS in 5m 20s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/94b11d61a99c40c4aaf51ded69ff5903 : SUCCESS in 49m 09s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/815e4a9f3e0741da9ae5434772301b0f : SUCCESS in 50m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7b6ae6f60aa64521b11bce61d1f99ec3 : SUCCESS in 27m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d6483501bdcb46a09f85e01c0216e9fb : SUCCESS in 26m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8dc399f3b24e4f3181118e03855ad447 : SUCCESS in 44m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf7b7d357b5348f793979a2a53e4163a : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f279c899353169abea4c779fd0d97df7d50f5cbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 07:24:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1758428aaad55c5e00ef5fa19cc54f4c8d3e6186","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 07:26:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1992dfe9587421292c5bad6cf86cc58 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9da790c7fbcb6d27748787da44a62290e5a3a528","revisions":{"9da790c7fbcb6d27748787da44a62290e5a3a528":{"kind":"REWORK","_number":1,"created":"2019-11-21 01:06:51.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/32/695332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/32/695332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/695332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/695332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/32/695332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/32/695332/1"}}},"commit":{"parents":[{"commit":"6399f9b7e0133202f2d63896b44628668cdc212f","subject":"Mask sensitive data when logging action results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6399f9b7e0133202f2d63896b44628668cdc212f"}]}],"author":{"name":"Bo Tran","email":"ministry.96.nd@gmail.com","date":"2019-09-08 13:32:25.000000000","tz":420},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-11-21 01:06:51.000000000","tz":0},"subject":"Fix error validate token when run cron trigger","message":"Fix error validate token when run cron trigger\n\nA trust client can\u0027t do validate token when run cron trigger\nThis patch will fix that.\n\nChange-Id: I793fbfec03032d9ff7137c11becb6d1c18ec54bc\nCloses-Bug: #1843175\n(cherry picked from commit 51b7dd0c6d5c4d0de354719765968ca883b9f8fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9da790c7fbcb6d27748787da44a62290e5a3a528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9da790c7fbcb6d27748787da44a62290e5a3a528"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
