)]}'
{"id":"openstack%2Fmistral~695333","triplet_id":"openstack%2Fmistral~stable%2Fstein~I2571246947ab5ae06731fc7d35de7a5954325a82","project":"openstack/mistral","branch":"stable/stein","topic":"releasenote-bug-1843175-stable/stein","hashtags":[],"change_id":"I2571246947ab5ae06731fc7d35de7a5954325a82","subject":"Add release note for fix error of cron trigger run","status":"MERGED","created":"2019-11-21 01:07:39.000000000","updated":"2019-11-21 07:28:05.000000000","submitted":"2019-11-21 07:24:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695333-1574321096922-64a7fc44","meta_rev_id":"3ed41fdf3add8f3ccc5ee75160bdcbe5c1fcaadd","_number":695333,"virtual_id_number":695333,"owner":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-21 07:24:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},{"value":2,"date":"2019-11-21 06:27:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2019-11-21 01:07:39.000000000","_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},{"value":1,"date":"2019-11-21 06:27:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-21 01:07:39.000000000","updated_by":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"reviewer":{"_account_id":31093,"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","username":"ohuzei"},"state":"REVIEWER"},{"updated":"2019-11-21 06:27:05.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-21 07:24:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"760850c9a7480c95226d82b7404d128924357e71","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2019-11-21 01:07:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5506c9bf422511bfba7f343a86f7fd19eff76c03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 01:36:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12a362e48e7b4db1a6b84f0c3907b3b9 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/918fc0c930e24ac4a5d04c2a8fab983e : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/115bd1da3d9f4f17b0e48449ba4856c3 : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4324a507b035463f87f437429780a93a : SUCCESS in 8m 53s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/943abaad7e7842f192257ec1f7621086 : SUCCESS in 28m 20s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/33a2132d33a94d19816f3fff22d34a58 : SUCCESS in 26m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e21dc7dfd88055183ce8489c5a3622f873706ff9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-21 06:27:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e733c121d6cd25670ab07880dfaec0a997e70524","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 06:27:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e284d52b130dd6d9ad9bf4958b3769382a6d7d00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 07:24:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a5fe7963d534056abed961a70337ec1 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78660cb7ede745c4a90c0110a73e979b : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b2728e8d24b4a098d42426636d28cbf : SUCCESS in 5m 32s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d97e45d4bffd46ea856205cae930668e : SUCCESS in 28m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d71faf0bad64400ea4c363a77ea4d271 : SUCCESS in 27m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9f25fda2704f4c13caa35b55d9da206208bd59d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 07:24:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ab90b49653771257e6ee6fc1ff81255a4b2617bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 07:28:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb441dd9ee9e467eb3816581e92065b8 : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffa57ca8787245568c1759a5692aeea8 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66da715c84fadba2120a59f02b68ba92ae6807b8","revisions":{"66da715c84fadba2120a59f02b68ba92ae6807b8":{"kind":"REWORK","_number":1,"created":"2019-11-21 01:07:39.000000000","uploader":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"ref":"refs/changes/33/695333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/33/695333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/695333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/695333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/33/695333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/33/695333/1"}}},"commit":{"parents":[{"commit":"6399f9b7e0133202f2d63896b44628668cdc212f","subject":"Mask sensitive data when logging action results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6399f9b7e0133202f2d63896b44628668cdc212f"}]}],"author":{"name":"Oleh Huzei","email":"oleh.huzei@ormuco.com","date":"2019-10-16 15:32:28.000000000","tz":-240},"committer":{"name":"Vlad Gusev","email":"vlad.esten@gmail.com","date":"2019-11-21 01:07:39.000000000","tz":0},"subject":"Add release note for fix error of cron trigger run","message":"Add release note for fix error of cron trigger run\n\nIt relates to the change I793fbfec03032d9ff7137c11becb6d1c18ec54bc,\ncommit 51b7dd0c6d.\n\nChange-Id: I2571246947ab5ae06731fc7d35de7a5954325a82\n(cherry picked from commit 948ceb150b66b86ba56b11821e1bb9cb8a7209cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66da715c84fadba2120a59f02b68ba92ae6807b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/66da715c84fadba2120a59f02b68ba92ae6807b8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
