)]}'
{"id":"openstack%2Fmistral~695378","triplet_id":"openstack%2Fmistral~master~Ie3381c7226d50591192ae61697bdb0d134ddf4f1","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"Ie3381c7226d50591192ae61697bdb0d134ddf4f1","subject":"Remove the TripleO standalone job","status":"MERGED","created":"2019-11-21 09:29:35.000000000","updated":"2019-12-16 14:04:24.000000000","submitted":"2019-12-16 14:00:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695378-1576504856745-fead5b33","meta_rev_id":"b3f899e5e8c5c92f6c696528c0693c3c11da4b7d","_number":695378,"virtual_id_number":695378,"owner":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-16 09:04:53.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-16 14:00:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2019-11-22 21:08:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2019-11-22 05:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2019-11-22 05:22:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 05:22:30.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-11-22 21:08:20.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-12-16 14:00:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96cc24ac115dae96d0bc0bac3d0d35c577d59cc5","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-11-21 09:29:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dfcf68343e8d46c05e02a1729fc51ac8fc88298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 11:47:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6744187ab464129a2f8dda70fec6ac3 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/befeb9d01e9d4db19bcfe20e24c52fa6 : SUCCESS in 8m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46b4e58fee8b4932a28e43e3cc00aa14 : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288c88b0aed24b758ce1f7bd8a934a17 : SUCCESS in 6m 58s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/683622a7311a49c992c9c4f532ef5149 : SUCCESS in 1h 01m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/64d31a3ee34c4d8aa5dc73d0cb31278e : SUCCESS in 43m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f764136bd3474018bcea9c954ea1270b : SUCCESS in 45m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2ca42a2cf76140fbb68a909a294e3f15 : SUCCESS in 29m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7c8414942fff4825b34e0183a5d10d39 : SUCCESS in 30m 03s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f6817f9fa1a140bfb0979585ef4fbe71 : SUCCESS in 2h 16m 12s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6ac59a18280d4641980a312e23eacdfb : SUCCESS in 57m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7925b2d55ce46739e1a985ebba75312 : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"32d3a956c73a5afed24daf1f275c9a40791a999a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-22 05:22:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"57192e9ad5e0ffcc05ff6b2181ba938961b84226","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 05:22:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ecd7b69d7de196ee28e9c061b6a9f3ab2909d6c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 07:13:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb717bfcc46a4cd6a587282b43fc89e3 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/651073ddb5d643788b61b16b290354ee : SUCCESS in 7m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c46d01476918459e884351a2b4340513 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e0a73c3c7249b4b2f669d4ca6b4e88 : SUCCESS in 6m 49s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0ff502946a7f4f82aa8e87b7ae9e8ba3 : SUCCESS in 38m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1eb77d12f60c431290ebad4e24c8076c : SUCCESS in 41m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/72d1fa7dd57046fd85cc0da1b0148a90 : SUCCESS in 28m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c7ad692b42e44ba6a45068351ba697a0 : FAILURE in 27m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/33351833aee14bc3aab0c1239f9606f4 : SUCCESS in 36m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/230e396bf2d942b58c9c52d8ba0ae04f : SUCCESS in 11m 07s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/bf9502d6fc7142c1a8b445dac1efb04f : SUCCESS in 1h 49m 08s","accounts_in_message":[],"_revision_number":1},{"id":"247be764169ccd0822781ab714beed57154a51dd","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-22 07:15:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a2156a84f35993c1af8d1fc119552ef87bba0d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 09:56:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18df6519bc774f31b27ee47a2486c229 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1d7d41c8d604a559f094da466430c47 : SUCCESS in 7m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c88fd5467de497cb4a7ff567edf8ff3 : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311cbadd33b74596a97cafae2a536e42 : SUCCESS in 8m 14s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6dcc07443a7543b39722cd4a0740a677 : SUCCESS in 40m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fe2a140637fc41d88bd57c5fc73a2cd5 : SUCCESS in 49m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a761aac0e020406fbc8f174542412f0e : SUCCESS in 43m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2bea5cb6ee74497a9359f78fbe059b7f : FAILURE in 28m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/029a3b49e6634610a3def91ea7b15623 : SUCCESS in 27m 26s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/14204463eb014c86bbc751c962d36884 : SUCCESS in 2h 34m 53s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5c8c70e597774509a06c83fb4949fc4a : SUCCESS in 1h 04m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b09b15858acb4212a194062f22a3f106 : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":1},{"id":"49203d400e68ff48db3cd9b30294ac8d12d145de","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-11-22 21:08:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"978b2e3ddfbd24363af5611c7127ea06f9c662ed","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2019-12-16 09:04:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"606dbf40bc8c88a39c7f4ac63afec7afa8221eb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 11:58:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f5118b9b4e439f9842085515929cf9 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52c1010af5f44da9bd8e2c4130e63337 : SUCCESS in 11m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98406dfd7cf44c01b0059fe66f7f471e : SUCCESS in 11m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f716ee7bbf4587ad98da1f798f8331 : SUCCESS in 11m 05s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fa84c4d699db4f509b4219fdddaa4631 : SUCCESS in 48m 42s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9423a13b4ca1469297ff1be250f59784 : SUCCESS in 48m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9598d563217c45b2983385ee68546c9d : SUCCESS in 49m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1e06ca95703b4d4e90470a49181b0900 : SUCCESS in 33m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/734f66e598a747f9a23dcb15931956bb : SUCCESS in 28m 08s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/09d219bc884343729a85bfcd96cc2d3c : SUCCESS in 2h 47m 18s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7fa6f89074d54af88aedca31e6ab1da5 : SUCCESS in 1h 09m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c23222fd4194f4083a500501955b8a5 : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7503493795a89d9b4c45e130353767bc095b1890","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 11:58:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09db9deb23fe37e00c87af197362fa2c26252cd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 14:00:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6199a4cb20e1464011bd406bfd0d212c8cac246c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 14:00:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1da53d43d1f24e84854e3a4b8a51462d : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3c06ab36b2d4f83b15377322f6cc6cb : SUCCESS in 9m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b20cf35056a4ac8a78366cf3b64dace : SUCCESS in 9m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8606b5c667e745b9a8501c9c05d75204 : SUCCESS in 9m 31s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6512998e352f42e4b78e1457a7c7db09 : SUCCESS in 1h 00m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6611614d945343cbb68ae2ed7fcbb234 : SUCCESS in 1h 03m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7145ba3c2587490ba480a54b927eef32 : SUCCESS in 29m 57s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d710f460a15b461ea9b4020a4966b909 : SUCCESS in 28m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e599564aad554ae7862394411d48aec1 : SUCCESS in 57m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ea7f12a97c54df69786f8361a6f3a92 : SUCCESS in 11m 42s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/dc7628a48c674f10af4f636eec536f96 : SUCCESS in 1h 56m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b3f899e5e8c5c92f6c696528c0693c3c11da4b7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 14:04:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea7ea953ebd4799b66d76507fed5dfd : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"822ea0ec22f894f17955a88cbc5db59fc93dc717","revisions":{"822ea0ec22f894f17955a88cbc5db59fc93dc717":{"kind":"REWORK","_number":1,"created":"2019-11-21 09:29:35.000000000","uploader":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"ref":"refs/changes/78/695378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/78/695378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/695378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/695378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/78/695378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/78/695378/1"}}},"commit":{"parents":[{"commit":"a9b19b5f2502b918163643b453df5bdbc6474ce7","subject":"Merge \"Make it possible to set None to REST API filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a9b19b5f2502b918163643b453df5bdbc6474ce7"}]}],"author":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-21 09:27:09.000000000","tz":0},"committer":{"name":"Dougal Matthews","email":"dougal@dougalmatthews.com","date":"2019-11-21 09:27:09.000000000","tz":0},"subject":"Remove the TripleO standalone job","message":"Remove the TripleO standalone job\n\nDespite the comment in .zuul.yaml this job actually doesn\u0027t use Mistral.\nIt is installed, but not used. This can be seen in the Mistral logs\nwhich have minimal output (other than that of the workbooks being\nloaded).\n\nhttps://c867b5a19c86ae8c3ab7-a51f273e08e05c861a2fa6ace8476176.ssl.cf1.rackcdn.com/695212/2/check/tripleo-ci-centos-7-scenario003-standalone/4827980/logs/undercloud/var/log/containers/mistral/\n\nTh other TripleO job also covers this case and then uses Mistral. This\npatch removes the job which has been wasting resources for quite some\ntime now.\n\nChange-Id: Ie3381c7226d50591192ae61697bdb0d134ddf4f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/822ea0ec22f894f17955a88cbc5db59fc93dc717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/822ea0ec22f894f17955a88cbc5db59fc93dc717"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
