)]}'
{"id":"openstack%2Fmistral~695636","triplet_id":"openstack%2Fmistral~master~Ic1293db03d020f321d4e009ab7906da7622c5daf","project":"openstack/mistral","branch":"master","topic":"fix_heartbeat_log_messages","hashtags":[],"change_id":"Ic1293db03d020f321d4e009ab7906da7622c5daf","subject":"Fix log messages in the action heartbeats mechanism","status":"MERGED","created":"2019-11-22 07:47:14.000000000","updated":"2019-11-22 12:39:24.000000000","submitted":"2019-11-22 12:38:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695636-1574426288709-99007b45","meta_rev_id":"8dd97917403e1f4c8d51c3ff09c8931b2ce59150","_number":695636,"virtual_id_number":695636,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-22 12:38:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2019-11-22 08:43:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":2,"date":"2019-11-22 08:09:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2019-11-22 09:02:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 08:09:13.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-11-22 08:43:17.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-11-22 12:38:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"333fe82aa8d9b84c98e9ad8de1aa486ac60c7bc3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-22 07:47:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8651d53321565fa732989dd182cc5f26d07f2f19","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2019-11-22 08:09:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"98290b414460304017bda230dfee3fd393887802","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-11-22 08:43:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4c274fc236e6ad1f44f5d4fcf04b3eea590fda74","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-11-22 09:02:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a8d1ac9c806a5fa9653ab024b5a18f701784d4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 10:41:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f1d13a73fa64337911cea66cc48edca : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7eaa0c7c2314bc993098cc91b487cff : SUCCESS in 9m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6dfd613d1f747cc9b17a91840d2a3da : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7006e537ae59483e9c14715e299017af : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/563b4a6e7e3940438680e5f7481e4932 : SUCCESS in 9m 18s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9205c047f6e24bf993e24275cc7c5f05 : SUCCESS in 43m 01s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2e137f518fd04ae7993e365f4b112686 : SUCCESS in 43m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4646250621de4b60a67393f5ca5769b5 : SUCCESS in 45m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d9d302da9acc4df6b8a3e7513b7b4cb0 : SUCCESS in 28m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f830a715faff42f89e83557e0808c71a : SUCCESS in 28m 30s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/469879b475e8486ea9bfc219abf2de9b : SUCCESS in 1h 43m 57s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5c7a6f9447fb416d8e6de1be18204f97 : SUCCESS in 2h 48m 03s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/856a811ae36d4b60baf699dabb8359da : SUCCESS in 58m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfe44baec09b4dd49c02f770b2cbdf08 : SUCCESS in 10m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a241ae7bfb93c326c8693680136018c018a331df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 10:41:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"454a8b17b1ed961c54e44b741379bdeb7c691e6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 12:38:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0421834fcb134b7981cb257aae9f61f7 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c43e7c784724b5ca8b324132b35cfee : SUCCESS in 10m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abbfd69b16a24977830ccc7c5c8b927f : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c955e899eb34b428b3d4a0e688ef8a3 : SUCCESS in 8m 37s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e321d46ae97e4927a5d9bbc3ebf1dea4 : SUCCESS in 44m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1d3c531cb1b844fa89d988f0eccfcc5b : SUCCESS in 41m 37s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3a1ce99479024aa2b3a2991ad7f4917f : SUCCESS in 29m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8cf3e7283cd54822928258c3d8fc6ab7 : SUCCESS in 28m 12s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5bc99d0d708a40a7872c5a20608abb9b : SUCCESS in 1h 19m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/aac85c499695497cbad2f2c54209be6c : SUCCESS in 42m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b18e0838b2347e383435ca06b4d6958 : SUCCESS in 10m 19s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4afca49b52c54d61b068153912b70b1b : SUCCESS in 1h 56m 08s","accounts_in_message":[],"_revision_number":1},{"id":"09f3e4085d2389a0a56a1b516897c474b2ac86a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 12:38:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"96be43102212d8b1199f618cb7d17d85257061af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 12:39:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd06170c5e2405cb154e6b8a343e33d : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b58b9c2672684abbe2960e321a8c92ed7b94727","revisions":{"7b58b9c2672684abbe2960e321a8c92ed7b94727":{"kind":"REWORK","_number":1,"created":"2019-11-22 07:47:14.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/36/695636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/36/695636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/36/695636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/36/695636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/36/695636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/36/695636/1"}}},"commit":{"parents":[{"commit":"a9b19b5f2502b918163643b453df5bdbc6474ce7","subject":"Merge \"Make it possible to set None to REST API filters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a9b19b5f2502b918163643b453df5bdbc6474ce7"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-22 07:46:32.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-11-22 07:46:32.000000000","tz":420},"subject":"Fix log messages in the action heartbeats mechanism","message":"Fix log messages in the action heartbeats mechanism\n\nChange-Id: Ic1293db03d020f321d4e009ab7906da7622c5daf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7b58b9c2672684abbe2960e321a8c92ed7b94727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7b58b9c2672684abbe2960e321a8c92ed7b94727"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
