)]}'
{"id":"openstack%2Fmistral~698714","triplet_id":"openstack%2Fmistral~master~Icf460fea113e9279715cae87df3ef88a77575e04","project":"openstack/mistral","branch":"master","topic":"eyalb/yaml","hashtags":[],"change_id":"Icf460fea113e9279715cae87df3ef88a77575e04","subject":"Disable the use of anchors when parsing yaml","status":"MERGED","created":"2019-12-12 12:49:26.000000000","updated":"2020-06-17 11:02:10.000000000","submitted":"2019-12-17 15:28:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698714-1576596489666-71032318","meta_rev_id":"f4e1c3227f4537679f9174f7bbb8cdba334d80b1","_number":698714,"virtual_id_number":698714,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-17 15:28:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2019-12-17 10:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2019-12-17 11:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2019-12-17 11:15:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-16 12:38:43.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-12-17 04:42:17.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2019-12-17 08:03:30.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2019-12-17 08:03:30.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2019-12-17 08:03:30.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2019-12-17 10:37:54.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2019-12-17 11:15:27.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-12-17 15:28:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96ab3c69ec973a58af5c42459ca49c62b2448fc3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-12 12:49:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55271f4d12a8ba9888d973f27ad3ebebd5666c63","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-12 12:52:00.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0a7eead43021eb52f4ba0791fdccf7fd9d261191","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-12 12:52:49.000000000","message":"Topic changed from eyalb/yaml to bug/1785657","accounts_in_message":[],"_revision_number":2},{"id":"3f14d4ad0fbdc51a662b67f315d5e6c7595350c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 16:21:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e69793ac0f00449d9f21a28f67b6e5e2 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c30fc8d0f600435c8187bc8f1da58931 : SUCCESS in 9m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c73a53add2f7492db42c1caca439e1d1 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a24a4d9ab984e82abeab1ac8a98b9fe : SUCCESS in 11m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1188f0949f084a6bb0167c5d67ada197 : SUCCESS in 9m 29s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bb1918c5e4884007b1cb0b80e76a9466 : SUCCESS in 1h 00m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4948f18004e345a48b2e87a84df97f98 : SUCCESS in 1h 00m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0f33cdcce94146cab7f9f3e439c49fe2 : SUCCESS in 44m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/85575737f17a4c1bb80418b64f10c800 : SUCCESS in 28m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/61256d4ea7d64692b02fc649cd4d54f4 : SUCCESS in 29m 01s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/49724c08771f4489b2f27a1960917a8e : SUCCESS in 1h 24m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7fa80f7a3cac4710a8e6797fa9d25544 : SUCCESS in 3h 27m 56s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/271894a5fbbd4bf48e865e626feddd4e : SUCCESS in 1h 00m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42d7f9d2c66b423b9f3068a5aac5de92 : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":2},{"id":"56b1a9735b78ff84a6888ec934a8e3cefd3c0531","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-13 11:10:44.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nLooks good, I\u0027d just prefer we add a comment about why we did so. See inline.","accounts_in_message":[],"_revision_number":2},{"id":"1a9cd2ffd896149ecd2a68adfd2048d45c4b0bc8","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-15 07:44:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"eccccdc821a19e11390476724885c8c1d77260b6","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-15 09:23:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5a987d388f460a98174d807cf8d4e933adce709d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-15 11:19:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"3b0d8c3b3d8429485746f7229049a78bf902ee2d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-15 11:19:15.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"eb3137e540720f5bfa0b5174f4c6c4883373a3b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 13:39:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0f0a3decfd44bd9b5c6f729e2843d5b : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03d711ec2ad547e3846b310d2b33eea2 : SUCCESS in 8m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41437919091046d790e68d447ccb12ff : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1324f9d1890e4afab54d5fff232abce1 : SUCCESS in 6m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeec86da52e74067946bf73b7f80d7ae : SUCCESS in 9m 02s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3b145ae8797a42949f96f201bb54a417 : SUCCESS in 46m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b82db0559d164a5dbe865b6dfce4107b : SUCCESS in 44m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ed9d22f2d5e942f9837206d367668538 : SUCCESS in 47m 15s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b69a5eb509504ab0b2db0603267cc467 : SUCCESS in 28m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/558b8960bd28455988b4bac3cc38a02c : SUCCESS in 28m 06s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/44a51f3a619148b590096f9ed0195fef : FAILURE in 1h 13m 32s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7d3434463f4f4e609672dbf7e21669ef : FAILURE in 2h 15m 22s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d00cc725021d47418c62eb2ef9040cda : SUCCESS in 58m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a18d0d1bd62245ba8a4b0507c0b55f39 : SUCCESS in 10m 48s","accounts_in_message":[],"_revision_number":3},{"id":"8cbb24d96b33fdcd4ae12bcd9eab454affe4ebd4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-16 06:50:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ef1997c77aa6481dba2f69dbb4b67e11039ba632","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-16 06:50:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c3d2a0e2a4fc364ab337fc02c33d49ca646445cc","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-16 08:05:11.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2fab8f582519deb065a2de4bdd7a10206575d1fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 10:09:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f90fd8317e454e29b43da455f501f3bb : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8c5eaecf9924ab68c9ed503e3cd68ab : SUCCESS in 9m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb6799f9d5794216a6fe6851538ac585 : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d145f1cd07d54988b1bc95c9f6bd6eae : SUCCESS in 14m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d9e6913b34844bea4e761a80b055cd2 : SUCCESS in 9m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/73eff43ec20747e1a0d8b7a4aa4b475b : SUCCESS in 51m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/689370b09dee474a86242bd13a20e931 : SUCCESS in 47m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3a0deb544428415fa633c24dbbecd5e4 : SUCCESS in 45m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/946480db6fd940c5a385ccd32bf18144 : SUCCESS in 29m 08s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d97d8bb2abe041fd9c9c7d473280e195 : SUCCESS in 30m 02s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/89be759c02ab4672b205729ae0be780e : SUCCESS in 1h 34m 11s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4d36bd41bc034dffbce9ebfe952a04c7 : SUCCESS in 3h 10m 11s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0732b69fade646108df51f4dc45bbca3 : SUCCESS in 58m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7bdbbb9570e41f9bab892511cd58c80 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":3},{"id":"d3fe489b1f76c09aed756c03a2c8072722099188","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-16 12:38:43.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"df2d0ed09e57387f00d492ecfc1f1fa9e5fe47f3","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-12-17 08:47:07.000000000","message":"Patch Set 3: Code-Review-1\n\nThis is great. Thanks. The import should is probably be better renamed to safe_yaml everywhere, so that there is no confusion later in the code.","accounts_in_message":[],"_revision_number":3},{"id":"613d058854cbb76070729824e79d1abdd01e9615","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-17 09:04:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5e7baea7c8b0372661c51b6180f3c7b9e2ee8774","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-17 09:38:23.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"516952676762e15e04236d9ed92c3add0611d6dc","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-17 10:08:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"47d2721a41f203e8d5ffaa6eae73fef8dbeba8c8","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2019-12-17 10:37:54.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"eaf0c462a7b1c67fe3c7986a76b3accba42a7fa8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-17 11:15:27.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"48d4fff464eb6e8922f4a7c77973ca8e6f6c0010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 12:46:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e4b05f4fc8541b791f8d44659ea9d26 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c73458cf52014dd4b6fad75866bc6620 : SUCCESS in 9m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dedf6711fdbf4b77985db47638869aa0 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dafeca1c7101458faccb977fbf4a8183 : SUCCESS in 8m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/315612461f9940c4aa2f9ba761f16ec7 : SUCCESS in 10m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d66e4e4600a54c8db275612e26994341 : SUCCESS in 1h 01m 52s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/892fd8cb1d454ab385a7b1a2b0d1dbf3 : SUCCESS in 53m 50s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/30f3c441b7844172a623377036304b17 : SUCCESS in 49m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9286d01a64d84c31b2b3b6f7f682c993 : SUCCESS in 29m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/950afda71c5649fba1445e7c6c89478d : SUCCESS in 30m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b1d2768112b14b1bbc850ff663fde1f6 : SUCCESS in 2h 33m 27s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9f5af20216e5466e8fbbe66c573857ef : SUCCESS in 1h 00m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76c61fabb0a14cfe836287971a556bff : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":5},{"id":"a287daf00d2e161bf0354f30967a76916feead1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 12:47:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b1c314459e42bca12f7baa6cf7d08ff8ddef3931","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:28:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"f1005a275f7f89111ebd515cfa04e659512fbd88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:28:09.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/695acec56f7043519ebe2ed3b4462cef : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7683eda44a04fb79272a803ed7c935c : SUCCESS in 11m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f4ba012ed924c2e9e5a09012c0e6051 : SUCCESS in 12m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87afd4dd702a4401ba9007fc76dce816 : SUCCESS in 11m 22s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2ff573c6520c4263ac7474f42422c9c8 : SUCCESS in 56m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2f612a0a5aa345fb8209affa05e41fbb : SUCCESS in 55m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8a6cc695e00d4092bee326ccb428dac9 : SUCCESS in 34m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5db038bfa7c845fbb613bd6b922fb966 : SUCCESS in 33m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/090c9c1416384325b3a7cf0ce79d3444 : SUCCESS in 1h 17m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b775c3da86fa4c509e5488c81fc88892 : SUCCESS in 10m 56s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/29e5b16e26f448c4bff7558905d601d5 : SUCCESS in 2h 39m 56s","accounts_in_message":[],"_revision_number":5},{"id":"2ca281ecb05a8a849c155c993076146b93e0ad4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:30:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3285d867130648e98750dd8dae1c77a3 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":5},{"id":"6f56d3c9374d977ff47d604331075d2e5ee2c0f9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-06-17 11:02:10.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit f0824730fcc4bed3614071be2a1fc82a0ed29114","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"eac23d9e774f658f9d4743c99aa2743eb104c3f9","revisions":{"265c8b12007b15ed6157ad314042d8ae32cda020":{"kind":"REWORK","_number":1,"created":"2019-12-12 12:49:26.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/14/698714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/698714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/698714/1"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-12 12:49:13.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-12 12:49:13.000000000","tz":120},"subject":"Disable the use of anchors when parsing yaml","message":"Disable the use of anchors when parsing yaml\n\nThis can be used as a DDoS attack\n\nChange-Id: Icf460fea113e9279715cae87df3ef88a77575e04\nCloses-Bug: #1785657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/265c8b12007b15ed6157ad314042d8ae32cda020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/265c8b12007b15ed6157ad314042d8ae32cda020"}]},"branch":"refs/heads/master"},"75aa20e6ecbe23dd58c2facc6246a0a7b6a3b058":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-12-12 12:52:00.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/14/698714/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/698714/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/698714/2"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-12 12:49:13.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-12 12:51:08.000000000","tz":120},"subject":"Disable the use of anchors when parsing yaml","message":"Disable the use of anchors when parsing yaml\n\nThis can be used as a DDoS attack\n\nCloses-Bug: 1785657\nChange-Id: Icf460fea113e9279715cae87df3ef88a77575e04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/75aa20e6ecbe23dd58c2facc6246a0a7b6a3b058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/75aa20e6ecbe23dd58c2facc6246a0a7b6a3b058"}]},"branch":"refs/heads/master"},"e749b85ad6ae6b44ce406192559d03aadaa9aa5b":{"kind":"REWORK","_number":3,"created":"2019-12-15 09:23:15.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/14/698714/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/698714/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/698714/3"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-12 12:49:13.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-15 09:22:56.000000000","tz":120},"subject":"Disable the use of anchors when parsing yaml","message":"Disable the use of anchors when parsing yaml\n\nThis can be used as a DDoS attack\n\nCloses-Bug: 1785657\nChange-Id: Icf460fea113e9279715cae87df3ef88a77575e04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e749b85ad6ae6b44ce406192559d03aadaa9aa5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e749b85ad6ae6b44ce406192559d03aadaa9aa5b"}]},"branch":"refs/heads/master"},"b1db6cac443a710518ac276142c3545e075e8054":{"kind":"REWORK","_number":4,"created":"2019-12-17 09:04:08.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/14/698714/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/698714/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/698714/4"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-12 12:49:13.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-17 09:03:55.000000000","tz":120},"subject":"Disable the use of anchors when parsing yaml","message":"Disable the use of anchors when parsing yaml\n\nThis can be used as a DDoS attack\n\nCloses-Bug: 1785657\nChange-Id: Icf460fea113e9279715cae87df3ef88a77575e04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b1db6cac443a710518ac276142c3545e075e8054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b1db6cac443a710518ac276142c3545e075e8054"}]},"branch":"refs/heads/master"},"eac23d9e774f658f9d4743c99aa2743eb104c3f9":{"kind":"REWORK","_number":5,"created":"2019-12-17 10:08:21.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/14/698714/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/14/698714/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/14/698714/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/14/698714/5"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-12 12:49:13.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-17 10:08:10.000000000","tz":120},"subject":"Disable the use of anchors when parsing yaml","message":"Disable the use of anchors when parsing yaml\n\nThis can be used as a DDoS attack\n\nCloses-Bug: 1785657\nChange-Id: Icf460fea113e9279715cae87df3ef88a77575e04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eac23d9e774f658f9d4743c99aa2743eb104c3f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eac23d9e774f658f9d4743c99aa2743eb104c3f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
