)]}'
{"id":"openstack%2Fmistral~699368","triplet_id":"openstack%2Fmistral~master~I07f3f7e83f7521c320f6b21b806692289c8a01ab","project":"openstack/mistral","branch":"master","topic":"mistral-coding-guidelines","hashtags":[],"change_id":"I07f3f7e83f7521c320f6b21b806692289c8a01ab","subject":"Add the 1st version of Mistral coding guidelines into the docs","status":"MERGED","created":"2019-12-17 08:25:56.000000000","updated":"2020-01-10 13:32:45.000000000","submitted":"2020-01-10 13:31:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699368-1578663073144-46a14a57","meta_rev_id":"b86ad2f4f3b12034a58fea687736c6ad49cced2f","_number":699368,"virtual_id_number":699368,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2020-01-10 09:58:43.000000000","_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 13:31:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-01-10 09:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2020-01-09 10:37:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-01-10 09:15:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":1,"date":"2020-01-10 13:20:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-10 06:58:16.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 10:37:10.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-01-10 06:58:16.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-01-10 09:15:12.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-01-10 13:20:54.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-01-10 13:31:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"abb4d9bb14c562a0599b32dc8e5e774b900e807b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-17 08:25:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3f50e89c81894eac43e159b2a2c1d8872e48089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 08:40:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bdbea61efc8490faaaea2d20d6c0bfa : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe587daa5da44050a3233264ad94c0b6 : SUCCESS in 9m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0c524dcca837fc4ae45a33e1bcedc358cb525b05","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-17 09:25:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e1d99b0d1984a5d712f1313259a1703d93ddeb86","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-17 10:05:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f6921326c5f50087f1731c65212d8f4c10ada5ab","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-17 10:05:35.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"338f2e36af4a1735d6d2aa98195d728150f76a36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 10:16:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c58d42e27e2c43eaba7ecd1f014cc371 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad91be59d1e84ff89d81c797ddf77fd7 : SUCCESS in 9m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a240b5e2816b54b36a7acbbbfe8c015f5a45fc00","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-17 12:15:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e49907f9a1526f49e55e3b93313a572037cfec8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 12:25:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4476b8c97dc34a7f81950651997b2c2a : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba446c6499604021b292545e20de929e : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":3},{"id":"0f5c5c5ffb84a1441cde007897f8cfaef3764237","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-18 09:01:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"51fd6505b95d33d2971641c8f496c42ad609a0fd","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-18 09:02:25.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"ac86f4df141fbfe8d4a5713b153e6316f9c213bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 09:17:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76048292b3474129950a4603567efabf : FAILURE in 11m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba12b2f905b438ea23a2d688d94a12b : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":4},{"id":"07701ebf74a1137efb0d980e4b940680e8a308a5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-18 10:01:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4ec694cd5c34d7a1a10c612999292e4fa090f5c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 10:15:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3a0198575b41d494300e4aeeb91eee : SUCCESS in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1271f02c50274ccaabf09e77deade511 : SUCCESS in 11m 12s","accounts_in_message":[],"_revision_number":5},{"id":"33363bd53d427b268c307f3914510e70ce0068db","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-19 06:55:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4b2fd93ae2ce84e1f77dc2dc7a2650c5dffbd2d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 07:06:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6de55e63db1c4b7496bda2a0000019ae : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5b19049e5a4adcb342357b3e525699 : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":6},{"id":"b7dbb724f63b12ce1aeede5c6a1d804acffc4662","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-24 08:54:02.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bba7441526325a4a4cf7f00368e6a1770d34e6fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 09:04:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41da9a9ce6b44e3d9e5a5078a82cb8ad : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b8280acca1b4822ba307f949d63451d : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":7},{"id":"21bcf1fbabe542ead812dcd3db85c83cafd658ea","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-24 09:25:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"924d9afbb8952938adaa365ac6b7b6a3c02494cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 09:33:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c95b107ab02e4815a1d201b73819fbcb : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa61248ff1c543739e76dce55e80e934 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":8},{"id":"a80b4c363a70e5b8c5244834454086dd21767b2d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-25 07:55:20.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7e31f1d6cbe9d5c72345115c3fa5f005ce15e6e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-25 08:04:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7098f20c2ce34a1699092595701088aa : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc7aaaad8184326942fabf2103be502 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":9},{"id":"07b690a5906a9cb86472f6c670d3e4b8902d92c0","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-05 09:02:10.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"fa95ea560c445f8649ba006243be6cfa2f881460","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-09 10:17:50.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"eeeb6f7be6103e81d441dd8b73b58a8d351203f8","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-09 10:37:10.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"f0d2de69bc3d1a8b4af024482531f7ed629d836e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 10:45:15.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9e21f0aa67439bbc1322977d6ec35e : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c086ec13ec47a1b8f41d726fb65ff5 : SUCCESS in 10m 40s","accounts_in_message":[],"_revision_number":10},{"id":"8702a6eadb0dca2038c71df1eab15cc41858c3ba","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-01-10 09:02:28.000000000","message":"Patch Set 10: Code-Review+2\n\nVery good! I like it. Is this something you created or has it been adapted from some other project?","accounts_in_message":[],"_revision_number":10},{"id":"55afa747462b11af405bb96ce0bce1ef9d55879c","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-10 09:15:12.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"51fc3014bb11bee60a8547703443ca0f4326e793","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-10 09:58:43.000000000","message":"Patch Set 10:\n\nWell, it is based on my previous Java experience, to be honest (although I realize it might not be a popular thing to say within this community). During Mistral development I was collecting these practices and finally decided to write a document. Of course, for Python it\u0027s different, I tried to adapt them keeping Python\u0027s nature in mind. And this document is not full yet, I just think that it\u0027s already big enough for one patch. I\u0027ll be adding more things step by step when moving forward.\n\nOne good source of info that has also inspired me is the book called \"Implementation Patterns\" written by Kent Beck. It describes similar things, for Java.","accounts_in_message":[],"_revision_number":10},{"id":"64ffcffd2cfcc1548e05c02017a9f08afdf3e671","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-01-10 13:20:54.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"34715a65aa92f6c0312cf88a6bc98258ad9dcfec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 13:21:06.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"57eea12405991f0c991b95f8e1f7a73d8031756e","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-01-10 13:21:40.000000000","message":"Patch Set 10:\n\n\u003e Well, it is based on my previous Java experience, to be honest\n \u003e (although I realize it might not be a popular thing to say within\n \u003e this community). During Mistral development I was collecting these\n \u003e practices and finally decided to write a document. Of course, for\n \u003e Python it\u0027s different, I tried to adapt them keeping Python\u0027s\n \u003e nature in mind. And this document is not full yet, I just think\n \u003e that it\u0027s already big enough for one patch. I\u0027ll be adding more\n \u003e things step by step when moving forward.\n \u003e \n \u003e One good source of info that has also inspired me is the book\n \u003e called \"Implementation Patterns\" written by Kent Beck. It describes\n \u003e similar things, for Java.\n\nTHUMBS UP.","accounts_in_message":[],"_revision_number":10},{"id":"1cfda1901c93237b3188b668db85c7675c182e71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 13:31:12.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff01983064846afbf1df784f9efe107 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e35a440caa734235b0f73abd2fa20004 : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":10},{"id":"48ac6ffd8f820cbf26bae70adc55770244d19b17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 13:31:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"03a46ce381c581c16b4ed4ca0075e5d5b47bb2bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 13:32:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f73a8ee0b0e34a219efb76956788e131 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"63ab3d4062f9035a18be763c9dd28412bc87a7c8","revisions":{"8195a96df59a355e1a3c6f1e6707bd11322519fe":{"kind":"REWORK","_number":1,"created":"2019-12-17 08:25:56.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/68/699368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/699368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/699368/1"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-12 10:50:18.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-17 08:25:15.000000000","tz":420},"subject":"WIP: add Mistral coding guidelines section into the documentation","message":"WIP: add Mistral coding guidelines section into the documentation\n\nChange-Id: I07f3f7e83f7521c320f6b21b806692289c8a01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8195a96df59a355e1a3c6f1e6707bd11322519fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8195a96df59a355e1a3c6f1e6707bd11322519fe"}]},"branch":"refs/heads/master"},"ef08453de62b7f64bed822199047ed196eff6179":{"kind":"REWORK","_number":2,"created":"2019-12-17 10:05:22.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/68/699368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/699368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/699368/2"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-12 10:50:18.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-17 10:05:03.000000000","tz":420},"subject":"WIP: add Mistral coding guidelines section into the documentation","message":"WIP: add Mistral coding guidelines section into the documentation\n\nChange-Id: I07f3f7e83f7521c320f6b21b806692289c8a01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ef08453de62b7f64bed822199047ed196eff6179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ef08453de62b7f64bed822199047ed196eff6179"}]},"branch":"refs/heads/master"},"df01f9919b1967677a0bcc10313b5730d5e6726d":{"kind":"REWORK","_number":3,"created":"2019-12-17 12:15:14.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/68/699368/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/699368/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/699368/3"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-12 10:50:18.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-17 12:14:58.000000000","tz":420},"subject":"WIP: add Mistral coding guidelines section into the documentation","message":"WIP: add Mistral coding guidelines section into the documentation\n\nChange-Id: I07f3f7e83f7521c320f6b21b806692289c8a01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/df01f9919b1967677a0bcc10313b5730d5e6726d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/df01f9919b1967677a0bcc10313b5730d5e6726d"}]},"branch":"refs/heads/master"},"2fff4de82149232b4cead2ddb55aa1ed7b2e90fd":{"kind":"REWORK","_number":4,"created":"2019-12-18 09:01:53.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/68/699368/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/699368/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/699368/4"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-12 10:50:18.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-18 09:01:37.000000000","tz":420},"subject":"WIP: add Mistral coding guidelines section into the documentation","message":"WIP: add Mistral coding guidelines section into the documentation\n\nChange-Id: I07f3f7e83f7521c320f6b21b806692289c8a01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2fff4de82149232b4cead2ddb55aa1ed7b2e90fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2fff4de82149232b4cead2ddb55aa1ed7b2e90fd"}]},"branch":"refs/heads/master"},"6f179d3ad6a77892723536dfefc1959aa87d1d23":{"kind":"REWORK","_number":5,"created":"2019-12-18 10:01:55.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/68/699368/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/699368/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/699368/5"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-12 10:50:18.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-18 10:01:40.000000000","tz":420},"subject":"WIP: add Mistral coding guidelines section into the documentation","message":"WIP: add Mistral coding guidelines section into the documentation\n\nChange-Id: I07f3f7e83f7521c320f6b21b806692289c8a01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6f179d3ad6a77892723536dfefc1959aa87d1d23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6f179d3ad6a77892723536dfefc1959aa87d1d23"}]},"branch":"refs/heads/master"},"7c19164828989967169a608fbf412c0a64b7890c":{"kind":"REWORK","_number":6,"created":"2019-12-19 06:55:35.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/68/699368/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/699368/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/699368/6"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-12 10:50:18.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-19 06:55:13.000000000","tz":420},"subject":"WIP: add Mistral coding guidelines section into the documentation","message":"WIP: add Mistral coding guidelines section into the documentation\n\nChange-Id: I07f3f7e83f7521c320f6b21b806692289c8a01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7c19164828989967169a608fbf412c0a64b7890c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7c19164828989967169a608fbf412c0a64b7890c"}]},"branch":"refs/heads/master"},"6b2896641e0483bbcfe66caa6902b4209c1379c8":{"kind":"REWORK","_number":7,"created":"2019-12-24 08:54:02.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/68/699368/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/699368/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/699368/7"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-12 10:50:18.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-24 08:53:29.000000000","tz":420},"subject":"WIP: add Mistral coding guidelines section into the documentation","message":"WIP: add Mistral coding guidelines section into the documentation\n\nChange-Id: I07f3f7e83f7521c320f6b21b806692289c8a01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6b2896641e0483bbcfe66caa6902b4209c1379c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6b2896641e0483bbcfe66caa6902b4209c1379c8"}]},"branch":"refs/heads/master"},"cb1f1e46a3375657638bd373967d35e61c5fe01b":{"kind":"REWORK","_number":8,"created":"2019-12-24 09:25:10.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/68/699368/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/699368/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/699368/8"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-12 10:50:18.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-24 09:24:18.000000000","tz":420},"subject":"WIP: add the 1st version of Mistral coding guidelines into the docs","message":"WIP: add the 1st version of Mistral coding guidelines into the docs\n\nChange-Id: I07f3f7e83f7521c320f6b21b806692289c8a01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cb1f1e46a3375657638bd373967d35e61c5fe01b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cb1f1e46a3375657638bd373967d35e61c5fe01b"}]},"branch":"refs/heads/master"},"e07a67f6cefdc43b1edccb1f6369c69a1049a132":{"kind":"REWORK","_number":9,"created":"2019-12-25 07:55:20.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/68/699368/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/699368/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/699368/9"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-12 10:50:18.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-25 07:55:00.000000000","tz":420},"subject":"WIP: add the 1st version of Mistral coding guidelines into the docs","message":"WIP: add the 1st version of Mistral coding guidelines into the docs\n\nChange-Id: I07f3f7e83f7521c320f6b21b806692289c8a01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e07a67f6cefdc43b1edccb1f6369c69a1049a132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e07a67f6cefdc43b1edccb1f6369c69a1049a132"}]},"branch":"refs/heads/master"},"63ab3d4062f9035a18be763c9dd28412bc87a7c8":{"kind":"REWORK","_number":10,"created":"2020-01-09 10:17:50.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/68/699368/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/68/699368/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/68/699368/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/68/699368/10"}}},"commit":{"parents":[{"commit":"0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e","subject":"Merge \"Add a release note for graceful scale-in feature\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e0ab09d544d8e5f51f88ba0b3af229d8fb9091e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-12 10:50:18.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-09 10:16:56.000000000","tz":420},"subject":"Add the 1st version of Mistral coding guidelines into the docs","message":"Add the 1st version of Mistral coding guidelines into the docs\n\nChange-Id: I07f3f7e83f7521c320f6b21b806692289c8a01ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/63ab3d4062f9035a18be763c9dd28412bc87a7c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/63ab3d4062f9035a18be763c9dd28412bc87a7c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
