)]}'
{"id":"openstack%2Fmistral~700507","triplet_id":"openstack%2Fmistral~master~I772651e33eada4a5c2149bfa867095c277eddeed","project":"openstack/mistral","branch":"master","topic":"coordination","hashtags":[],"change_id":"I772651e33eada4a5c2149bfa867095c277eddeed","subject":"Add coordination support for devstack","status":"MERGED","created":"2019-12-24 11:40:51.000000000","updated":"2019-12-26 13:27:38.000000000","submitted":"2019-12-26 13:26:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700507-1577366784157-0548d2f5","meta_rev_id":"afa936551984d82bdadfae44244c95927138574d","_number":700507,"virtual_id_number":700507,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-26 13:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-26 07:26:24.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2019-12-25 09:01:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2019-12-26 05:11:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-24 15:32:20.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2019-12-26 05:11:37.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2019-12-26 13:26:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc68330408f1993e3a22c97749c7df5a39787880","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-24 11:40:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3eb89b805f1b3b7dfecfda52e6f2f32b5e33ce94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 15:10:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9944bdab9ebb4e48b131e6524385facf : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0f9485bd3f2486683998a196773c858 : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b1697c99ac714379a77dcc7d5b6fcead : SUCCESS in 8m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f04c03ed6c447bea1843079f23a0e45 : SUCCESS in 17m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9b145503dda48babdfc8f69c14879fa : SUCCESS in 9m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4a7fc510d7e741a0b10405ff230c4536 : SUCCESS in 50m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/656ab021609a49d7ac08b9e0a3b96e1a : SUCCESS in 1h 01m 33s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2426da4b19764d8a8d9a9ee7f8b6ff8b : SUCCESS in 52m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9e00cbc480144e02884237f5b206ea6c : SUCCESS in 31m 15s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/19a7cf8c2d814a7db3440a7063f4cb6c : SUCCESS in 28m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1b7bc36da8a84fccb329dcaed38a9bba : SUCCESS in 2h 54m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1208333d6ff84578955e7c33d5675b1b : SUCCESS in 1h 02m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e0d4f78633f42a7a49bc189a85173da : SUCCESS in 10m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a92f7c9ce88fde9afd9cda4e1631774d47a81922","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2019-12-24 15:32:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4d4b0ebd02163cc9ca45bfd7f155b7a011aa3e77","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-25 04:56:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d2965937230a84dc786c112b089722904d45bb3e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-25 07:55:07.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a3ddb9bb16cc39fd677ce6107b826717aad09ea4","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-25 08:45:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3869462cee8503f92a5929e684a8bded5a85d9ea","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-25 09:01:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ebbdc844437662192c463d26763b3d795288ae62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-25 11:45:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9337b63f0a44ae3be67e9c6811983d9 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a50771139c234f5db7840c7fcb581c52 : SUCCESS in 7m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c748bcd17b146e08035cf6e0cd1d737 : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602bfa8c718d45f286bdf3504207a1e9 : SUCCESS in 6m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c91010d8e292458e99ba51ddfbbf2811 : SUCCESS in 8m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fbb8512b0b414e37afce42da76ec53e6 : SUCCESS in 43m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e7b4512e18c44d0fbc17705e7e71303c : SUCCESS in 44m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1fcdacc8557248bf8e850ba944191030 : SUCCESS in 42m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a6e716a9c0ee43e383326cea3e2e575f : SUCCESS in 28m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/71e5bbbbff484fef95cfc1a0bd7de075 : SUCCESS in 28m 29s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7c9d4c1fd33b41ed83ff6810cb046f55 : SUCCESS in 2h 54m 09s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0df2fd5c81c14ec1934e09724f592e5e : SUCCESS in 1h 02m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/393c2b44d45c436c817ea4159d68cdea : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":2},{"id":"be486cbd7324a0033e4d6b81317b7bf2336f6dec","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-26 05:11:37.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9cd0f8426abfb9edc2efc5c632262bd3b5ea0d73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 05:11:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b3ba12a84e314811dc15db1b136d313189fff215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 07:24:46.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff8630dc36654007b0d798aafc51337b : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0adcff353ce04bd5916a574731e275ea : SUCCESS in 9m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/533d335080b34214b3800f97023bbb69 : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d793b169e8d140b7a2fa40783b98cd60 : SUCCESS in 12m 36s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a7bf6ed417b646dcb0a6e221bdd077b0 : SUCCESS in 49m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cd41b675cfb94e7fa5667da1592c416f : SUCCESS in 48m 28s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/715f4ae1e95d423d88c7c86cdbdcb280 : FAILURE in 33m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bbab69fc3309452caf04d543e7c67940 : SUCCESS in 32m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cfc132bcf06a4b9085e4b8f3ddeb384b : SUCCESS in 48m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a0a783d58a74e14984e8ff0ad53c74f : SUCCESS in 13m 37s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/dd678e9381894f4ba2c7e3d0bd2c8089 : SUCCESS in 1h 53m 07s","accounts_in_message":[],"_revision_number":2},{"id":"697fedd972e204f6bd1deb8cd893e464fd1c9b3d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-26 07:26:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a508ea8d24513734f6eb8d5b93c94917d466734c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-26 08:35:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"cc4055d68d1d17e9a1498595905e09d0d7be0fd3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-26 08:35:57.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"81e639c192c0f1d21e2ef8dbe31efd62ddcd4984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 11:38:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aab2d40f1b504926be2f23e083e24870 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e8f23b5b6ae44168c741595421057c5 : SUCCESS in 11m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ef7a7e0043d4b379d9e933ae9a77c93 : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bbad5e34979411791d7a5bee620e6d6 : SUCCESS in 9m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adfd5b8f89d54dfdaaa60c9e75aa849f : SUCCESS in 9m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/67b2c0591ac04e51871b270b44c11815 : SUCCESS in 57m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/639e9d9915db49128ad7c3737a0f7d56 : SUCCESS in 54m 14s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b6a8f76165c64cbb8c8589cbdf49ae81 : SUCCESS in 49m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f76f76f883ff4be5ae3435b330037162 : SUCCESS in 31m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fa60129a4dfe41029a8911b44f3c8c21 : SUCCESS in 30m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d58f9da79b1c443693dbd27a98af7f93 : SUCCESS in 2h 40m 47s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/84c00bc618f0407186c169165e5ebbfa : SUCCESS in 1h 01m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6df3ef208bd44984b2332871e13c5540 : SUCCESS in 10m 14s","accounts_in_message":[],"_revision_number":2},{"id":"075e73253864adaadb9b16b493873752bea7195d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 11:38:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"239a8ad9cfcdc1722f2a01b7ba18d14ece11b0fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 13:26:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c46692bd9094cb5a0599c5b49f8a144 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5ac5a767e4c4cef919ea28cda5cd589 : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a9ab67d7d6d44bcbc64e8163d77803e : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c34440c042104072871f01c343ca3fa4 : SUCCESS in 6m 44s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/58626a3e946146a4a1079edb8dd43fbe : SUCCESS in 44m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4272fe4ea89643f0a12c3c2bcf8d10a6 : SUCCESS in 44m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/798cc9ff16a24d52a40e055f93fc8276 : SUCCESS in 28m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/44942fba2c064bc6a21df00659f23e3e : SUCCESS in 28m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/cf78d311b6ac4f8c98679261aff302ac : SUCCESS in 46m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62bc1c0ba1ed4ee68d94e42505acf08d : SUCCESS in 9m 34s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f85aa3f8f8e84a598554315dbd7eda54 : SUCCESS in 1h 47m 40s","accounts_in_message":[],"_revision_number":2},{"id":"30e40b4d8db475e611634c7e3603af6b30198469","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 13:26:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3afd5d496b2c132d282cc077807f8eafc179724f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-26 13:27:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1942c6433cd94ef88c5f0968ab26318a : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d838607b2f2d112e867ce50dfae397e8eb0af633","revisions":{"f067fd459bfdea02d5634542f71f236239f2304f":{"kind":"REWORK","_number":1,"created":"2019-12-24 11:40:51.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/07/700507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/07/700507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/700507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/700507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/700507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/07/700507/1"}}},"commit":{"parents":[{"commit":"541aabbfe8a7afd876adcfb209d18aafdef2d5f6","subject":"Merge \"Use MISTRAL_SERVICE_HOST as the host ip for standalone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/541aabbfe8a7afd876adcfb209d18aafdef2d5f6"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-24 11:40:37.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-24 11:40:37.000000000","tz":120},"subject":"Add coordination support for devstack","message":"Add coordination support for devstack\n\nThis way we can test the service api later on gate\n\nChange-Id: I772651e33eada4a5c2149bfa867095c277eddeed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f067fd459bfdea02d5634542f71f236239f2304f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f067fd459bfdea02d5634542f71f236239f2304f"}]},"branch":"refs/heads/master"},"d838607b2f2d112e867ce50dfae397e8eb0af633":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-12-25 08:45:30.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/07/700507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/07/700507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/700507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/700507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/07/700507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/07/700507/2"}}},"commit":{"parents":[{"commit":"541aabbfe8a7afd876adcfb209d18aafdef2d5f6","subject":"Merge \"Use MISTRAL_SERVICE_HOST as the host ip for standalone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/541aabbfe8a7afd876adcfb209d18aafdef2d5f6"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-24 11:40:37.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-25 08:38:29.000000000","tz":120},"subject":"Add coordination support for devstack","message":"Add coordination support for devstack\n\nThis way we can test the service api later on gate\n\nIn order to use etcd in gate few changes were made:\n* All identifiers must be byte type (group type, member id)\n* Tooz has a built-in mechanizm for heartbeat no need to implement it\n* Need to use eventlet monkey patch since etcd client uses blocking\n  methods\n* Services name must be identical to LAUNCH_OPTIONS used in cli\n* Gate coordination url should be define with a schema of etcd+http\n  which is the etcd gateway and works better then just etcd\n\nChange-Id: I772651e33eada4a5c2149bfa867095c277eddeed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d838607b2f2d112e867ce50dfae397e8eb0af633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d838607b2f2d112e867ce50dfae397e8eb0af633"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
