)]}'
{"id":"openstack%2Fmistral~700695","triplet_id":"openstack%2Fmistral~master~I0eecec4b4e64381cac005622b16c6d9e4bed4df6","project":"openstack/mistral","branch":"master","topic":"bug/1857871","hashtags":[],"change_id":"I0eecec4b4e64381cac005622b16c6d9e4bed4df6","subject":"Fix keycloak authentication","status":"MERGED","created":"2019-12-27 09:21:43.000000000","updated":"2020-01-20 17:20:01.000000000","submitted":"2020-01-17 10:10:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700695-1579255805484-46a14a57","meta_rev_id":"d5981d175bb449fc764ccfca3f965c3e986b0b33","_number":700695,"virtual_id_number":700695,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"date":"2020-01-15 07:24:53.000000000","_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-17 10:10:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2020-01-14 13:59:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-01-15 09:53:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2020-01-15 10:53:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-15 08:12:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"date":"2020-01-15 07:02:23.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2020-01-17 08:06:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-14 13:59:52.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-01-15 07:02:23.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-01-15 08:12:08.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-01-15 10:53:43.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-01-17 08:06:50.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-01-17 10:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a41d064b36e3068dc35c58d9e02354ce52cf80c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-27 09:21:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62a50d846875341a124befdc9b80858c12eeb43d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-27 09:35:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ccc5ce5d5af613f6771c222461261fddda39a6c5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-27 11:16:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d23c570d192542af010dea3a3e22377a93dae8f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-27 13:50:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f9d26f5446a41a8a72fbd444b77b567 : FAILURE in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1ac701149c9451380655d78a5131049 : FAILURE in 8m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02c68f5cf1f3400084b1b6abe54c6678 : FAILURE in 9m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aae01ceee1554e939c7322eff74fd9bd : SUCCESS in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/494a5781a7834b7b81729c2d62299fe1 : FAILURE in 8m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/635067318cfc4871a3f4df468576c71f : SUCCESS in 40m 55s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e562c7d1fa40457689d92c4637c8bd64 : SUCCESS in 40m 44s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e36b89d8b5fa4df5bac740f41fe4b1a0 : SUCCESS in 40m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/773a81c9f4d34eb288fdd2d8ec85b6c5 : SUCCESS in 39m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/42b2458262c84a438ea1dec8347c899e : SUCCESS in 40m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f9aa6fc35f6f4bd7946f68c35beed938 : FAILURE in 36m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9c3212361dbe4cc58f70bc2b271364c2 : FAILURE in 28m 24s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e08a9f29703949ed9565c8e0137213da : SUCCESS in 2h 21m 32s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/383f4fb0816944f99335506e083d687b : SUCCESS in 56m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c6da6bcb10f4b309d76d85415d0e4ef : FAILURE in 9m 44s","accounts_in_message":[],"_revision_number":3},{"id":"0d5e736248c397746a41c470184921353da68bfa","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-30 05:18:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5c576d79ce88b49fa57975c005ab23d0e271ff42","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-30 05:28:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"40f14d8d810c007bce87776e969a01d5e66b1da2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-30 07:03:08.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6970b8070a2b68469fcb6121fff66dec2ff1018f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 09:49:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/211dce04cd854f3d84b59f222bebb768 : SUCCESS in 12m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13bcac0d548c43e69373cb48c60da413 : SUCCESS in 13m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a14d6751f64461abc5a4cf06d5c2e71 : SUCCESS in 15m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a782b64db14117b05065b2b9b968e1 : SUCCESS in 17m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3b943f50b42a4e21ba33442d01d01811 : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e778ea33c2764de9a77549bc8c539710 : SUCCESS in 14m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fb92743c25d142ceab3296e5c438c9b4 : SUCCESS in 47m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcbbf113c3374747b36b63e3fa9a7785 : SUCCESS in 50m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfab4652f2794fc299c921503844c491 : POST_FAILURE in 47m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9ab2ee579516497aba71daf2d038917e : SUCCESS in 49m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4f8d0a75c3db46a6a16070029142e3cd : SUCCESS in 53m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/45194d51cbcf44709267b3b6b3ed2b22 : SUCCESS in 30m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ad04ca50334243a6849c1e64a4bd7813 : SUCCESS in 34m 22s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/03509fb6e9494f29a6c863dac276293c : SUCCESS in 2h 42m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5e1da5d33eaf49c1a36db12ffd6b1188 : SUCCESS in 1h 03m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/621bbef4e5f64b3dad1813a84e70e069 : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":6},{"id":"1bd8186ac7258fbe7f1d9479c5792a21885f11a7","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-30 11:03:53.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"ddf84170e5b38f292eff580e9506c6cb94b1b222","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-30 11:47:01.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"03f306f58af0e327f18f39b9533275129d42eef2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 14:57:57.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f198b6ba4ab4c8bb81c2bc84e08d2af : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5505cb5a79054d439d13f65643f48471 : SUCCESS in 8m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f1b71acf8164966ad8e9a431d7b95bc : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d49b7eb6e6344a893afee30afc09caf : SUCCESS in 7m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2e497c93d4cc4657aeef4b3bd185d456 : SUCCESS in 2m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c131eabf950240638c3d8e5e68aba518 : SUCCESS in 9m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/16a7bd6986884e169bae30f358f31038 : SUCCESS in 40m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63b93bed62684e70a9cae66b9dc8b7ce : SUCCESS in 37m 59s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a36dd7cf8ac4a438269e22275c0a7b0 : SUCCESS in 36m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/525bfe76b7ad4ed7babf7f30bf3a8d78 : SUCCESS in 52m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/21fc8bc90bb347858cf8a918cfc8d6e7 : SUCCESS in 38m 37s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1ced092b337244a2887bffb4af2c762e : SUCCESS in 27m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f1a613f43a1943dabd89896486b03172 : SUCCESS in 28m 19s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0c619a5e92584bb49a5f6ab97354d6bc : TIMED_OUT in 3h 22m 33s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1727895b20054379a9c4a90f12f0e723 : SUCCESS in 59m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6d260ef0f61494f91c09d478491f5ee : SUCCESS in 10m 45s","accounts_in_message":[],"_revision_number":6},{"id":"650a760b6123f398cf47b3a1710871302d019a35","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-30 14:58:50.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"751a7453f29fc0faba254971453fb367ce8705d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 18:24:52.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39d99448f1754f0e850a126d7197b808 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43ff80d6217d47c09ff1df2f8057e5d9 : SUCCESS in 7m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01d96468e8ed4f0c99d5f14fe92a061f : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ddc6c807ed44a2a64bdaf969aa7500 : SUCCESS in 7m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bb09382c3b914a28ba01f96a51d870c5 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4707f00d04fd47b2ac1bf96bf356d193 : FAILURE in 9m 19s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/210be0a39e364961b11af5122c067cb0 : SUCCESS in 42m 19s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bca4eff56884ff5a1a65a909cfa14b8 : SUCCESS in 46m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82d22437b25c44e6bffe013a9bc9f04f : SUCCESS in 46m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2c5f6ea3cebc4b1b9a34c79f79dce9e1 : SUCCESS in 43m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5ca5ed9332d9433cbdbb17f91eb0e233 : SUCCESS in 43m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/47fdbcbab2734220acc54c4e8a151a65 : SUCCESS in 28m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/20325adf41c4459f87de61d246439ee8 : SUCCESS in 28m 41s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f9e25c9fde4849aabcc252d7fd7b9c16 : TIMED_OUT in 3h 23m 46s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/dc674b75c84a4af9a12a93e340bfba09 : SUCCESS in 59m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4c6676050a84d47ba79ed70df366f28 : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":6},{"id":"95ac283f17c8bec005a5a6373e92d43b7acc9a83","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-30 18:29:08.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"7d42d486fd2bbf076b3a77c609fe4e6ac981399e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 21:09:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73fc8d4bcee457383a53987ec7bfc4c : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/198fd1f9ab14443bbe64661fb06a96b1 : SUCCESS in 8m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2fc6fa7db74e4680a0ce9bc9c09b5f7f : SUCCESS in 8m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d19ed455bdd4b8e8798effc3b46f0b3 : SUCCESS in 6m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8b7c67d9b71045e4a86c3a3d770f17f4 : SUCCESS in 3m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c80524fcac746f390cdbe000ab73a6d : SUCCESS in 9m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/03e5de666165401f8d33536ea6d2e54f : SUCCESS in 1h 09m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fb7d138da29409ba8f551fb7c49e7b9 : SUCCESS in 40m 00s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e246a6f31714c90b3b99971bb4c2edb : SUCCESS in 35m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/559b6e4521254541b63aa8720317e33c : SUCCESS in 47m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/18949e8b68b846f6800d8680fe7b7b54 : SUCCESS in 49m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5e65aa93d62845cebccca48abffd5471 : SUCCESS in 28m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4dda3b8eae474061b684b534a06d8a27 : SUCCESS in 37m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8206a363683b49f4b8d7cbccb6606630 : SUCCESS in 2h 33m 38s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d4bc57c62d9e41ee867f60a3c0178aed : SUCCESS in 57m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0da15e3b91454f6cb4766708248fe038 : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":6},{"id":"c90de4b15b2476ffdb408c05e78ffc4b9d977faa","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-31 07:11:35.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"13a29cb8a3b4b876ea4fdd2886e740a50ed45ba2","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-05 08:18:28.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"55dc0273ae59bdbdc43afff053a68455e9861ab8","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-05 08:24:47.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a131c69ec54860fb4a6776f7ec8fd2d2407bcaf4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-09 08:38:03.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f800db86de507c5aa9bd2450bf21569a49cc1c93","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-13 07:16:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"565096202fa2b78585fb1c2ad9ff5af78dacafd8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-13 07:17:45.000000000","message":"Patch Set 7: Workflow-1\n\nMaking it WIP until it\u0027s tested against a real Keycloak server.","accounts_in_message":[],"_revision_number":7},{"id":"0753e07ab8b32c5edc9dee8ceddc5077e6f9f5ab","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-13 07:19:40.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"75e317cf2ffc607946d9cb0a841c64bd153fb0f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 10:13:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74348a10a7e74096bfbeb0b76aeca77e : FAILURE in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98cc4b95068844bab15f5e582c892622 : SUCCESS in 8m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ff316c332e94c9099ef9ee6a845d3dd : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e867d5b6b7d4766bfdabceb5bfaca20 : SUCCESS in 8m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b387741f2874e98a65aa51505423151 : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f7a95db07664de285501e206ebc24f9 : SUCCESS in 8m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/834a8dfc0ebb43ce93bba1f58cd13916 : FAILURE in 11m 19s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3a8314be54a4beaa6c11030a640b17f : FAILURE in 8m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6083681798ad4e2eb7fe59256c5d96f5 : FAILURE in 11m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e96d63bce09e438ebb53172a2e9324fd : FAILURE in 12m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ae0453d3a86b4f6abc55cba3048986c9 : FAILURE in 8m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/309af3f5f0d3466487353573f7107d72 : FAILURE in 29m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/377f7a53d5f942f5967c2f2dad9490bd : FAILURE in 27m 20s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0fa81c15c5b243a4a36c90607cacdced : SUCCESS in 2h 50m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f05a8a25f2994695b371dae2e53c5815 : FAILURE in 11m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b1de83f279543109856cdd68e24a1fc : SUCCESS in 10m 07s","accounts_in_message":[],"_revision_number":7},{"id":"3b09cda082d818eb2c036942133b548c534c5885","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-14 06:34:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"068b0f02c43ad5551d30eb4255a269b0ae4bd80c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-14 06:59:48.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"711399d660003a331fac8f477bc08d7872e19b2a","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-14 07:26:25.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"36722e203bb0f8f4578c309c2956f02c8866cc54","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-14 08:51:21.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9cdcf17b6c169f076b4c6447b62afe5f6daa60cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 12:19:43.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea88c24950b64ec2a0cd70ea5fa11374 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5382951325154ac6abcf5861edce4685 : SUCCESS in 9m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec1ebf394d2e4cd683973bfc2306611c : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b08e9912c113485a87647c8971571a22 : SUCCESS in 6m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dec1fc08745e41409994ec82cf6d8bdd : SUCCESS in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fc04661908542ddb0102295a57795b4 : SUCCESS in 9m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/801a0caed54a47afac4d001c1f4d345e : SUCCESS in 50m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3cb8b7c608e7494195e7660bb5fad61b : SUCCESS in 42m 45s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e832de601294551a29c4095e80de92e : SUCCESS in 45m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/05f19921238b4b77be378fdd63293a72 : SUCCESS in 1h 02m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c5fc6f5d2e614458b2fd76aa61624eb7 : SUCCESS in 44m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/14d48c2f2bae44f19a82e65412275c1a : SUCCESS in 29m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/736cd4aa23594312b02e47e3a675480a : SUCCESS in 27m 14s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/dcad2bf3309b4a6091623a36467ee91a : SUCCESS in 2h 26m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4d2907d7a7244795a59c910eb1abba99 : SUCCESS in 1h 00m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7778339c33d04a368baaff1659273b39 : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":10},{"id":"857f31ad7ce2b189af41bb52784856472ba740b3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-14 12:20:28.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"2b737501f711ff40f02738fa918c99b276c96dc5","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-14 13:59:52.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"9644f432a220ea73b4d31ebb271f7cae2f7f2218","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-15 07:24:53.000000000","message":"Patch Set 10:\n\nThis is now tested against a real Keycloak but only w/o certificate verification in the outgoing HTTP requests since I only have Keycloak with a self-signed certificate (not signed by a Certificate Authority), so my machine always fails Keycloak certificate validation, if requested (insecure\u003dfalse in the configuration).","accounts_in_message":[],"_revision_number":10},{"id":"6c7bfef94a4bdae8ca15f7694f252fe73f96a33a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-15 08:12:08.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"860baa9adb3302f372087f9e8196fb3e5b39523f","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-15 09:53:02.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"5901b07ca9b938ed411930048018188f4bb7f319","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-01-15 10:53:43.000000000","message":"Patch Set 10: Code-Review+1\n\nLooks good to me but I don\u0027t have proper keycloak knowhow. so only +1-ing it","accounts_in_message":[],"_revision_number":10},{"id":"3062ed317b534fede505b3f2e5afe33b4061f58e","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-17 08:06:50.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"085d93920b7922514594915787f43b411ea66e72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 08:07:02.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"a7bfbac4b12d6c5e07d1d6e8fe05e82520e282c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 10:10:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"f273031c77e4f99f1e632df96d4398cc4fffa9f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 10:10:05.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9076749090ff4a5c91ae96dfb81dbc93 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b11abbf49894e77bb13ec65ca98de24 : SUCCESS in 9m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53993bafd8dc4cd09abd499368c9dad9 : SUCCESS in 9m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33e738f3cd564190a5aa08eac1547a41 : SUCCESS in 9m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/507f6c67367e4e46bbccc6eb01eecdc1 : SUCCESS in 3m 35s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e6718214a9474d24856ca01dc0cc1054 : SUCCESS in 40m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22480a29cb994a4bb48af8cc1ab74313 : SUCCESS in 38m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/704bb62083b44d68a22614e756208b0c : SUCCESS in 48m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/84024067d42a45a4b8d31d8d5de7bbca : SUCCESS in 37m 17s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e889db8e68e84115916daf007049d9b5 : SUCCESS in 31m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ae2f941a5d454d9b930b196c6f156e57 : SUCCESS in 30m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/74ae5c08c3744e67a78a52b3ce1d506c : SUCCESS in 52m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a913001b693461eb21b5aff6515abdc : SUCCESS in 10m 27s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0b9e306ca3514c5cbfe128c31f5a053d : SUCCESS in 2h 02m 38s","accounts_in_message":[],"_revision_number":10},{"id":"0d1bba4f6fd209bcad871ae0004d8c7068947d05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 10:13:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0046cd88dd35430a9681765e87ac8e1c : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":10},{"id":"bbdfe4a11065427844bd725894b7e314d727b16a","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-01-20 17:20:01.000000000","message":"Created a revert of this change as I5a8977d27f779b88cc29afb78e92d95fd51d9202","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"c888a46ccc46346deee8d483a706918341bc0a7f","revisions":{"4564e95c0f687eaf0d15737af414813526daf6d2":{"kind":"REWORK","_number":1,"created":"2019-12-27 09:21:43.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/95/700695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/700695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/700695/1"}}},"commit":{"parents":[{"commit":"541aabbfe8a7afd876adcfb209d18aafdef2d5f6","subject":"Merge \"Use MISTRAL_SERVICE_HOST as the host ip for standalone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/541aabbfe8a7afd876adcfb209d18aafdef2d5f6"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 09:20:19.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 09:20:19.000000000","tz":420},"subject":"WIP: fix keycloak authentication","message":"WIP: fix keycloak authentication\n\n* Implement offline access token validation using Keycloak public key.\n\nChange-Id: I0eecec4b4e64381cac005622b16c6d9e4bed4df6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4564e95c0f687eaf0d15737af414813526daf6d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4564e95c0f687eaf0d15737af414813526daf6d2"}]},"branch":"refs/heads/master"},"bf9818d06094fb0f19381b86df5ddcd68305f4f0":{"kind":"REWORK","_number":2,"created":"2019-12-27 09:35:44.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/95/700695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/700695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/700695/2"}}},"commit":{"parents":[{"commit":"541aabbfe8a7afd876adcfb209d18aafdef2d5f6","subject":"Merge \"Use MISTRAL_SERVICE_HOST as the host ip for standalone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/541aabbfe8a7afd876adcfb209d18aafdef2d5f6"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 09:20:19.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 09:35:29.000000000","tz":420},"subject":"WIP: fix keycloak authentication","message":"WIP: fix keycloak authentication\n\n* Implement offline access token validation using Keycloak public key.\n\nChange-Id: I0eecec4b4e64381cac005622b16c6d9e4bed4df6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bf9818d06094fb0f19381b86df5ddcd68305f4f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bf9818d06094fb0f19381b86df5ddcd68305f4f0"}]},"branch":"refs/heads/master"},"47d6f62d924f83ce0850bc7eabb64073e77b9bc5":{"kind":"REWORK","_number":3,"created":"2019-12-27 11:16:51.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/95/700695/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/700695/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/700695/3"}}},"commit":{"parents":[{"commit":"541aabbfe8a7afd876adcfb209d18aafdef2d5f6","subject":"Merge \"Use MISTRAL_SERVICE_HOST as the host ip for standalone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/541aabbfe8a7afd876adcfb209d18aafdef2d5f6"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 09:20:19.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 11:16:33.000000000","tz":420},"subject":"WIP: fix keycloak authentication","message":"WIP: fix keycloak authentication\n\n* Implement offline access token validation using Keycloak public key.\n\nChange-Id: I0eecec4b4e64381cac005622b16c6d9e4bed4df6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/47d6f62d924f83ce0850bc7eabb64073e77b9bc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/47d6f62d924f83ce0850bc7eabb64073e77b9bc5"}]},"branch":"refs/heads/master"},"a850841095db1dca26e1d7bc02c7288475f0c836":{"kind":"REWORK","_number":4,"created":"2019-12-30 05:18:30.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/95/700695/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/700695/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/700695/4"}}},"commit":{"parents":[{"commit":"541aabbfe8a7afd876adcfb209d18aafdef2d5f6","subject":"Merge \"Use MISTRAL_SERVICE_HOST as the host ip for standalone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/541aabbfe8a7afd876adcfb209d18aafdef2d5f6"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 09:20:19.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-30 05:18:03.000000000","tz":420},"subject":"Fix keycloak authentication","message":"Fix keycloak authentication\n\n* Implement offline access token validation using Keycloak public key.\n\nChange-Id: I0eecec4b4e64381cac005622b16c6d9e4bed4df6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a850841095db1dca26e1d7bc02c7288475f0c836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a850841095db1dca26e1d7bc02c7288475f0c836"}]},"branch":"refs/heads/master"},"de4ea6b7058f61898cf4a4aac1ec28cd2f8489fb":{"kind":"REWORK","_number":5,"created":"2019-12-30 05:28:42.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/95/700695/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/700695/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/700695/5"}}},"commit":{"parents":[{"commit":"541aabbfe8a7afd876adcfb209d18aafdef2d5f6","subject":"Merge \"Use MISTRAL_SERVICE_HOST as the host ip for standalone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/541aabbfe8a7afd876adcfb209d18aafdef2d5f6"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 09:20:19.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-30 05:28:13.000000000","tz":420},"subject":"Fix keycloak authentication","message":"Fix keycloak authentication\n\n* Implement offline access token validation using Keycloak public key.\n\nCloses-bug: #1857871\nChange-Id: I0eecec4b4e64381cac005622b16c6d9e4bed4df6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/de4ea6b7058f61898cf4a4aac1ec28cd2f8489fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/de4ea6b7058f61898cf4a4aac1ec28cd2f8489fb"}]},"branch":"refs/heads/master"},"df1090bf7435422a390ea40d2d6bda6f8efc3702":{"kind":"REWORK","_number":6,"created":"2019-12-30 07:03:08.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/95/700695/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/700695/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/700695/6"}}},"commit":{"parents":[{"commit":"541aabbfe8a7afd876adcfb209d18aafdef2d5f6","subject":"Merge \"Use MISTRAL_SERVICE_HOST as the host ip for standalone\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/541aabbfe8a7afd876adcfb209d18aafdef2d5f6"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 09:20:19.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-30 07:02:53.000000000","tz":420},"subject":"Fix keycloak authentication","message":"Fix keycloak authentication\n\n* Implement offline access token validation using Keycloak public key.\n\nCloses-bug: #1857871\nChange-Id: I0eecec4b4e64381cac005622b16c6d9e4bed4df6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/df1090bf7435422a390ea40d2d6bda6f8efc3702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/df1090bf7435422a390ea40d2d6bda6f8efc3702"}]},"branch":"refs/heads/master"},"f7fcac6eafe4b507d7223e8496e2024e5f6841e4":{"kind":"REWORK","_number":7,"created":"2020-01-13 07:16:54.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/95/700695/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/700695/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/700695/7"}}},"commit":{"parents":[{"commit":"fef538cce793518f4e7bfe645f0ccbd6dc55ad0a","subject":"Merge \"Add the 1st version of Mistral coding guidelines into the docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fef538cce793518f4e7bfe645f0ccbd6dc55ad0a"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 09:20:19.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-13 07:16:39.000000000","tz":420},"subject":"Fix keycloak authentication","message":"Fix keycloak authentication\n\n* Implement offline access token validation using Keycloak public key.\n\nCloses-bug: #1857871\nChange-Id: I0eecec4b4e64381cac005622b16c6d9e4bed4df6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f7fcac6eafe4b507d7223e8496e2024e5f6841e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f7fcac6eafe4b507d7223e8496e2024e5f6841e4"}]},"branch":"refs/heads/master"},"37cd3db07cf1a4eedd1a909110eda1072d204977":{"kind":"REWORK","_number":8,"created":"2020-01-14 06:34:50.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/95/700695/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/700695/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/700695/8"}}},"commit":{"parents":[{"commit":"fef538cce793518f4e7bfe645f0ccbd6dc55ad0a","subject":"Merge \"Add the 1st version of Mistral coding guidelines into the docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fef538cce793518f4e7bfe645f0ccbd6dc55ad0a"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 09:20:19.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-14 06:34:13.000000000","tz":420},"subject":"Fix keycloak authentication","message":"Fix keycloak authentication\n\n* Implement offline access token validation using Keycloak public key.\n\nCloses-bug: #1857871\nChange-Id: I0eecec4b4e64381cac005622b16c6d9e4bed4df6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/37cd3db07cf1a4eedd1a909110eda1072d204977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/37cd3db07cf1a4eedd1a909110eda1072d204977"}]},"branch":"refs/heads/master"},"33dd3d121acfad607ee15bd72217481b18b33653":{"kind":"REWORK","_number":9,"created":"2020-01-14 06:59:48.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/95/700695/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/700695/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/700695/9"}}},"commit":{"parents":[{"commit":"fef538cce793518f4e7bfe645f0ccbd6dc55ad0a","subject":"Merge \"Add the 1st version of Mistral coding guidelines into the docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fef538cce793518f4e7bfe645f0ccbd6dc55ad0a"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 09:20:19.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-14 06:59:30.000000000","tz":420},"subject":"Fix keycloak authentication","message":"Fix keycloak authentication\n\n* Implement offline access token validation using Keycloak public key.\n\nCloses-bug: #1857871\nChange-Id: I0eecec4b4e64381cac005622b16c6d9e4bed4df6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33dd3d121acfad607ee15bd72217481b18b33653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/33dd3d121acfad607ee15bd72217481b18b33653"}]},"branch":"refs/heads/master"},"c888a46ccc46346deee8d483a706918341bc0a7f":{"kind":"REWORK","_number":10,"created":"2020-01-14 08:51:21.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/95/700695/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/95/700695/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/95/700695/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/95/700695/10"}}},"commit":{"parents":[{"commit":"fef538cce793518f4e7bfe645f0ccbd6dc55ad0a","subject":"Merge \"Add the 1st version of Mistral coding guidelines into the docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fef538cce793518f4e7bfe645f0ccbd6dc55ad0a"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-27 09:20:19.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-14 08:51:06.000000000","tz":420},"subject":"Fix keycloak authentication","message":"Fix keycloak authentication\n\n* Implement offline access token validation using Keycloak public key.\n\nCloses-bug: #1857871\nChange-Id: I0eecec4b4e64381cac005622b16c6d9e4bed4df6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c888a46ccc46346deee8d483a706918341bc0a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c888a46ccc46346deee8d483a706918341bc0a7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
