)]}'
{"id":"openstack%2Fmistral~700809","triplet_id":"openstack%2Fmistral~master~Ie3052d8cdc047f8a46e139f533a63f8c36d2906d","project":"openstack/mistral","branch":"master","topic":"bug/1808953","hashtags":[],"change_id":"Ie3052d8cdc047f8a46e139f533a63f8c36d2906d","subject":"Don\u0027t use eventlet.monkey_patch under wsgi","status":"MERGED","created":"2019-12-31 13:14:27.000000000","updated":"2020-01-09 12:56:52.000000000","submitted":"2020-01-09 12:54:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700809-1578574491724-1b68ee4e","meta_rev_id":"94cddde9b2114619fa1e88394080cb04ea7c6465","_number":700809,"virtual_id_number":700809,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 12:54:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-09 08:07:15.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-01-09 06:50:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-01-09 06:50:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-09 06:50:06.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-01-09 12:54:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"50866c763b6c18485b910b33f684c42e761f3842","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-31 13:14:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bef814315a05d304457c565bcdcfe129ca78f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 16:07:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a129745c98174fd2a6d3d29b0f0e4522 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ea1d7eb4546466f9070fd8e9f3ecb8e : SUCCESS in 8m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/476666b9ad2f4c7384fc9e465b08e9a1 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c62033c1edc42cb9ccfd79bc3b5d57c : SUCCESS in 7m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32d0e422427144f0bdba8b0b4e6f8d87 : SUCCESS in 8m 42s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/de77431552034aa6a73773d7647ce220 : FAILURE in 20m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc4f2c00b6f4436a9aebfd1882815af0 : SUCCESS in 35m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6361bb4776994d93b85bf25e0a35fb68 : SUCCESS in 37m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6ea08f8c5806428cb578b0602edc680b : FAILURE in 22m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/387f0e81e94345a29e16b4e773625dc4 : FAILURE in 25m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6c926ec3e0204443a522fa63e7e52cb8 : SUCCESS in 28m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/eb55e59b36c540b7ac2b3e16f153bb18 : SUCCESS in 26m 43s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2c46bdfc1ff643b3936fdea41143a685 : SUCCESS in 2h 48m 26s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b83347c4147d45328477935fc0d9a213 : SUCCESS in 59m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae733cb91f48432283aef4b31e362d47 : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":1},{"id":"df232712205161bba00216aba21490b21a39a120","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-12-31 16:09:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c28c1f8cb3a43a3e31e5b74143b753985aa6fd15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 19:15:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44c5d8045c024522b8f89418c3d6d98a : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91d9a79ec20e42b59aa971c4a0ead799 : SUCCESS in 8m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b91ae5fd6e2a42deb2d38fcc4458afaf : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30ec9b4efa4249118543af91c3fac377 : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c443226b70384e628f6da63bee431b29 : SUCCESS in 8m 59s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/33b1d73dc1fe48a0abe84ea8f7356bf5 : SUCCESS in 42m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3965ba7ff5db40688681d51ae5cd8bd5 : SUCCESS in 42m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6750f5da3a24430b8b71beb1cdc4fc06 : SUCCESS in 42m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/79759373b03846538917527b89c76f49 : SUCCESS in 41m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a476f846c3e445dbaf3002c77b09ace0 : SUCCESS in 42m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/80aa0b601fd147b2ae1dd55c69a74385 : SUCCESS in 28m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9d5f9e18bec6421d97e4b9fed58090ea : SUCCESS in 26m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/224ecdbeb30149a99b7dbd31110e9b08 : SUCCESS in 3h 01m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/633ed47826674b03a580c3842b7fa554 : SUCCESS in 55m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b03229205d2e46b99cab7ee3f24d345d : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":1},{"id":"732663a5edfb838c55b03433d0db092330453433","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-09 06:50:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f49fe2fcded182ba7e20f2ac609bb4c7dffebccc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 06:50:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a34d7128ebe24f62840510896fab8196eb78297b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 08:02:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c33f0e44468b4f939fd44d15d49458b2 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11e61728fb7b4de093db2a2c30125937 : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58f2c6f831ca439ba17ca93acff44e0e : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e4da033e4d14de2a6935eb188d1fa63 : SUCCESS in 11m 55s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b219ebddefb94da2a8547610f0ec9a78 : SUCCESS in 43m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a6a6d07264f480a8975f7b9d542fba3 : SUCCESS in 42m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cf452785e144a60add0ced26671f3a7 : SUCCESS in 48m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0391037136e349f7a127385e374f8830 : SUCCESS in 49m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/dc1accd600eb4326a68a13e2a2129ed8 : FAILURE in 29m 07s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/12d9827b617346aa8d136aaa3209f8b5 : SUCCESS in 31m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d1a0daec630e473ba7c01d081c4fffce : SUCCESS in 50m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56687b440e2946aebdc5389048c51285 : SUCCESS in 10m 35s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/85834d8426b24fadad97d7ebadf45e6f : FAILURE in 52m 38s","accounts_in_message":[],"_revision_number":1},{"id":"af9157049a002621b816cff3c4c727ca926939b2","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-09 08:07:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"08b4036c8eb865b7ab6d34a60169bcf6bee4936b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 11:09:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0f5974a459e47b294003d90b0e5d511 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d9d3d5808574a9fbe795bed0332428e : SUCCESS in 11m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1283c185c60466a9f5f82e425fc75d4 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0734e70c002b48108c1dffcb9882c262 : SUCCESS in 8m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b6258c733f44da98a7f6fe35ed24733 : SUCCESS in 10m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/438a16b868bb4e2d85340669bd38c88a : SUCCESS in 50m 02s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b137557f652457bbc3b75a844bc3a52 : SUCCESS in 54m 27s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12d5f8e39f394d0fb148e60c3b5725f8 : SUCCESS in 1h 00m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6b616788fb3246b1a3605420bcbdbe70 : SUCCESS in 1h 00m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/376e69cbe8a54c1f826515843fd9f961 : SUCCESS in 56m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ce1678e0227843b08a946363f3d6d2f0 : SUCCESS in 33m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a24e5a7524204bf7b9f2fe1f0837b7eb : SUCCESS in 31m 39s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a2f39322fadb4a67b9d13f6555ccc6fb : SUCCESS in 2h 20m 36s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/785ce5d7bf6c4418b7da10fb4aa82200 : SUCCESS in 1h 22m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1845e54b48cd49dd877515e2d3374ecd : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f4e8ca2a4156340dfb79834d5ec0a73bfdc6bf7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 11:09:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7dac12fdae7919b789ac1d666313ca5e5b193153","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 12:54:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6e8bb39794aedf4d659b7093bb57439f913555f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 12:54:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/455cbdf70fea4ae3a307ec5bb4b3744b : SUCCESS in 9m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e395ce24fe949b4a535f18353012601 : SUCCESS in 12m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b036ac3cf4241ae884ce31462e71fdb : SUCCESS in 12m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef0d7b320fe74245aa6b476caa56853c : SUCCESS in 11m 50s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e92655810f1e4e67ac878fc30d29770a : SUCCESS in 55m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ce02b6b279b40249281181ae8667c6f : SUCCESS in 49m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/378a5319bfb34ad28ab3a2a28ab54795 : SUCCESS in 1h 03m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/32b10fd51578494ca0aaa142a5151726 : SUCCESS in 42m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4db8e28b994945a9afcc6b72b2d849f6 : SUCCESS in 28m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ec84ca3f74234658b914c7c0eac4543d : SUCCESS in 28m 08s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ba4b18e6a47a41e9ad220aaf36ad4d70 : SUCCESS in 44m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7889707925034ebf9c1e5906a07d79d2 : SUCCESS in 11m 26s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/1713c196399d4f8185ef5f150d6d8fc8 : SUCCESS in 1h 40m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f37e190972e6b7d377a3cf278f62ffab23831926","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 12:56:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d145afd09d348b99c1fd13a9a9fe358 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1981e63b3f63ac463bed52b7b2deea4aae2b53b0","revisions":{"1981e63b3f63ac463bed52b7b2deea4aae2b53b0":{"kind":"REWORK","_number":1,"created":"2019-12-31 13:14:27.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/09/700809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/09/700809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/09/700809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/09/700809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/09/700809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/09/700809/1"}}},"commit":{"parents":[{"commit":"6e9a70db25ddbce2ce7ec4d6fd057cec5d988417","subject":"Merge \"[train][goal] Run \u0027mistral-devstack-tempest-ipv6-only\u0027 job in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e9a70db25ddbce2ce7ec4d6fd057cec5d988417"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-31 13:14:10.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-12-31 13:14:10.000000000","tz":120},"subject":"Don\u0027t use eventlet.monkey_patch under wsgi","message":"Don\u0027t use eventlet.monkey_patch under wsgi\n\nThis causes errors and is not recommended.\nIt also solves the problem of coordination not\nworking properly under wsgi\n\nChange-Id: Ie3052d8cdc047f8a46e139f533a63f8c36d2906d\nCloses-Bug: 18008953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1981e63b3f63ac463bed52b7b2deea4aae2b53b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1981e63b3f63ac463bed52b7b2deea4aae2b53b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
