)]}'
{"id":"openstack%2Fmistral~701164","triplet_id":"openstack%2Fmistral~master~I4cdfc1abf2634a3a922377d7ceda0585ed0c4061","project":"openstack/mistral","branch":"master","topic":"bug/1820349","hashtags":[],"change_id":"I4cdfc1abf2634a3a922377d7ceda0585ed0c4061","subject":"Use 406 error instead of 500 error when Service API is not supported","status":"MERGED","created":"2020-01-06 08:23:44.000000000","updated":"2020-01-09 10:22:04.000000000","submitted":"2020-01-09 10:20:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701164-1578565209769-c94174cc","meta_rev_id":"4575f09d433bde9c1b4bfab610188d0f2b97af9f","_number":701164,"virtual_id_number":701164,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 10:20:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2020-01-06 08:26:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-01-09 07:45:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2020-01-09 07:45:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 08:26:42.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-01-09 07:45:49.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-01-09 10:20:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c31047ab92c1f5f652c7fabadc2d093d85bc74e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-06 08:23:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1dc7be131b02b93609775a61dc46cb58686f5192","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-06 08:24:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f2dcd3e3a72c926cd78b95749078da3aecc9d61a","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-06 08:26:42.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6227a3a529567663bb847b17cffbafe15313465c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 11:03:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abd35c93394547a3b484ac26aca1129d : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24135fc41ca04db580002fda11451faf : SUCCESS in 8m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e8834cc8a4e4284805586cbad621c8e : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0611f30324f34bbcb92439a9764ec1d1 : SUCCESS in 6m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78b2d289ec0847eba9c8ca34618cc73e : SUCCESS in 8m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/49fc0210e9904ab3887386a96a71f4ba : SUCCESS in 42m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1f626bde3fe4f86b67ae78390cb3e5b : SUCCESS in 42m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cd37a50d70e4cfaa66943d5e7da937e : SUCCESS in 42m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c117c1eac1b44aaba6664c0664b74fb8 : SUCCESS in 42m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/30e5a355d3964e8a88dbf03ffcf34cf4 : SUCCESS in 41m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1eeb34aff2ea4c1ca0ac027b9583258f : SUCCESS in 28m 22s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a4affe3809fe49488bd27ffd92b3be6e : SUCCESS in 28m 31s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a3ff04a0a09544a89d03cf9e748bf4ea : SUCCESS in 2h 37m 28s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ff47ccb5e57f4ea8941559cf9b2a3bd3 : SUCCESS in 59m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c947a3c9960478ba73fe726b4f728c8 : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":2},{"id":"9402f49113c7588ea3883f60f202b46f8ec782e5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-09 07:45:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b526fa915c4d5a0dbe5e04e3649f9ddfe0b83697","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 07:46:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"541ded4e2f9ee293872075d777da8b3f015a98e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 10:20:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bb7c9253992d3ac972d1e90dca7d4a0e9b3f95c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 10:20:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/757c8b8a9df84664ae75d9a4c13fe086 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfb720ea6d4b493b940334f389d0739d : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34a1fd4f943c4351bffb0e4028d01d67 : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1448ea255a2429eb18054a03cbfb554 : SUCCESS in 9m 21s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9759373e1394496188f0b9ede297baad : SUCCESS in 55m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/508fb4042b4748d0b1e2e61b3de15486 : SUCCESS in 46m 09s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/717b926c354b42908d1445a0bddde945 : SUCCESS in 47m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ba1bcc941b0f4ba295b045c265e9937f : SUCCESS in 44m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/799bd0f7d2e94382aa6eb3756b2bde9e : SUCCESS in 29m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/20578ec2140e4412be104f3f70a61b12 : SUCCESS in 29m 36s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a0cc9c3b4dec46519c7d7878388161ff : SUCCESS in 45m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/703022eb1bb9459698b5ff64c6f4153a : SUCCESS in 10m 48s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c412771c028b43378deddf9f7950ef83 : SUCCESS in 2h 01m 57s","accounts_in_message":[],"_revision_number":2},{"id":"b2589ba2be7bf51bb4f3e4612ca4c6efe99e1449","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 10:22:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b62d55a45fb34e3bafb4b5dbd42693c4 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb94ca1fede74c84bd7192f19a7f8497f0756432","revisions":{"e4e152dc29a26e553c1198d64a74ca8f42c275d0":{"kind":"REWORK","_number":1,"created":"2020-01-06 08:23:44.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/64/701164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/64/701164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/701164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/701164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/701164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/64/701164/1"}}},"commit":{"parents":[{"commit":"6e9a70db25ddbce2ce7ec4d6fd057cec5d988417","subject":"Merge \"[train][goal] Run \u0027mistral-devstack-tempest-ipv6-only\u0027 job in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e9a70db25ddbce2ce7ec4d6fd057cec5d988417"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-06 08:18:08.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-06 08:18:08.000000000","tz":120},"subject":"Use 406 error instead of 500 error when Service API is not supported","message":"Use 406 error instead of 500 error when Service API is not supported\n\nChange-Id: I4cdfc1abf2634a3a922377d7ceda0585ed0c4061\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4e152dc29a26e553c1198d64a74ca8f42c275d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4e152dc29a26e553c1198d64a74ca8f42c275d0"}]},"branch":"refs/heads/master"},"bb94ca1fede74c84bd7192f19a7f8497f0756432":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-06 08:24:39.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/64/701164/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/64/701164/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/701164/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/701164/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/64/701164/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/64/701164/2"}}},"commit":{"parents":[{"commit":"6e9a70db25ddbce2ce7ec4d6fd057cec5d988417","subject":"Merge \"[train][goal] Run \u0027mistral-devstack-tempest-ipv6-only\u0027 job in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e9a70db25ddbce2ce7ec4d6fd057cec5d988417"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-06 08:18:08.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-06 08:24:05.000000000","tz":120},"subject":"Use 406 error instead of 500 error when Service API is not supported","message":"Use 406 error instead of 500 error when Service API is not supported\n\nCloses-Bug: 1820349\nChange-Id: I4cdfc1abf2634a3a922377d7ceda0585ed0c4061\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bb94ca1fede74c84bd7192f19a7f8497f0756432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bb94ca1fede74c84bd7192f19a7f8497f0756432"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
