)]}'
{"id":"openstack%2Fmistral~701167","triplet_id":"openstack%2Fmistral~master~I55c96c1eea495d8dfb7e87d8f8498fce5dbbd068","project":"openstack/mistral","branch":"master","topic":"duplicated","hashtags":[],"change_id":"I55c96c1eea495d8dfb7e87d8f8498fce5dbbd068","subject":"Fix duplicated words issue like \"from from parsed token\"","status":"MERGED","created":"2020-01-06 08:30:39.000000000","updated":"2020-01-09 09:43:16.000000000","submitted":"2020-01-09 09:43:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701167-1578562996798-d3fc192b","meta_rev_id":"9e40f3b2cc558a2ca83fb5788d44855c19bd855e","_number":701167,"virtual_id_number":701167,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 09:43:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2020-01-06 18:58:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-01-09 06:54:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-06 08:31:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2020-01-09 06:54:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 08:31:30.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-01-06 18:58:26.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-01-09 06:54:48.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-01-09 09:43:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3e0773d14c9f33b94b1f28fd377455ae0044bdf","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-01-06 08:30:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4ffea4d846be1008cde63bd90e6125ace02070e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-06 08:31:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"acefff3bc6fe177f85cb4b76fccd3619d1adfeea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 11:08:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fba0a2df3eb24a23a99b45aae0e29471 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf4f8d2b8c1a4dbb86b1e009a72882e9 : SUCCESS in 10m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c174e6fd29449038d87804b97755ece : SUCCESS in 9m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e521f2b4a1fc4683afc18a6dd3073273 : SUCCESS in 11m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bda384ad52944fb8f7263c2e8ebd27c : SUCCESS in 9m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fd412fa5de054d8a9bbf013902527e4b : SUCCESS in 1h 03m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dc0e14857364ff89d943f8297605942 : SUCCESS in 51m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/179d4f1a7e5d4139b2d263594ca6fbc6 : SUCCESS in 1h 03m 09s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d2340b47721d4f61897d98068f09790b : SUCCESS in 41m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3923d2e2f68d485fa8f95836b602c0fa : SUCCESS in 53m 05s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ef4f7d0c47de4bc595b57f39116e2b71 : SUCCESS in 28m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dd260d49d10043e7b1c4bc39f3120032 : SUCCESS in 30m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2dfbbb12fae64fa88a3c18f4fd19083a : SUCCESS in 2h 36m 51s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1c20af9783fd42d0834a429625851e4c : SUCCESS in 1h 11m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c08dacc9085e42ebbbce08119287a7fa : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":1},{"id":"550237dc342d8eda8a6905e7034adb55a3e8ce93","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-06 18:58:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2591b5fb72669b0f65865ffaa0da5dddbf374a73","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-09 06:54:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b561a1fc15d58cb7b486b82b91211cae04f1e0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 06:55:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cc6206da9ec395cb777e882472b95a29a29e117","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 09:43:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"08566e67168a124b79d9bcd923d1fad433f6a9b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 09:43:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62be73a1b3324404aaadef40bde56002 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a09d5fd064da40a89b3fffa7810251e7 : SUCCESS in 9m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ec9b763517c4aff94dbb6adbfe1bba1 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/499db17400e24d6087f356d407755357 : SUCCESS in 9m 19s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/07988a18aabf4445bb68ae8715c7dab2 : SUCCESS in 47m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df65d4fcce1345e992d41fdef12ce6c3 : SUCCESS in 48m 48s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5dd1cc5792b5486ca6162b8974a96c1c : SUCCESS in 49m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a0f89ef9cc46489f901eba973cee15fd : SUCCESS in 50m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bc988b810ea74d6d931fb6720b7289f8 : SUCCESS in 30m 19s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fe3e6d20e5b44260865b74f7afd48fa2 : SUCCESS in 31m 50s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a80bda372d4b41a48c8632237afc2658 : SUCCESS in 43m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7250f467ae2423fad8db4ea3f56ceb8 : SUCCESS in 11m 59s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/42e1ed5c04814aef8bf1c58e3ed9e9b4 : SUCCESS in 1h 45m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c6771954f21db05bff1c04a8a83b6c0d491e52a","revisions":{"6c6771954f21db05bff1c04a8a83b6c0d491e52a":{"kind":"REWORK","_number":1,"created":"2020-01-06 08:30:39.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/67/701167/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/67/701167/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/701167/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/701167/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/67/701167/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/67/701167/1"}}},"commit":{"parents":[{"commit":"6e9a70db25ddbce2ce7ec4d6fd057cec5d988417","subject":"Merge \"[train][goal] Run \u0027mistral-devstack-tempest-ipv6-only\u0027 job in gate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e9a70db25ddbce2ce7ec4d6fd057cec5d988417"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-01-06 06:06:46.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2020-01-06 06:06:46.000000000","tz":480},"subject":"Fix duplicated words issue like \"from from parsed token\"","message":"Fix duplicated words issue like \"from from parsed token\"\n\nThis is to fix the duplicated words issue like\n\"Get roles from from parsed token\".\n\nChange-Id: I55c96c1eea495d8dfb7e87d8f8498fce5dbbd068\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6c6771954f21db05bff1c04a8a83b6c0d491e52a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6c6771954f21db05bff1c04a8a83b6c0d491e52a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
