)]}'
{"id":"openstack%2Fmistral~702991","triplet_id":"openstack%2Fmistral~master~Id866ed13764b1d4db75cf1a819b53a7e8955b34a","project":"openstack/mistral","branch":"master","topic":"httpaction_json_param","hashtags":[],"change_id":"Id866ed13764b1d4db75cf1a819b53a7e8955b34a","subject":"Add json param to HTTPAction","status":"MERGED","created":"2020-01-16 21:40:54.000000000","updated":"2020-01-23 19:11:01.000000000","submitted":"2020-01-23 19:09:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702991-1579806547383-9ea6cd4b","meta_rev_id":"88b2c2b91e2081f478e66115a58c0130f62cd460","_number":702991,"virtual_id_number":702991,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-23 14:06:26.000000000","_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-23 19:09:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2020-01-23 08:33:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-23 08:38:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-23 08:33:02.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-01-23 08:38:01.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-01-23 19:09:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66a84b91a0e60954673eea4649162b4909b58f3e","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-16 21:40:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e42ba7bdaa17e739a13036816c25dc5ef8ab1619","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-16 21:43:36.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"98d0158f46f88a2553c2119e359450f993b72c3e","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-16 21:44:32.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9a0b3d6b633a8ddb5c6ea057a77f8b2863c422d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 00:25:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ecafd29b8b043268f5f175456638658 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/074e68166af741d990d55c35df55975b : FAILURE in 9m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6fe57e553d54da082af70d33a010c50 : FAILURE in 8m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e24a2316ca4b07bbc5056ee5951fd9 : SUCCESS in 7m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd0d6283d33647d0a1db88c554e04aaa : FAILURE in 12m 36s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ecf7390e5d5a4675b4f1bc910aed1998 : SUCCESS in 41m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79778cda99c44f7e96da1befd9f10b36 : SUCCESS in 50m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4470bbcb5f54371b81a12990ae24e6c : SUCCESS in 45m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fabc9b65994c49b9800a6b74613bad05 : SUCCESS in 45m 47s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/32e86a88683641dbb51ddeb262c0021c : SUCCESS in 44m 06s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e02b4713dde34c2da2f58b658780a983 : FAILURE in 29m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/085c07f945074f83b8c4a556b6696e23 : FAILURE in 28m 29s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4811cac9002348dbb800184fe4bafcbc : SUCCESS in 2h 39m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b7c6048274094c7e977716ba4c227f7e : SUCCESS in 56m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8625839dfd194da4bdb4b1de20ceb018 : FAILURE in 10m 57s","accounts_in_message":[],"_revision_number":3},{"id":"b776e98c8cac35f21a1efb821fc968172e433f12","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-17 08:23:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f1b785cbe1aa6dbe3e9433380eee390d4e6583fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 12:24:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13065b06c1484b839d7a0fe1ac950f07 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0a2f20fa8884569b2685594ee7668bb : SUCCESS in 11m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5f91ac4387a41d48d9ba5abf02d476f : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6179642d360424188d3d4b7374bdde5 : SUCCESS in 9m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f579242992d44c9eb4911a5ac490907e : SUCCESS in 10m 03s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/45e326811afb4f56a89b5a96856ff4bf : SUCCESS in 45m 05s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13082fc72f85474f8ec7dd5b2ad19087 : SUCCESS in 47m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25b420d9176d430ebec1a937e792510f : SUCCESS in 46m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9210a5c44bf546e891eaf712c0ca023a : SUCCESS in 46m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4027907c58ba4a758f90decf9c2e53f5 : SUCCESS in 54m 47s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e9db95ea1a3748019be0c580155c1f75 : SUCCESS in 31m 51s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2b86683b93444a248c0706f25b533af7 : SUCCESS in 29m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/669a951b4c9f4ca5adf24a8662ea2e43 : FAILURE in 1h 41m 12s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/923cac8dc7ad4aa5905b48ced68ebc72 : SUCCESS in 1h 07m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/369369041f654543a41abdff8acaa26f : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":4},{"id":"a58e34a77d70bdd97cb0a77192f570e572230398","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-19 11:37:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4d5ef113020c5835ce833e74fececc80adfe6fdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 14:35:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3451e00c8de841b3944c40c2fa5218cf : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8ce6b2e28be4dc188271cb015be4d2e : SUCCESS in 8m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf9721f7b3b846e88fe01ae999e0f67a : SUCCESS in 8m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70ed24f445694a4d9da8ac1ea216a5e7 : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33f35a27b2fb4fc49bad4acbeb52e6ca : SUCCESS in 8m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/82b98ec1fb204a4a9da3d70ed07d47e8 : SUCCESS in 44m 45s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2115dac5c09a4d6f87d984b0d99e31e9 : SUCCESS in 45m 31s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78fae54ba7f343aea11c6a4d1278c8e3 : SUCCESS in 42m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c9cd0f38e5074bcd9b63b8375aa6233d : SUCCESS in 44m 39s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/073fdd1d5ec1499fb28067a49ac2de1f : SUCCESS in 42m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bf6f75b60a1e4a6fbf8c2e4d9b884435 : SUCCESS in 30m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9074d5d24f8447679aef359c39c4f60b : SUCCESS in 30m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/487705c3c0da419d94d6ddac8cd4b58a : SUCCESS in 2h 53m 41s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ba35db91579f4efbb3b3d24fda33082c : SUCCESS in 59m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/871520f22a9349ceb7195fc58a7257fe : SUCCESS in 11m 11s","accounts_in_message":[],"_revision_number":4},{"id":"e52e3acf7392a40fa820688de0d28e8e5d53c148","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-20 21:45:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b799001eaa2d1aa1f29019a448ad685d57f2f0a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 00:47:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b5eab161d004112873e1b560d4bd817 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9053a105e61432d8c1a00f81d714e35 : SUCCESS in 9m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2f071f44ceb45fc98fe0c6433421c18 : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6a93c1c6fd24af2ab5ac336687f083e : SUCCESS in 8m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ee61b6b23de4f4aa009888739b4c1af : SUCCESS in 9m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/38674c0c48ed4cf5b4b58269025fb331 : SUCCESS in 43m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a9070ef7df84cddb4825fc62f9ed43d : SUCCESS in 43m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9580bc962c69487fac3177b765eaa824 : SUCCESS in 45m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/570be2f186754fc99f898e8b3c47a372 : SUCCESS in 41m 50s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b0e9f28fb7144094a4981681e83ff9b7 : FAILURE in 46m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/034b109d1ba44514b5c35523f276a207 : SUCCESS in 29m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/94711b5279184b6cbec87449fe06c684 : SUCCESS in 28m 14s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/586146bfb3a144bf9b4ee3babe70fa15 : SUCCESS in 2h 56m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d22511af58084746bc85f6ad0fd2ae30 : SUCCESS in 1h 10m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eea8792ff33c4803bde21a3233298780 : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":5},{"id":"f804ae1b1b1e2219a2073c4f945ceb0cd9281fc6","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-21 06:52:47.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"bb5fdc1de310c3442bcab2317fa281c685155039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 09:49:58.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fe26987adf6408fae63757fcdfaaa5c : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/491b3759b46e4cf3a26d0dff253d5f6f : SUCCESS in 8m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf2d493f46b04297a31715909f44b871 : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b85629cf7324ccdb377ab7e3586999d : SUCCESS in 10m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e4dc292fa814a88980bb0cd4fd93231 : SUCCESS in 9m 44s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ef3de0160e3a41afb0c7457af0ee12ba : SUCCESS in 43m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df411032107444d79a61f569017a7b6b : SUCCESS in 46m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10e68cf211c84993a58f1a12ead8ed13 : SUCCESS in 42m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/57dd550b10e54ee0b1a70eb6ac7cd69d : SUCCESS in 41m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/038ef839abaf4321a68d07594fc876aa : SUCCESS in 48m 24s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bec4acfe141540f698f375e2d15f3b45 : FAILURE in 31m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/05a286c3275c4b71b4a43059c798a069 : SUCCESS in 29m 48s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/78a276352876478da60bfb5c12743b14 : SUCCESS in 2h 52m 20s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/caa67a603bdd41eaab8e1133e0993e96 : SUCCESS in 57m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95aeadb9cf7c4a20b67794a20e25c624 : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":5},{"id":"43d6278499b4b3596667a96cb537efd7787ff33b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-21 10:47:50.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"8accbfeb67bbf8aca9456f40395d8fa38cf23274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 12:52:06.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d46b84c71a4f95adb71327e8141ca9 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba4594abae0942e0ad2f45c2cfd433a4 : SUCCESS in 8m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c98d5c55a324ae38fa37530a5e5a002 : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2be1c34204454f8195e9eca1ea2f32be : SUCCESS in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8133e8226f6b49f08d47619fc19f1996 : SUCCESS in 9m 02s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/22625863636a4f7b828c5cc3b9a183f5 : SUCCESS in 42m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31dfc2513aa440cca80051bafbe74cc1 : SUCCESS in 47m 03s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8f2a802fc21443dab0627f2dd5f517e : SUCCESS in 44m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b5ea538cba8748638dbf7979345760cc : SUCCESS in 41m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0ef332c9f4ac4434b07038cab10f6ce3 : SUCCESS in 52m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/29571662f7c84f029971d70800f25d59 : SUCCESS in 29m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6c79d4ceac994b6c9705609aa5f9a661 : SUCCESS in 27m 56s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4a169d3b3ad94f1dbac227b6e6a1e233 : FAILURE in 1h 55m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7078bd60be034c5890f570c51df28154 : SUCCESS in 58m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4616a43b67f8497288c60ba5e593369a : SUCCESS in 11m 26s","accounts_in_message":[],"_revision_number":5},{"id":"848c8d9a5c6b101e4d86aaf08e26f6b319ea0b00","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-21 15:25:27.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"fd6b4eba865c3a3107e64027f14bf528dd0d1251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 18:13:14.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f340441ac244ef83b366165a07df49 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d909cea682147028e3dacc98d7b9d2d : SUCCESS in 11m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c74c8026fc444dd590309afac32f9571 : SUCCESS in 11m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c5cb67b01941b9a5593bf3589ca6e7 : SUCCESS in 9m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47643497e8c3469eba4c79ff6464bd0b : SUCCESS in 8m 24s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4241fa3f28b04a77b7a49b9a6dc748e2 : SUCCESS in 43m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7bc94e03e9d4b45b016bf66af959fda : SUCCESS in 1h 04m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e380b1572b044cf7a3254a90904847d5 : SUCCESS in 40m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ae9c5403c62a4505abf14eca1ddb3126 : SUCCESS in 57m 14s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6338fae19cda4de2a1d8eb68b04b1df8 : SUCCESS in 58m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3ca8b27593724fb7b6d9f6c35f937f1e : FAILURE in 29m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8b29c589473b4108bfbce5e673b989d1 : SUCCESS in 28m 10s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/669df1036d4a402295e136d7515ce373 : SUCCESS in 2h 41m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f9c28da3addf4cf8b7dd28ea002c2be5 : SUCCESS in 57m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/139f53a2ea914df28df0113775ae3c95 : SUCCESS in 11m 08s","accounts_in_message":[],"_revision_number":5},{"id":"6967bd7090c291905e5ad390be109f67aeaff6c2","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-22 08:50:57.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"5b16f51b2ca22c03805761c3bef32771cc731d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 11:56:03.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/399c7041af324730926d41d759c465f1 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d2b5dfaef404562905415e502dac389 : SUCCESS in 9m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf013fbb31c54724b0fe274bd4c4d978 : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e65ecb7b12c4be3bdb9ef8fde93add9 : SUCCESS in 7m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98b2e3358b024ec281edca0225116dcf : SUCCESS in 9m 26s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8d1ad58c8ff14e70b589d8b36b281f08 : SUCCESS in 44m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27a0fc99499342f095c2606c9692a629 : SUCCESS in 44m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a6a6b3bbdec4761b85b0315500fbd99 : SUCCESS in 47m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3366e3581b284cf4b277e4716cf6f811 : SUCCESS in 45m 50s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bb886e19877847609b9271e7436ad54c : SUCCESS in 49m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ad06a633354f4af3873bd8b731e17749 : FAILURE in 31m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/88cf70c722574f95a047d40b242fca38 : SUCCESS in 27m 28s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0666e611427e4e0b8fc7d97cdc2fd839 : SUCCESS in 3h 00m 18s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/78f47a4ccd8847a98198699b1d5d7590 : SUCCESS in 1h 10m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/477744a5933b4580b69e366326a963d8 : SUCCESS in 11m 19s","accounts_in_message":[],"_revision_number":5},{"id":"7357cd32f8208a7d720d3d3646b587c71ff781e4","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-22 12:26:18.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"95a2ff61fb84141a477892b633c76f75de3c91ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:24:34.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa50e9f0630a40e4b9449c4900a48001 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8fbbb0707524dd1bf66832963c96d0b : SUCCESS in 8m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6c832fdd18e4f9e861fe05b59a71888 : SUCCESS in 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29f4f7311c5d45f3b12d0794fcccb70f : SUCCESS in 7m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01ff32ad2b6d4bc99957df72640831b6 : SUCCESS in 9m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/87a3e8bbec1f4f1eb73112646b829a97 : SUCCESS in 43m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/928ab8cfb1294c2bb3ac361ce579e990 : SUCCESS in 40m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/400ee9c1b7b241a88f659ac22fd294ee : SUCCESS in 1h 04m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/58130d3c07144b148e1307430653e998 : FAILURE in 1h 01m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/727310e92f604a3eba52910c512ec4f1 : SUCCESS in 48m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9db25e111caa44779cfb065de4a4b38d : SUCCESS in 29m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/48adaca814214648a3f6a90d12af1968 : SUCCESS in 33m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d05b7a367caa47ac8e63703c7d96a3e9 : SUCCESS in 2h 53m 07s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/08c10cbf3478486b9d9394c58789659e : SUCCESS in 58m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fb745a0a680449eaed4e4bfe741fbb1 : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":5},{"id":"6efad22761b2f63049250c7b2e30b9650f5e45da","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-22 18:26:14.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b83bcd00a26009004adddbc32cbf5b28808f334e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 21:08:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b35b57963594f4687657a4de81f0266 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dd5e20224fa47cba012a2a43811d998 : SUCCESS in 9m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/503b4d7ee02542ff8f9789061010ff2d : SUCCESS in 12m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c8ef249d8a04ac380f3d3feeb5efb73 : SUCCESS in 9m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb1f389053c247d6afe4abf54f096492 : SUCCESS in 10m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/29efdca83ec842e0bd541b88266fa24a : SUCCESS in 43m 47s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b9e294cc2d44989becc0f8069f0246d : SUCCESS in 45m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91b9489c20504869bb84e079925ddad6 : SUCCESS in 42m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/48b7b07e189840d2813620808153621c : SUCCESS in 47m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b9820f4dcf754fc987663409939801e9 : SUCCESS in 48m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/67104c393dc54b5ebef94003af349547 : SUCCESS in 31m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b6378b8a92d542a49f9a3162dcab049c : SUCCESS in 28m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cd49e2677dbc43ababfd6e25b069f0e1 : SUCCESS in 2h 38m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/39ecd7e3a59645e2bc1b78c435db44e6 : SUCCESS in 1h 02m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eee9a62e11aa44a79952be835f764601 : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":5},{"id":"670768b31e3d16416be164de465db400ba7ca7f7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-23 07:18:57.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"65cba2f2a8e25078a606d382ae537b8e70af24b7","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-23 08:27:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"acab4ac9ca9132c37d2e6c7a96362f49341a7f86","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-23 08:33:02.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"12e831fbf232bd3edc7fae73d2e650cca0935b25","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-23 08:38:01.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"57defaae08c900673e01e76578a1fc3294d5bb76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 11:48:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aca33c9e4c34e4e81c0a591006189f8 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/964b5fb435084a398a027a5d6813ec6d : SUCCESS in 8m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2eb443702634e82aa96100e0082bec1 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af146a2b4994cb99339262855de842e : SUCCESS in 8m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05b8b2818a544dc6a52e4a8e5fcf8474 : SUCCESS in 12m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/04ae8d40be4a4467b39d4da72ba4711d : SUCCESS in 41m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6d22a7a622644b3881ea91cadc907e2 : SUCCESS in 49m 49s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2c8bf79f42f4802a2d688466df6ec56 : SUCCESS in 53m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c4b33be7cfd84c6e8844850f9d787d59 : SUCCESS in 43m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ad0a330d8b2345fb916780903b5bd90e : SUCCESS in 41m 45s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ff4bef0b0e664a8a8e3ac03484cee8a1 : SUCCESS in 28m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5020c1d5527f4e8d8407e9f9d39d0ac6 : SUCCESS in 27m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f592ee1238d641fa99849c8c20018e2d : SUCCESS in 3h 12m 59s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5a2202be57fe456ea39263a1753acd1d : SUCCESS in 1h 00m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56d6ee5efb074c75adff5a01a5bfc505 : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":6},{"id":"e6652c69682c6ab0f38c737f74d9c94a9213fe04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 11:48:57.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"591aa44a639e197bb89adad0de4b91020cc0b2d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 13:46:06.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a82e314f8343f980a0621d5b36d7cd : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/382223bac07b4aaf9e86fcfad1d70ee3 : SUCCESS in 9m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5422c268743941c38705a05508fab3fd : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe8a2a4d57644fc198406222ef3b6978 : SUCCESS in 8m 49s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/324bd6bd68024e10a9afecc1b8a6ab3c : SUCCESS in 44m 20s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46368780c75f490c932eb0478de3c8e5 : SUCCESS in 46m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/747123c34eab49ed96b3e0e1bd13315a : SUCCESS in 48m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/686d8b0979bc4318b44d611ab4ecdf00 : SUCCESS in 1h 02m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3fcb189c45c44e7190d0ca35af29164d : SUCCESS in 30m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/34c89457c5134dc4892ad8fd698214b4 : FAILURE in 31m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0a53519e08ae4510b7e4980966523eee : SUCCESS in 43m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/199c5b1a5733484a954062404c8bdf1e : SUCCESS in 10m 49s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/636056cc39534afea01d596a2a3fb269 : SUCCESS in 1h 52m 28s","accounts_in_message":[],"_revision_number":6},{"id":"f805c64cc5622c6451e69324d133ed7aa9197806","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-23 14:06:26.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"d9d7e720b62ed38a1d52b3fec63a8cff25763f09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 17:10:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c5d42737024a6e841f39e764138499 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2568d5790db4e93bcd63397ff29ad98 : SUCCESS in 8m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ed0f5cb93d44e30b16a26cb078a89f3 : SUCCESS in 11m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f8a36d6942949bb8c40ec7b36c28d29 : SUCCESS in 10m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d86437eca6a475e977f464b29ebf720 : SUCCESS in 9m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/180c23bf27f04238a18030d6163be480 : SUCCESS in 48m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1c0ba03f4584e039f372c31cb7115bd : SUCCESS in 47m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34cd9e82dd5744cb92158839ba33ea19 : SUCCESS in 1h 01m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/080a2b18c8de4204a0783fe3bef0fb3f : SUCCESS in 43m 11s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/338f25743e464aa5817ff8c1e02b0f7e : SUCCESS in 39m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a1b0038f72034c34ad46cf474b1fa754 : SUCCESS in 31m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0a86679a344844429631535578d5b9be : SUCCESS in 27m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/53ec08689ba44487bf2e7c025f4b9f1f : SUCCESS in 2h 56m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c2a5f2098090492d8486ac18b7080ec1 : SUCCESS in 1h 07m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b22c4c19684f4be88f83767a45eb24ba : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":6},{"id":"f7e53905b1b455cc8380c856377751bc8847b864","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 17:10:28.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c1f8526e5133685594d03ff621947005594e4a06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 19:09:07.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2184f9cdf5dd40d28d1c5aa4aec12244 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07e1148312814c7b852c802dc3735ac3 : SUCCESS in 9m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d156862202a4f70bb334c4d8230ac3f : SUCCESS in 12m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74cc93b4e1964b13ae2cc2e163c8cb58 : SUCCESS in 10m 17s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/eebdac8ccf6e4da5b28260017c8f1904 : SUCCESS in 45m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8a7f5f2155a4668bb8356f476b6c0f7 : SUCCESS in 42m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6f9c90364184d9582fc547b13eee88c : SUCCESS in 40m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7e8e1e6d5a844f1a854f2c4c86a2c281 : SUCCESS in 43m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3dbccbb2301c44c1824156f3b4ca3201 : SUCCESS in 34m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d974861ae26b4527a11458f04b76bd28 : SUCCESS in 32m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7c8d13255ccf48c1848fdf968a4c04e3 : SUCCESS in 44m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65bbd8b853664b6b80ccd04799985076 : SUCCESS in 12m 13s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3e9d65d87bf4404d8a0338d7ec35717b : SUCCESS in 1h 53m 47s","accounts_in_message":[],"_revision_number":6},{"id":"5074e1dc30ca63883e32e2282c8c7a97cbe2b817","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 19:09:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"f1614209ac0e1dbcb3a1fc3e55a1371685d2d92c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 19:11:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7548e0bb9644af6955e6e72e0bf9862 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"fb3db8ddb79c843299262a2abb7beacf0df8f9f7","revisions":{"c6a004f9645a67d4608fe4931413c9ba87ff4f34":{"kind":"REWORK","_number":1,"created":"2020-01-16 21:40:54.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/91/702991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/702991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/702991/1"}}},"commit":{"parents":[{"commit":"55f181f8aaf8cdc7d519c1e446d0053cd73368a6","subject":"Fix requirements remove py2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55f181f8aaf8cdc7d519c1e446d0053cd73368a6"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-16 21:40:37.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-16 21:40:37.000000000","tz":180},"subject":"Add json param to HTTPAction to simplify work with jsons and to avoid errors with json arrays.","message":"Add json param to HTTPAction to simplify work with jsons and to\navoid errors with json arrays.\n\nChange-Id: Id866ed13764b1d4db75cf1a819b53a7e8955b34a\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c6a004f9645a67d4608fe4931413c9ba87ff4f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c6a004f9645a67d4608fe4931413c9ba87ff4f34"}]},"branch":"refs/heads/master"},"5118930a0af5c83004089b8954dd64554dd879be":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-16 21:43:36.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/91/702991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/702991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/702991/2"}}},"commit":{"parents":[{"commit":"55f181f8aaf8cdc7d519c1e446d0053cd73368a6","subject":"Fix requirements remove py2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55f181f8aaf8cdc7d519c1e446d0053cd73368a6"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-16 21:40:37.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-16 21:41:27.000000000","tz":180},"subject":"Add json param to HTTPAction To simplify work with jsons and to avoid errors with json arrays we should add support of request\u0027s json param alongside the data param.","message":"Add json param to HTTPAction\nTo simplify work with jsons and to avoid errors with json arrays\nwe should add support of request\u0027s json param alongside the data\nparam.\n\nChange-Id: Id866ed13764b1d4db75cf1a819b53a7e8955b34a\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5118930a0af5c83004089b8954dd64554dd879be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5118930a0af5c83004089b8954dd64554dd879be"}]},"branch":"refs/heads/master"},"2dff4f1f5f95448619fffb8de6f2a920958d325f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-16 21:44:32.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/91/702991/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/702991/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/702991/3"}}},"commit":{"parents":[{"commit":"55f181f8aaf8cdc7d519c1e446d0053cd73368a6","subject":"Fix requirements remove py2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55f181f8aaf8cdc7d519c1e446d0053cd73368a6"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-16 21:40:37.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-16 21:44:19.000000000","tz":180},"subject":"Add json param to HTTPAction","message":"Add json param to HTTPAction\n\nTo simplify work with jsons and to avoid errors with json arrays\nwe should add support of request\u0027s json param alongside the data\nparam.\n\nChange-Id: Id866ed13764b1d4db75cf1a819b53a7e8955b34a\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2dff4f1f5f95448619fffb8de6f2a920958d325f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2dff4f1f5f95448619fffb8de6f2a920958d325f"}]},"branch":"refs/heads/master"},"5d1fc03e0d78a7afc25d8cd38bbae598c969a5dc":{"kind":"REWORK","_number":4,"created":"2020-01-17 08:23:54.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/91/702991/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/702991/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/702991/4"}}},"commit":{"parents":[{"commit":"55f181f8aaf8cdc7d519c1e446d0053cd73368a6","subject":"Fix requirements remove py2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55f181f8aaf8cdc7d519c1e446d0053cd73368a6"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-16 21:40:37.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-17 08:23:41.000000000","tz":180},"subject":"Add json param to HTTPAction","message":"Add json param to HTTPAction\n\nTo simplify work with jsons and to avoid errors with json arrays\nwe should add support of request\u0027s json param alongside the data\nparam.\n\nChange-Id: Id866ed13764b1d4db75cf1a819b53a7e8955b34a\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5d1fc03e0d78a7afc25d8cd38bbae598c969a5dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5d1fc03e0d78a7afc25d8cd38bbae598c969a5dc"}]},"branch":"refs/heads/master"},"ce90186130867bfdf5956d1bfb3c8506286ae7be":{"kind":"REWORK","_number":5,"created":"2020-01-20 21:45:29.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/91/702991/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/702991/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/702991/5"}}},"commit":{"parents":[{"commit":"55f181f8aaf8cdc7d519c1e446d0053cd73368a6","subject":"Fix requirements remove py2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55f181f8aaf8cdc7d519c1e446d0053cd73368a6"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-16 21:40:37.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-20 21:45:08.000000000","tz":180},"subject":"Add json param to HTTPAction","message":"Add json param to HTTPAction\n\nTo simplify work with jsons and to avoid errors with json arrays\nwe should add support of request\u0027s json param alongside the data\nparam.\n\nChange-Id: Id866ed13764b1d4db75cf1a819b53a7e8955b34a\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ce90186130867bfdf5956d1bfb3c8506286ae7be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ce90186130867bfdf5956d1bfb3c8506286ae7be"}]},"branch":"refs/heads/master"},"fb3db8ddb79c843299262a2abb7beacf0df8f9f7":{"kind":"REWORK","_number":6,"created":"2020-01-23 08:27:43.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/91/702991/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/702991/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/702991/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/702991/6"}}},"commit":{"parents":[{"commit":"55f181f8aaf8cdc7d519c1e446d0053cd73368a6","subject":"Fix requirements remove py2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/55f181f8aaf8cdc7d519c1e446d0053cd73368a6"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-16 21:40:37.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-01-23 08:27:00.000000000","tz":180},"subject":"Add json param to HTTPAction","message":"Add json param to HTTPAction\n\nTo simplify work with jsons and to avoid errors with json arrays\nwe should add support of request\u0027s json param alongside the data\nparam.\n\nChange-Id: Id866ed13764b1d4db75cf1a819b53a7e8955b34a\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fb3db8ddb79c843299262a2abb7beacf0df8f9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/fb3db8ddb79c843299262a2abb7beacf0df8f9f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
