)]}'
{"id":"openstack%2Fmistral~703022","triplet_id":"openstack%2Fmistral~master~Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8","project":"openstack/mistral","branch":"master","topic":"bp/mistral-restructure-docs","hashtags":[],"change_id":"Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8","subject":"The first iteration of restructuring Mistral doc","status":"MERGED","created":"2020-01-17 07:14:10.000000000","updated":"2020-01-28 12:38:27.000000000","submitted":"2020-01-23 07:12:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703022-1579763576269-5aeb02ef","meta_rev_id":"f9237dd996d3a228ea1ec4eed260a04a58697b3a","_number":703022,"virtual_id_number":703022,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"date":"2020-01-22 07:08:55.000000000","_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-23 07:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2020-01-23 06:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":2,"date":"2020-01-22 08:55:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"date":"2020-01-22 07:07:49.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-01-22 07:07:49.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2020-01-22 07:07:49.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-23 06:51:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-22 07:07:49.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2020-01-22 07:07:49.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2020-01-22 07:07:49.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-01-22 08:55:31.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-01-23 06:51:46.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-01-23 06:53:00.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-01-23 07:12:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ae58624b391b72a6c6755e5a1435fe097ce98e0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-17 07:14:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c307a52f9007bd61e8abf9ec65fd49e028ca889d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 07:29:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1b262cc041f4250a4d36fdc9563d229 : FAILURE in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea04a149e0a24031ba3fdd253c3fa316 : FAILURE in 14m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8a272e22efe8a900fd151cfda1837755e175d7fe","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-17 10:37:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"18216e9ec4b6f3d7bd7d0c2fcba8e18c5c378038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 11:07:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae6cf5cfbf9042979f21a7b6eb5e19af : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/871311630dd64b19aa9abc370d69add9 : FAILURE in 6m 42s","accounts_in_message":[],"_revision_number":2},{"id":"98459422192b180dd5276296c01f2e98122b3fc0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-17 11:48:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"119a04f0ba78525637f367ecef3a7cef3352ce1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 11:59:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a910abd8d38b401b9fc6ab701dbac2a6 : FAILURE in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70a3da4320634df5a3e1827b42bd8986 : FAILURE in 7m 07s","accounts_in_message":[],"_revision_number":3},{"id":"b596c8372d5877c7067421e91502a7b3d04f69c6","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-17 12:09:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b950c1afb85a87e954ba4ca5c4976e58c798a167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 12:22:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18e7ee5d5ee64a70b57756afd514fc60 : FAILURE in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5027c9fa7994d8abc18d0a3573f4220 : FAILURE in 6m 03s","accounts_in_message":[],"_revision_number":4},{"id":"3e99d3ccfd18458599c6435c8a0a9f06f441f7c7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-21 05:17:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"af2ec2e61d7897fe436ba0b21e9f3ec2ed843dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 05:28:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22ee4df86a9b4694a22a3f4f6699960b : FAILURE in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/042899b23cd04f1c92929cd0b83d9117 : FAILURE in 9m 41s","accounts_in_message":[],"_revision_number":5},{"id":"696ed8ded38f92291e6e71b2c42c9ac2166c85df","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-21 06:12:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e9fefaf40763481d97bd0491e625a13a63c47add","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 06:29:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f1b8377da0049edb6ac23556d560ed9 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c1539f46d9a43b9b40ee27afdb9e9ba : FAILURE in 8m 34s","accounts_in_message":[],"_revision_number":6},{"id":"73733646dc215b720c1ba74afad08a25cb95458a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-21 07:39:06.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1d80fe4c593f7b10437cf91849ba161d27d742ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 07:46:57.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a2dc0f37e3e40c080f78021824126f6 : FAILURE in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55565da8471e45ab9bb8c21030424a9b : FAILURE in 6m 28s","accounts_in_message":[],"_revision_number":7},{"id":"7501c3c24aef577ed873d4c5b7d7e3b873fec78e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-21 08:11:07.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b038cc83f46d756f47cfc1723c8a813355ea6ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 08:19:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e55dcbad434da3be69e52b04a9a0a2 : FAILURE in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47a32c92e4ce45c4958443a7faef6d09 : FAILURE in 6m 37s","accounts_in_message":[],"_revision_number":8},{"id":"0362c05b3b6f586fcba167358a0b6feeb5a969fd","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-21 09:06:30.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"acf0a4dbae534620966862cdb3f90aab86d78992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 09:16:24.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c798b11add640d29f770fb8149c4150 : FAILURE in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f6840c2783a4f79baec70e8a4d21240 : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":9},{"id":"2c8cd29073e15b2097dec09240a280857e51df02","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-21 09:27:26.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5c4797905468091781b492b8e75415c5e72a674d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 09:39:23.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25ae150b88374d17864bec3ad2aa1988 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99cedf1b374843b1a49ef742ee8b1953 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":10},{"id":"bc1cea017413e35313bf872f77bfee3d02931bcf","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-21 10:07:13.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"cc3b002ab111ace87f69a47f905c4aeeaee004bf","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-22 07:06:47.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a8c44ff7f24f47f6942afe7d11a84fee5c23ec41","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-22 07:07:31.000000000","message":"Uploaded patch set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"a0c22872fcce553550afeb3f2ee44cc2e34e7765","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-22 07:08:55.000000000","message":"Patch Set 12:\n\nThe patch is big so I\u0027d recommend to look at the result (click on the docs gate when it passes) rather than trying to review all source code changes.","accounts_in_message":[],"_revision_number":12},{"id":"02dfbe4cfdafe69486ac017b3da7cb480e92c390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 07:25:29.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9e19a6623744b1a1bdf9ebde5eb2f6 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba8eb6de81e8486cb85daa1de69792bd : SUCCESS in 17m 30s","accounts_in_message":[],"_revision_number":12},{"id":"07495052201b192f085696f55a3f7931b92417ad","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-22 08:55:31.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"bdf0253d824b34d44344fbe014df7852605d4060","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-23 06:51:46.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"e3edde93c1938cdb5156782922ae3a492a038624","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 06:51:59.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"09addc49f726d662d13ef81b30d32fef3bf80afc","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-01-23 06:53:00.000000000","message":"Patch Set 12: Code-Review+2\n\nThank you Renat!","accounts_in_message":[],"_revision_number":12},{"id":"fa955b885894f7052c924d26fbd90afb3304234b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 07:12:56.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6952dcdfe284ae8b052673ac0c62997 : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd09baa9c380464caa71b7cd46db06e5 : SUCCESS in 18m 31s","accounts_in_message":[],"_revision_number":12},{"id":"bec6a761aec0631cd6f5b82e115af9dddc46d762","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 07:12:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"b97bab1f482d071f5bafbca8fb595a3abdc1a9fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 07:15:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8c0737ea86b46b2b7fbc5af457a4fce : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":12},{"id":"a71dd0d2d29fe8baab3a75fad82ec464420869d1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-28 11:54:21.000000000","message":"Patch Set 12:\n\nthis removes *standarized* links to INstall docs and others:\n2020-01-28 10:17:17,017 ERROR    has_install_guide set for mistral but https://docs.openstack.org/mistral/latest/install/index.html does not exist (404)\n2020-01-28 10:17:17,018 ERROR    has_config_ref set for mistral but https://docs.openstack.org/mistral/latest/configuration/index.html does not exist (404)\n2020-01-28 10:17:17,018 ERROR    has_in_tree_api_docs set for mistral but https://docs.openstack.org/mistral/latest/api/index.html does not exist (404)\n\nSee https://docs.openstack.org/doc-contrib-guide/doc-index.html and a previous OpenSTack community goal for details.\n\nI\u0027ll remove the links now to your content to unblock publishing of openstack-manuals.","accounts_in_message":[],"_revision_number":12},{"id":"9e82b5563bde240c7f3fb3f68b4d64c481f8069e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-28 11:56:51.000000000","message":"Patch Set 12:\n\nSee also https://review.opendev.org/704540","accounts_in_message":[],"_revision_number":12},{"id":"70ee2c7eed1c6a43109fa9dd238f32320e38919d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-28 12:37:48.000000000","message":"Patch Set 12:\n\nYeah, don\u0027t let me stop you if you really want to do this, but we expect to find the admin guide at \u0027doc/source/admin\u0027, user guide at \u0027doc/source/user\u0027, config guide at \u0027doc/source/configuration\u0027, etc. This was all documented back in Pike [1]. You can do this but at AJaeger notes, it prevents us from linking to your docs at e.g. https://docs.openstack.org/train/user/. Could you consider sticking to the prescribed structure and just working within those limits?\n\n[1] https://specs.openstack.org/openstack/docs-specs/specs/pike/os-manuals-migration.html","accounts_in_message":[],"_revision_number":12},{"id":"a58feb5c3167b596e4d958435646d04e1518967f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-28 12:38:27.000000000","message":"Patch Set 12:\n\nPing me on #openstack-doc (nick: stephenfin) if you\u0027ve questions about why we do things that way or anything else doc\u0027y","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"9fb213c31518f201cfa548f6248da01b61ff0a07","revisions":{"ee491c8697b1ed7da71e53ba23ce506a69bbcf5a":{"kind":"REWORK","_number":1,"created":"2020-01-17 07:14:10.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/703022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/703022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/703022/1"}}},"commit":{"parents":[{"commit":"e4d54ec88199cad704e570435ab8a06f9cc30e06","subject":"Remove mistral-devstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4d54ec88199cad704e570435ab8a06f9cc30e06"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-16 12:09:36.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-17 07:13:54.000000000","tz":420},"subject":"WIP","message":"WIP\n\nChange-Id: Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee491c8697b1ed7da71e53ba23ce506a69bbcf5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee491c8697b1ed7da71e53ba23ce506a69bbcf5a"}]},"branch":"refs/heads/master"},"72fa616c5df4b2a7f7219fe50e9287302ce44710":{"kind":"REWORK","_number":2,"created":"2020-01-17 10:37:05.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/703022/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/703022/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/703022/2"}}},"commit":{"parents":[{"commit":"e4d54ec88199cad704e570435ab8a06f9cc30e06","subject":"Remove mistral-devstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4d54ec88199cad704e570435ab8a06f9cc30e06"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-16 12:09:36.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-17 10:36:50.000000000","tz":420},"subject":"WIP","message":"WIP\n\nChange-Id: Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/72fa616c5df4b2a7f7219fe50e9287302ce44710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/72fa616c5df4b2a7f7219fe50e9287302ce44710"}]},"branch":"refs/heads/master"},"353ccbc90d85efdcae9988f859ca39b7309a9965":{"kind":"REWORK","_number":3,"created":"2020-01-17 11:48:21.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/703022/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/703022/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/703022/3"}}},"commit":{"parents":[{"commit":"e4d54ec88199cad704e570435ab8a06f9cc30e06","subject":"Remove mistral-devstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4d54ec88199cad704e570435ab8a06f9cc30e06"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-16 12:09:36.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-17 11:48:00.000000000","tz":420},"subject":"WIP","message":"WIP\n\nChange-Id: Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/353ccbc90d85efdcae9988f859ca39b7309a9965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/353ccbc90d85efdcae9988f859ca39b7309a9965"}]},"branch":"refs/heads/master"},"8dffcbb8ec909abb704737c050fe13e95c892cb4":{"kind":"REWORK","_number":4,"created":"2020-01-17 12:09:10.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/703022/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/703022/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/703022/4"}}},"commit":{"parents":[{"commit":"e4d54ec88199cad704e570435ab8a06f9cc30e06","subject":"Remove mistral-devstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4d54ec88199cad704e570435ab8a06f9cc30e06"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-16 12:09:36.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-17 12:08:52.000000000","tz":420},"subject":"WIP","message":"WIP\n\nChange-Id: Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8dffcbb8ec909abb704737c050fe13e95c892cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8dffcbb8ec909abb704737c050fe13e95c892cb4"}]},"branch":"refs/heads/master"},"e3174d9879370f9a15fd6f07c97bd4cf88d633eb":{"kind":"REWORK","_number":5,"created":"2020-01-21 05:17:14.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/703022/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/703022/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/703022/5"}}},"commit":{"parents":[{"commit":"e4d54ec88199cad704e570435ab8a06f9cc30e06","subject":"Remove mistral-devstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4d54ec88199cad704e570435ab8a06f9cc30e06"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-16 12:09:36.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-21 05:16:40.000000000","tz":420},"subject":"WIP","message":"WIP\n\nChange-Id: Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e3174d9879370f9a15fd6f07c97bd4cf88d633eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e3174d9879370f9a15fd6f07c97bd4cf88d633eb"}]},"branch":"refs/heads/master"},"b087ec59724425c489a2a7fb7595f16def976c16":{"kind":"REWORK","_number":6,"created":"2020-01-21 06:12:49.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/703022/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/703022/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/703022/6"}}},"commit":{"parents":[{"commit":"e4d54ec88199cad704e570435ab8a06f9cc30e06","subject":"Remove mistral-devstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4d54ec88199cad704e570435ab8a06f9cc30e06"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-16 12:09:36.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-21 06:12:26.000000000","tz":420},"subject":"WIP","message":"WIP\n\nChange-Id: Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b087ec59724425c489a2a7fb7595f16def976c16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b087ec59724425c489a2a7fb7595f16def976c16"}]},"branch":"refs/heads/master"},"efa26d6229f0bcc56e07fa5ff9bde18161950907":{"kind":"REWORK","_number":7,"created":"2020-01-21 07:39:06.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/703022/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/703022/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/703022/7"}}},"commit":{"parents":[{"commit":"e4d54ec88199cad704e570435ab8a06f9cc30e06","subject":"Remove mistral-devstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4d54ec88199cad704e570435ab8a06f9cc30e06"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-16 12:09:36.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-21 07:38:49.000000000","tz":420},"subject":"WIP","message":"WIP\n\nChange-Id: Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/efa26d6229f0bcc56e07fa5ff9bde18161950907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/efa26d6229f0bcc56e07fa5ff9bde18161950907"}]},"branch":"refs/heads/master"},"4825981e01efdf8b501ffef4ab877985b5b9c184":{"kind":"REWORK","_number":8,"created":"2020-01-21 08:11:07.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/703022/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/703022/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/703022/8"}}},"commit":{"parents":[{"commit":"e4d54ec88199cad704e570435ab8a06f9cc30e06","subject":"Remove mistral-devstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4d54ec88199cad704e570435ab8a06f9cc30e06"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-16 12:09:36.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-21 08:10:27.000000000","tz":420},"subject":"WIP","message":"WIP\n\nChange-Id: Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4825981e01efdf8b501ffef4ab877985b5b9c184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4825981e01efdf8b501ffef4ab877985b5b9c184"}]},"branch":"refs/heads/master"},"41373c4c6eb59828603bdabd490ce3b0300ef15a":{"kind":"REWORK","_number":9,"created":"2020-01-21 09:06:30.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/703022/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/703022/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/703022/9"}}},"commit":{"parents":[{"commit":"e4d54ec88199cad704e570435ab8a06f9cc30e06","subject":"Remove mistral-devstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4d54ec88199cad704e570435ab8a06f9cc30e06"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-16 12:09:36.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-21 09:03:11.000000000","tz":420},"subject":"The first iteration of restructuring Mistral doc","message":"The first iteration of restructuring Mistral doc\n\n* Grouped documentation articles into several main parts: user docs,\n  admin docs, developer docs.\n* Changed the index page so that it contains links to the index pages\n  of the main documentation parts like user docs, admin docs etc.\n* Fixed all the broken links\n\nPartially implements: blueprint mistral-restructure-docs\nChange-Id: Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/41373c4c6eb59828603bdabd490ce3b0300ef15a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/41373c4c6eb59828603bdabd490ce3b0300ef15a"}]},"branch":"refs/heads/master"},"291ca67acbdb9606a51766c8047b64fe67c618f5":{"kind":"REWORK","_number":10,"created":"2020-01-21 09:27:26.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/703022/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/703022/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/703022/10"}}},"commit":{"parents":[{"commit":"e4d54ec88199cad704e570435ab8a06f9cc30e06","subject":"Remove mistral-devstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4d54ec88199cad704e570435ab8a06f9cc30e06"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-16 12:09:36.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-21 09:27:08.000000000","tz":420},"subject":"The first iteration of restructuring Mistral doc","message":"The first iteration of restructuring Mistral doc\n\n* Grouped documentation articles into several main parts: user docs,\n  admin docs, developer docs.\n* Changed the index page so that it contains links to the index pages\n  of the main documentation parts like user docs, admin docs etc.\n* Fixed all the broken links\n\nPartially implements: blueprint mistral-restructure-docs\nChange-Id: Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/291ca67acbdb9606a51766c8047b64fe67c618f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/291ca67acbdb9606a51766c8047b64fe67c618f5"}]},"branch":"refs/heads/master"},"4c300da5c529efc08c0e56c2b8594ddec8d6f93e":{"kind":"REWORK","_number":11,"created":"2020-01-22 07:06:47.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/703022/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/703022/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/703022/11"}}},"commit":{"parents":[{"commit":"e4d54ec88199cad704e570435ab8a06f9cc30e06","subject":"Remove mistral-devstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4d54ec88199cad704e570435ab8a06f9cc30e06"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-16 12:09:36.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-22 06:59:07.000000000","tz":420},"subject":"The first iteration of restructuring Mistral doc","message":"The first iteration of restructuring Mistral doc\n\n* Grouped documentation articles into several main parts: user docs,\n  admin docs, developer docs.\n* Changed the index page so that it contains links to the index pages\n  of the main documentation parts like user docs, admin docs etc.\n* Fixed all the broken links\n* Removed index generation since it\u0027s not informative at all in its\n  current form and it exposes code internals (REST controller class\n  names)\n* Removed obsolete properties that are no longer used by the\n  \"openstackdocstheme\" project\n* Removed obsolete static html files\n* Moved part of the images to the folders where they are used\n\nFurther work:\n\n* Refactor main chapters (user, admin, developer) one by one and\n  make them look consistent. For example, there are several pages\n  that provide similar kind of information: overview, quick start,\n  main features etc. It is a mess that\u0027s been accumulated throughout\n  the last 4-5 years.\n* Fill the gaps. Add all missing pages like: event notifications,\n  workflow environment, etc.\n\nPartially implements: blueprint mistral-restructure-docs\nChange-Id: Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4c300da5c529efc08c0e56c2b8594ddec8d6f93e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4c300da5c529efc08c0e56c2b8594ddec8d6f93e"}]},"branch":"refs/heads/master"},"9fb213c31518f201cfa548f6248da01b61ff0a07":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2020-01-22 07:07:31.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/22/703022/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/22/703022/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/22/703022/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/22/703022/12"}}},"commit":{"parents":[{"commit":"e4d54ec88199cad704e570435ab8a06f9cc30e06","subject":"Remove mistral-devstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4d54ec88199cad704e570435ab8a06f9cc30e06"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-16 12:09:36.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-22 07:06:51.000000000","tz":420},"subject":"The first iteration of restructuring Mistral doc","message":"The first iteration of restructuring Mistral doc\n\n* Grouped documentation articles into several main parts: user docs,\n  admin docs, developer docs.\n* Changed the index page so that it contains links to the index pages\n  of the main documentation parts like user docs, admin docs etc.\n* Fixed all the broken links\n* Removed index generation since it\u0027s not informative at all in its\n  current form and it exposes code internals (REST controller class\n  names)\n* Removed obsolete properties that are no longer used by the\n  \"openstackdocstheme\" project\n* Removed obsolete static html files\n* Moved part of the images to the folders where they are used\n\nFurther work:\n\n* Refactor main chapters (user, admin, developer) one by one and\n  make them look consistent. For example, there are several pages\n  that provide similar kind of information: overview, quick start,\n  main features etc. It is a mess that\u0027s been accumulated throughout\n  the last 4-5 years.\n* Fill the gaps. Add all missing pages like: event notifications,\n  workflow environment, etc.\n* Move cookbooks from Wiki to this doc.\n\nPartially implements: blueprint mistral-restructure-docs\nChange-Id: Ia722a6885ad2fb97d63a34285b0a5b1a23da79e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9fb213c31518f201cfa548f6248da01b61ff0a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9fb213c31518f201cfa548f6248da01b61ff0a07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
