)]}'
{"id":"openstack%2Fmistral~703290","triplet_id":"openstack%2Fmistral~master~I367926727291f5f21b1849237ca29fdce7e926c0","project":"openstack/mistral","branch":"master","topic":"gnocchi","hashtags":[],"change_id":"I367926727291f5f21b1849237ca29fdce7e926c0","subject":"Gnocchi: fix client options","status":"MERGED","created":"2020-01-19 12:26:44.000000000","updated":"2020-01-22 17:23:57.000000000","submitted":"2020-01-22 17:21:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703290-1579713698524-5f1f21a0","meta_rev_id":"cac00ddcb4733d70a5bc5d3d3314658daa72c0c0","_number":703290,"virtual_id_number":703290,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 17:21:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-22 07:25:08.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2020-01-21 07:46:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2020-01-20 06:38:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2020-01-22 05:26:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 07:46:23.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-01-22 05:26:58.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-01-22 17:21:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"883436b915eab812dfa4678d3f9ba2d38dbfc50b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-19 12:26:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"323f36d883bf649b98bed505ab39753c9e83ee58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 13:42:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f46d8816ff5741aca220bcb75e657771 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fbe375cbaca40f6be0fba18e4dba680 : SUCCESS in 8m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7cd57bb4118d4b1ea73bdb1440370ae2 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7222818904f4321862f41e8191732ed : SUCCESS in 7m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/133ab540ae2c45aab5f944e138aea71f : SUCCESS in 7m 57s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1113965c9be643398239a88c6f9a39d5 : SUCCESS in 35m 58s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41104794008347c5b2f4f9abc27e3230 : SUCCESS in 35m 24s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9883d352f8949b382572e342532cf00 : SUCCESS in 37m 12s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4904d48a49a340deb372429e53dae873 : SUCCESS in 44m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e409c4eaf4314049b32de61045fa7550 : SUCCESS in 41m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4efe57c38ff0469bb39f761998e05d7d : FAILURE in 29m 36s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/700e10d96f0d45e583dd33dc12b018a1 : SUCCESS in 27m 47s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5c433d4d93574fd1ae59bc3d5dee50ec : FAILURE in 1h 10m 49s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ccb3cdafba5f44beb71b31b220efd367 : SUCCESS in 57m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd8a779a0d5c4bec8e55f91376ccfd77 : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"375d8222597f3d5b0df8f3df6ff0f5aaa008f567","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-19 14:01:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3cf38049d38ea18373c461fba34e098a002a7d63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 16:50:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e367d1371806429094502f60662a9702 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b4a159515af4f2ea1fb9cfb2f026dbe : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58e7f4fc28b3401b864ce9858cbe708e : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3be8376de224b3aab6a77598b338f0b : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/349cf58c338340d4b409aa81967f3d6b : SUCCESS in 9m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a249ea8f9e574824b920788004750366 : SUCCESS in 41m 54s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9fbe171bd044b87ab78d369a07e2585 : SUCCESS in 42m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12e2a3c9bcfc4c5ebdbaa7a670d0f912 : SUCCESS in 41m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/99207426c1f94e24b73ea34d0046b366 : SUCCESS in 43m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/468841518a68447294e95fc7de0a9fac : SUCCESS in 43m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/63d0fcc639764e40b9e2bb8a452d9372 : SUCCESS in 28m 20s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ee021c87a1c948d4a1430946d1e8918c : SUCCESS in 26m 06s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/792afae67b174a05a8a9b8abef293654 : SUCCESS in 2h 44m 11s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8141229cdf9044f39eee56c028d46b29 : SUCCESS in 56m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/372fd3d9c344466992097e99fb0740b5 : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c5fbbd0ce95977a5a66f5477ef9e11ad7a3d7193","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-20 06:38:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d39c70a45a7b18fb0726954bbb36f1f5520721fc","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-21 07:46:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"643b065ffbb596067c91506ce61401cb9f33d5c2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-22 05:26:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e59d9c499dbedfa0c4922de8d45e9c29a77676f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 05:27:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08ec7dfb46f71487e36b0b82ff95bcc83730f9db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 07:23:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/843ef6fba1f64fe4b8621cf492f204e5 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e10dd4ef05b640d8838199918bbaf038 : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ce5d3a1599b427a8bfad27ead8238d6 : SUCCESS in 12m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6fa32b0342f424ab71f159d36c4a303 : SUCCESS in 7m 34s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e2ec856819f447bba794e3a7df8219c6 : SUCCESS in 53m 47s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa866b808c3941d28614a50cf5727d82 : SUCCESS in 48m 44s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8baea5bd81e4eff821ee8db2f891640 : SUCCESS in 53m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8e2b692fcb39425eb2348eb913864040 : SUCCESS in 43m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d10847544b3f441da072fe8ced426140 : FAILURE in 31m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7891c0d2f3fd4cb4b8824178cbb50645 : SUCCESS in 30m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c347b43f9635407a833c5bc0eea4ac56 : SUCCESS in 49m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a6963f9c3d548568f4e7aa6e3fb08b0 : SUCCESS in 11m 29s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/aecb650a25b743879c3f58d3073de5b9 : SUCCESS in 1h 56m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5a62d0388d2f5f2a68efbec0bcd54ef521241d76","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-22 07:25:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8d88c87588ae608eda1917729c2b860b92d5e643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 10:25:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d70cb243354dc48cfa2e44b4eab841 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e35b98a9ab964ce4ae659dfebd216270 : SUCCESS in 9m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab33eba17cc149558396c42a82ca6ba3 : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab1a7c1d6b743bc8bc2f890c7b0b153 : SUCCESS in 8m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df0d747a96ad4486a7c4d11f6c833920 : SUCCESS in 10m 19s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6931a2a34a5d45c6a79a1f6dfc2f6ccc : SUCCESS in 46m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f12f4bdf243e4fd08d2bc58c6e5bade6 : SUCCESS in 46m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a45dd36bec04933b72ce12f6a05d3e1 : SUCCESS in 44m 59s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/87cc665c2179452482febd925ac86d3e : SUCCESS in 43m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f23e25a8ea6243d2a9a5808ac93bd702 : SUCCESS in 47m 06s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f7bde89a50c2427a85ef07d132a33486 : SUCCESS in 30m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4cb2ef62874b4e3ebc3d20e394c9c877 : SUCCESS in 27m 25s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/288cc169df5e48708b458115feb01a00 : SUCCESS in 2h 55m 55s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8c39e2dfe03449cfa92b974d8b63b10b : SUCCESS in 1h 02m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3481cc4f96894d5ea16082604ef54776 : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":1},{"id":"50d1f7b06faf9c4724d7af560810a1711983f3c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 10:26:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ec2a67c14c85150ffd410ebc8767c90c41c6eca","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-22 11:56:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"708b30613f7ad68a602aef169a3c3631c40fb9bf","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-22 11:56:11.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"0034fe6289a0fb87e2839275051012d4d37b968d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 14:59:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a0d1045aa0c43abb8da30e4c297cf35 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3449e774f39f47938f4e564d88dfb87e : SUCCESS in 12m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90bdb748234e4b9299b6f3916e63f6a8 : SUCCESS in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b0b9a57c10742e8be94b34612a85f24 : SUCCESS in 9m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebb0d591abbe499698f66d8b316dadc7 : SUCCESS in 11m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/14bec0474807404588f357c0fdeb5c4c : SUCCESS in 51m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69b8eaf15a3449198be14992b849cafa : SUCCESS in 37m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49683db065d3469e9b192210b171b806 : SUCCESS in 59m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/defc40f51c5a4834a72d209d86a4d100 : SUCCESS in 38m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/65181c8b5721428588da45f9d0958ddc : SUCCESS in 48m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b1e4fe1a727d47229608d244a6a9e2b9 : SUCCESS in 31m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/db4a65a1bcde4d8f971236fd1103b55c : SUCCESS in 27m 21s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1e562e04db63408781374f2d8ce92972 : SUCCESS in 2h 58m 38s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d15bc04d898b4d4692874a4d3ced1e18 : SUCCESS in 1h 02m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/817c5dc2f35f488387fa35f8492673b1 : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d15a9088fa1448f8253fe2ebca484368568951df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 14:59:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26b819ab42654a79e2ae4ef49cd9a272f222ea21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 17:21:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1280bc39663cf42187b620f9e0738093f38ca34e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 17:21:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704b8eb5f6cd408bb8329035bab03bca : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7cc890e65494b09922be3526fd5335e : SUCCESS in 9m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6e6c68115864d319a319b0a38c88e94 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a94f37f4db5842eaab94a40b032a338d : SUCCESS in 9m 06s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/38bb7043fa7f4a4a86bb76ae80717028 : SUCCESS in 49m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/964183e06b17494387b79fc32470ecb2 : SUCCESS in 47m 30s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e989aabd0f7434fbfd70e6cd7bab3a3 : SUCCESS in 45m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ede794ba0fbe42baac5f25c326e4f8bb : SUCCESS in 47m 45s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a23e5590148a44fea41fbe8ff6013b42 : SUCCESS in 30m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/37e7f0f78aa74a13abe063a9da6cac3f : SUCCESS in 28m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/87445b13c1c84e738ce8176dddf0f335 : SUCCESS in 38m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad7d44b5dcd2441899e086f947c5b440 : SUCCESS in 12m 31s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/88865f0bf8df499fa9f5fe63b64881e2 : SUCCESS in 2h 13m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e2222dcd0d90127dd2d3c2801bae984e8b3a21d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 17:23:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70632f6585e5424493efa448a36d63e6 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"947a986a72a1f2ccf139fd42a33897f8746f89ae","revisions":{"947a986a72a1f2ccf139fd42a33897f8746f89ae":{"kind":"REWORK","_number":1,"created":"2020-01-19 12:26:44.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/90/703290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/90/703290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/90/703290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/90/703290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/90/703290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/90/703290/1"}}},"commit":{"parents":[{"commit":"7f92cc8f5b704b5af8d430e5e7372756bbbc4d87","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7f92cc8f5b704b5af8d430e5e7372756bbbc4d87"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-19 12:26:26.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-19 12:26:26.000000000","tz":120},"subject":"Gnocchi: fix client options","message":"Gnocchi: fix client options\n\nBased on https://review.opendev.org/#/c/488307/3\n\nChange-Id: I367926727291f5f21b1849237ca29fdce7e926c0\nCloses-Bug: 1707131\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/947a986a72a1f2ccf139fd42a33897f8746f89ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/947a986a72a1f2ccf139fd42a33897f8746f89ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
