)]}'
{"id":"openstack%2Fmistral~703294","triplet_id":"openstack%2Fmistral~master~I4b3fc7e0ad4552022d44c3bdd8c87004f5e8ff46","project":"openstack/mistral","branch":"master","topic":"bug/1631541","hashtags":[],"change_id":"I4b3fc7e0ad4552022d44c3bdd8c87004f5e8ff46","subject":"wrapped the value of parameters in inputs(in wf get API) with \"\".","status":"MERGED","created":"2020-01-19 14:21:02.000000000","updated":"2020-01-28 10:01:13.000000000","submitted":"2020-01-28 09:59:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703294-1580205564049-cc57de1a","meta_rev_id":"d9e9b0b7c46663e3477d3041f1631940202d6271","_number":703294,"virtual_id_number":703294,"owner":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-01-27 08:42:41.000000000","_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 09:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2020-01-28 07:55:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":2,"date":"2020-01-28 04:42:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-28 08:22:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 04:42:41.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-01-28 07:55:00.000000000","updated_by":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"reviewer":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"state":"REVIEWER"},{"updated":"2020-01-28 08:22:56.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-01-28 09:59:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7fa38add300f939db74b77a5a12617543dbacfc1","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-19 14:21:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07add466cf1efc172d4c92627e193bce319ccd01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 17:33:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccac79b6e8fc4fb9ac147f199413267d : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ddd7139b6744446b138ab8f82786f68 : SUCCESS in 11m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d796bb444474bef973a32b7e092dd9b : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a95937463ce496c989c5cc9de6ffa49 : SUCCESS in 8m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56d490acb6fc42e19788e74317066835 : SUCCESS in 8m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e5e9abe964d040808a6bf44c1e33862d : SUCCESS in 40m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53991ab0770540e7a06318a1ec4e196a : SUCCESS in 55m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3be98323268e477d9a41c98d90092095 : FAILURE in 47m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/22e4c33e10d44d0593c47306cb91dcbb : FAILURE in 47m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/72744ce1c0ef413da24434835c332f84 : SUCCESS in 42m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/83b4e7ac16f14d15a358cde48d9bf408 : SUCCESS in 28m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/cd3d9836ba794806a91882aebf920bc6 : SUCCESS in 33m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4acdf013ae3149e5822d81385eaa93f6 : SUCCESS in 3h 03m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e208cd8174d3411f9d6a646145375efe : SUCCESS in 1h 13m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f422a90e155545f5a11865770efed27f : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3e643b39db5fc8ee4464e4f2921ee585a05490ad","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-27 08:42:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ffef55b80b4cfdf5d202e0a019fe1600b3db71f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 11:26:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2516e5dff2b540bc81acef54ce32aa4a : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3f776883f6a48a194e5316b46d34665 : SUCCESS in 8m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0055601bbaac45348474aa90111b1aa4 : SUCCESS in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a1a6ce5dfe4431d868fa081b33d1fd6 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/475849aabf07435f988c35604b65c72e : SUCCESS in 7m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2004129eb6344471bac9e81191fb96ed : SUCCESS in 8m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/77af89cc734f46bda69bd26ddd56867e : SUCCESS in 41m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9234958b2184273860fb6d36803b200 : SUCCESS in 41m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51d88275af8e44ec92eb1417d7bdab49 : SUCCESS in 42m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/69529723e5e14e0481cad6bf6edb0ebc : SUCCESS in 42m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c4060ff536d5429fb8b1448fb2d76f55 : SUCCESS in 40m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cc2731c93e804eb89ffb98c18af415fb : SUCCESS in 28m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8405871abba641779e90a7f6be780977 : SUCCESS in 26m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c280408187bc44e1821bfa12b1ef3b4e : SUCCESS in 2h 40m 00s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/15afa1b178094d39b98593aa9f94105a : SUCCESS in 59m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a31f7007884445f9e715c1dc2fc2707 : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":1},{"id":"12a3b0d27062b78fc4df0ebd358acbba05393764","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-28 04:42:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2931d28679ae643aa191227e8c2e6e65e4f8bf0a","author":{"_account_id":9029,"name":"Moshe Elisha","email":"moshe.elisha@nokia.com","username":"MosheElisha"},"date":"2020-01-28 07:55:00.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM as we also have the new \"interface\" field","accounts_in_message":[],"_revision_number":1},{"id":"08988e33b2f6099db8ca343439c41d7b1d6761c9","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-28 08:22:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f18fe96e4c9de4ad0cc3f10e38feabae6ce90d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 08:23:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fde9255341728e9f20de22e58f3437daf3b715d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 09:59:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4f02bdf8ab541d387a2c8fe053d5673 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37296e8494ce4959870b8c45e9f19127 : SUCCESS in 8m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0202bf6c7984588a52ef2da508cbed7 : SUCCESS in 10m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0564a198dd94099ab35b13236d240da : SUCCESS in 10m 51s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9d98e7941fa94a6fa21373347f81de6e : SUCCESS in 42m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80df078d3e3e48ee8bb712cf581eeb88 : SUCCESS in 41m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a0ad20644154857b44dbfbab732659a : SUCCESS in 43m 29s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/57ed53300d9c4bba88464c7842ba2e47 : SUCCESS in 40m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/cd0170fc4d5c42668d19a747da8f39fb : SUCCESS in 30m 15s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f3e62c9d5b854f6db7eb840074e4799f : SUCCESS in 29m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6beada36cf1c4bd384ae21c9784cfa85 : SUCCESS in 41m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4646d0d14b294313b6315233cc9ae42d : SUCCESS in 13m 01s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/09a77e9d05094779b744c3c033c14ff8 : SUCCESS in 1h 33m 28s","accounts_in_message":[],"_revision_number":1},{"id":"45f392bfcf47baaed7870b22b7ae83971d0128d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 09:59:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1455fb11d291d98d738a8b94e9d5e3d4b86f5457","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 10:01:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d10df15539b49208e65c7bf031da915 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a6309683b698dc8ec41955ae1ac77b0200ff6a9","revisions":{"4a6309683b698dc8ec41955ae1ac77b0200ff6a9":{"kind":"REWORK","_number":1,"created":"2020-01-19 14:21:02.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/94/703294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/94/703294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/94/703294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/94/703294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/94/703294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/94/703294/1"}}},"commit":{"parents":[{"commit":"5e90c610f8ea2a6de66a6e24cc3d7d1d5ef418ea","subject":"Merge \"Add \u0027interface\u0027 field to Workflow resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5e90c610f8ea2a6de66a6e24cc3d7d1d5ef418ea"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-01-19 14:20:51.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-01-19 14:20:51.000000000","tz":0},"subject":"wrapped the value of parameters in inputs(in wf get API) with \"\".","message":"wrapped the value of parameters in inputs(in wf get API) with \"\".\n\n the input field of workflow-get API is ambiguous, and not well defined,\n 2 different inputs can have the same value for the input field.\n\n for example :\n\n   input:\n     - first\n     - second: \"value, third\"\n\n  and:\n\n    input:\n     - first\n     - second: \"value\"\n     - third\n\n for both it was: input \u003d\"first, second\u003dvalue, third\"\n\n now it is:\n for the first: input \u003d\"first, second\u003d\\\"value, third\\\"\"\n and for the second: input \u003d\"first, second\u003d\\\"value\\\", third\"\n\n Closes-Bug: 1631541\n\nChange-Id: I4b3fc7e0ad4552022d44c3bdd8c87004f5e8ff46\nSigned-off-by: ali \u003cali.abdelal@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a6309683b698dc8ec41955ae1ac77b0200ff6a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4a6309683b698dc8ec41955ae1ac77b0200ff6a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
