)]}'
{"id":"openstack%2Fmistral~703603","triplet_id":"openstack%2Fmistral~master~Ie68abcb3a670b293e26d8d53f4fcb9c0d67e5077","project":"openstack/mistral","branch":"master","topic":"actions","hashtags":[],"change_id":"Ie68abcb3a670b293e26d8d53f4fcb9c0d67e5077","subject":"Install mistral-extra so we can test it in gate","status":"ABANDONED","created":"2020-01-21 12:47:25.000000000","updated":"2020-01-23 11:33:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"038accc36e7b4c1a55309ff1a5a0afe1679fccae","_number":703603,"virtual_id_number":703603,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-23 10:40:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-23 10:40:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d256e2a0c33f674c62aa18e27cc8b6b8bfbccd7c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-21 12:47:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b56f56dc6049b31b26444f419c39b32d5c0009d5","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-21 12:50:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f3e33419ad452aa14a2b24305c31b9d863b02220","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-21 13:04:16.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7b4298f22c8c076e891238f5980303cfee809874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 13:22:48.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"929ae53d0a463df6b3625aa479e1781328e046f2","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-21 13:23:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"209b698f357dea304626ab94399ef5f9b8053305","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-21 14:53:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7dabcd4a3f26f06eb420b16b90180a872455c77c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 16:04:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3114ecf3734944009f6a8d736db7d282 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/416c810327e54be2b1ca7a82b39de257 : SUCCESS in 9m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/868a98d0b72c4b66b68417379ffa8641 : SUCCESS in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d6198156e2e41a8b12e86efbca07a45 : SUCCESS in 8m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d27970a979e948b2a5ea4b96fb2484e9 : SUCCESS in 11m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/012628f9137e4324b87313a6d0edc9f4 : SUCCESS in 50m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af88ef37a8cf4505bcef840f2bc5353f : SUCCESS in 47m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aff7660c363d4acdbd9435923dd1643e : SUCCESS in 55m 24s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7ec916c6df38483da96aea9b1a1e0c9c : SUCCESS in 50m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c1c1828f626e4efa943235b2341b756f : SUCCESS in 44m 05s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0b039c021e53480692cfa7c461fd5ff7 : SUCCESS in 32m 29s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2d6f94661fe745098ab65b978b7d36af : SUCCESS in 30m 21s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/28a239cd949a47ca953d88eebae73807 : FAILURE in 42m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b1e6f030955f4407b6a76c6591e3061e : FAILURE in 1h 07m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cf12de2464e464697d2b325f5b974ae : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":4},{"id":"38f6143794c14860bf36b06ddb16cd1ce953e382","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-22 18:56:11.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"82681fbd9c1dc5594c6c1aac24a2a52bbf740cfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 20:15:23.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e15cef629b704f8cb439903f92eb7fb1 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15777efdfb1a445ea43d98f77d36baa6 : SUCCESS in 10m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e321518a18ba4b2a88233d1d44a87ee7 : SUCCESS in 15m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3d63f4b2cc342d09e30ec31a16bfe5b : SUCCESS in 12m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3278134abf3a4e7ea2db9ddee56d3d42 : SUCCESS in 10m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8b3a3d101a174ed7946d5059cb0a95b7 : SUCCESS in 50m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1807e48f3046451295048541c40acc92 : SUCCESS in 44m 48s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25e3854cf1b8462eaaa221b6f21425e7 : SUCCESS in 44m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1a657db01d8d4d9f9abade06e6d9403c : SUCCESS in 41m 26s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/556870dfd363446889ee38ed73ea2512 : SUCCESS in 47m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b43b26cb273448c781d00d77a87df34a : SUCCESS in 32m 15s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b6cdb37d9e86440c99f4a3427d67998f : SUCCESS in 28m 21s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cad0eb6502fa439295ddf7706d8f498c : FAILURE in 39m 21s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/87a6e83b362848d5ae10aa53ef118afe : SUCCESS in 1h 12m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7aab7a40afa44599811bd9ff5014bd3 : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":4},{"id":"7bd35edebe7a4882018708ba4c01143daea703cd","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-23 09:33:33.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"9773f635837b6e90dac31c2a5d91729ce7cd615b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 10:40:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8937a21a24f4c529dbff6731b643ae3 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/938003ace8f04520b0bd30f7942eb6ae : SUCCESS in 10m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6317cb8d2c34420eb5ac8cf79438325a : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f880a75df5a412e936133ebc5461add : SUCCESS in 8m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17cd0c4ed68e420186c36688de08d8b2 : SUCCESS in 10m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b1372a8887cb4e059e450cb59ab8eb4f : SUCCESS in 45m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c071f8cb81654e71aa9fde6e5361cc75 : SUCCESS in 49m 24s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccda8105544a44ce9a34c9881ddd9b14 : SUCCESS in 59m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/51cb14cbb6684e1ba6fef1060e6407b1 : SUCCESS in 46m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f02df1056f6b454c9b7797175f28d2d2 : SUCCESS in 48m 01s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/95a50fa511aa4303b3cb76b0f6214ca4 : SUCCESS in 30m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/067dff3fc2d345dcb1717ae7db084ba7 : SUCCESS in 28m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0a781dd4dc0644878518b844164eff30 : FAILURE in 39m 15s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ab62a943685d432d8549380d303e44f6 : SUCCESS in 59m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f12c6e449f164244be235f9b19766d82 : SUCCESS in 11m 42s","accounts_in_message":[],"_revision_number":5},{"id":"b99b72d8d7c3970fc5916e69e1a8dbffd8c4b63c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-23 11:33:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8c6999079df0eec2e91237cec2b9b95b7eda0c45","revisions":{"2bd8e27fbea68588d75651ad5342a8505d9009bd":{"kind":"REWORK","_number":1,"created":"2020-01-21 12:47:25.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/03/703603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/703603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/703603/1"}}},"commit":{"parents":[{"commit":"7f92cc8f5b704b5af8d430e5e7372756bbbc4d87","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7f92cc8f5b704b5af8d430e5e7372756bbbc4d87"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 12:47:03.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 12:47:03.000000000","tz":120},"subject":"Install mistral-extra so we can test it in gate","message":"Install mistral-extra so we can test it in gate\n\nAfter moving all openstack actions to moistral-extra\nwe want to test it in gate and devstack\n\nChange-Id: Ie68abcb3a670b293e26d8d53f4fcb9c0d67e5077\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2bd8e27fbea68588d75651ad5342a8505d9009bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2bd8e27fbea68588d75651ad5342a8505d9009bd"}]},"branch":"refs/heads/master"},"2642903ff4b1097cd92017be1dd3a42584d6b10c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-21 12:50:57.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/03/703603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/703603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/703603/2"}}},"commit":{"parents":[{"commit":"7f92cc8f5b704b5af8d430e5e7372756bbbc4d87","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7f92cc8f5b704b5af8d430e5e7372756bbbc4d87"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 12:47:03.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 12:50:36.000000000","tz":120},"subject":"Install mistral-extra so we can test it in gate","message":"Install mistral-extra so we can test it in gate\n\nAfter moving all openstack actions to moistral-extra\nwe want to test it in gate and devstack\n\nDepends-on: https://review.opendev.org/#/c/703296/\nChange-Id: Ie68abcb3a670b293e26d8d53f4fcb9c0d67e5077\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2642903ff4b1097cd92017be1dd3a42584d6b10c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2642903ff4b1097cd92017be1dd3a42584d6b10c"}]},"branch":"refs/heads/master"},"40ba5464c70c30b77532e61eee6ec1a2158e460e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-21 13:04:16.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/03/703603/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/703603/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/703603/3"}}},"commit":{"parents":[{"commit":"7f92cc8f5b704b5af8d430e5e7372756bbbc4d87","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7f92cc8f5b704b5af8d430e5e7372756bbbc4d87"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 12:47:03.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 13:04:00.000000000","tz":120},"subject":"Install mistral-extra so we can test it in gate","message":"Install mistral-extra so we can test it in gate\n\nAfter moving all openstack actions to mistral-extra\nwe want to test it in gate and devstack\n\nDepends-on: https://review.opendev.org/#/c/703296/\nChange-Id: Ie68abcb3a670b293e26d8d53f4fcb9c0d67e5077\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/40ba5464c70c30b77532e61eee6ec1a2158e460e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/40ba5464c70c30b77532e61eee6ec1a2158e460e"}]},"branch":"refs/heads/master"},"b3f422b78ed229bd1d0151a08b55046bc6b68696":{"kind":"REWORK","_number":4,"created":"2020-01-21 14:53:53.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/03/703603/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/703603/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/703603/4"}}},"commit":{"parents":[{"commit":"7f92cc8f5b704b5af8d430e5e7372756bbbc4d87","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7f92cc8f5b704b5af8d430e5e7372756bbbc4d87"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"subject":"Install mistral-extra so we can test it in gate","message":"Install mistral-extra so we can test it in gate\n\nAfter moving all openstack actions to mistral-extra\nwe want to test it in gate and devstack\n\nDepends-on: https://review.opendev.org/#/c/703296/\nChange-Id: Ie68abcb3a670b293e26d8d53f4fcb9c0d67e5077\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b3f422b78ed229bd1d0151a08b55046bc6b68696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b3f422b78ed229bd1d0151a08b55046bc6b68696"}]},"branch":"refs/heads/master"},"8c6999079df0eec2e91237cec2b9b95b7eda0c45":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-01-23 09:33:33.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/03/703603/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/703603/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/703603/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/703603/5"}}},"commit":{"parents":[{"commit":"d5ae8ad1bced6603e4f6b146a4288d92736fd35d","subject":"Merge \"Disable key sorting in workflow definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5ae8ad1bced6603e4f6b146a4288d92736fd35d"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-23 09:28:38.000000000","tz":120},"subject":"Install mistral-extra so we can test it in gate","message":"Install mistral-extra so we can test it in gate\n\nAfter moving all openstack actions to mistral-extra\nwe want to test it in gate and devstack\n\nDepends-on: https://review.opendev.org/#/c/703296/\nChange-Id: Ie68abcb3a670b293e26d8d53f4fcb9c0d67e5077\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8c6999079df0eec2e91237cec2b9b95b7eda0c45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8c6999079df0eec2e91237cec2b9b95b7eda0c45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
