)]}'
{"id":"openstack%2Fmistral~703823","triplet_id":"openstack%2Fmistral~master~Iffd819a8934a7973c1d8a9b7a8a9fb1780411203","project":"openstack/mistral","branch":"master","topic":"actions","hashtags":[],"change_id":"Iffd819a8934a7973c1d8a9b7a8a9fb1780411203","subject":"Make the action generator an entry point","status":"ABANDONED","created":"2020-01-22 15:41:10.000000000","updated":"2020-01-23 11:33:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3890403059ba7b334cc2d24ea3f06fb6202dc54c","_number":703823,"virtual_id_number":703823,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-01-23 10:48:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-23 10:48:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"731a15a9d58dcbc8bba38c608891abea1c602ca9","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-22 15:41:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25e9301b08b4b868e7b0f338a5c53de34ff06bc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 17:00:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7bc5b56c26de4b16a93fc09fea29aef9 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65261c1e3e8449a5b093a07a2e961a59 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32a165e451a74b7ebb1eab78f8d03fb3 : SUCCESS in 8m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62bcfb0136904ba396afbf4c8de5f010 : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac47506b4c4498dae9ecc20073a9ec9 : SUCCESS in 10m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52d95478236c4feba180d7b03f335c8f : SUCCESS in 9m 11s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4517ff8f69cd4d9881144df4d9442a21 : SUCCESS in 1h 01m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf0b695c964341e8bff9cc63c62d8f71 : SUCCESS in 59m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d68dfedbe0e244e586b44b629523f5b8 : SUCCESS in 1h 02m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/67e69d9a37184b30a31de968737db480 : SUCCESS in 1h 01m 58s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ec2522e15ffd49f1b1322ed2d914afa7 : SUCCESS in 54m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/34b9fa873aad426d87e9154fd4d4e025 : FAILURE in 33m 36s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/edc82448d9df41d8b4a89a53866c6cb4 : SUCCESS in 28m 57s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ba84f96858a548fc951c3329f24aba2e : FAILURE in 27m 48s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5d4cf56e40544bb6b77e518471e0113b : SUCCESS in 1h 05m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/236051d9347d4af09b0ba3deca879a8b : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a159fdae42a5003ee56e38917d75a05a28d95fb3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-22 18:56:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ff3f4fa64eb114a871f8a7d6698c3744db6fa286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 20:13:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8727bebce0764c77a132a361fa68510d : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffb6b9b6c82b4d2690d29f52c0f5cc6b : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3028fdeb49a64f3bb21fbb4b581e61ff : SUCCESS in 10m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0895c4cfc07a4822a1964d51bb3c91f5 : SUCCESS in 12m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5349ba531c6c426badb5e9f147931085 : SUCCESS in 10m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4feb3f4892a4f71893e6464427c4811 : SUCCESS in 10m 50s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/54b0451efbdc4aad82e82bdf84ba23dc : SUCCESS in 54m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee074b46b757474ca50d313504de1b95 : SUCCESS in 1h 01m 18s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80781d92e71844c5a8673bc712aad251 : SUCCESS in 46m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b8f396b92e7b4627925e7a0cf7e4c6fa : SUCCESS in 45m 15s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/53af62ab492941318b1c14258fcf9c3a : SUCCESS in 42m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/43a6049a69594050b1d92bf127231b2e : SUCCESS in 30m 52s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0031182fb57548ef9f1c3e03838641a0 : SUCCESS in 27m 45s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a300da03ef194f6082c06354ce6aed70 : FAILURE in 42m 31s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/3ec2d6d550de40a9be70a5971444c3dc : SUCCESS in 1h 05m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb6fdb1c72de4b20a5db264719267b6b : SUCCESS in 11m 42s","accounts_in_message":[],"_revision_number":1},{"id":"32380426694b596839f06151a8c3cbb45ab0457a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-23 09:33:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"149a90762a1254c8237a1a0a21849078de71afcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 10:48:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/454231be06814a0eaa8895b8cf6536a2 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082cce07c45846ff8d3b6ade7a3cf2ab : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a11893f34ce547879670c8f41211dc4c : SUCCESS in 11m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ca4d2b4b20546cf93d4a85166c4323e : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acc456e3a2d14f44ba229b0fc82553f6 : SUCCESS in 11m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d595d1c6ea5e4be18e795b5c52287f44 : SUCCESS in 11m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d28f3f6030c64a2fbe1ac932968a67ba : SUCCESS in 45m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/637f7e4805c247e1923bbb4f84e65846 : SUCCESS in 46m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b5881d1fbeb42b6960d2700bece638b : SUCCESS in 46m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fba7b567ae63487694c5dc839f61559e : SUCCESS in 46m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/626a69d23b274fdca3ecfb8d75c60248 : SUCCESS in 43m 24s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/842ed9cd0dbf48018bc1e027df6d1b96 : SUCCESS in 30m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b41a0e739fec4a5d9d0333204e217270 : SUCCESS in 28m 50s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7777fa7e86d94a6abaa978791fcd53f6 : FAILURE in 33m 46s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ab79a5c09ace43d08940f8ee7ee454c5 : SUCCESS in 1h 06m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7be94f6427034e7fb2e0bcc20831f2de : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":2},{"id":"0ef713140ca490e89863bd41361107c1e3c0f34f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-23 11:33:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2d4fea504248ed099442d92e5f99130b092fafbb","revisions":{"8527f472e95e44e4b82eee6fc4f8e86095e6b312":{"kind":"REWORK","_number":1,"created":"2020-01-22 15:41:10.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/23/703823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/703823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/703823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/703823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/703823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/703823/1"}}},"commit":{"parents":[{"commit":"b3f422b78ed229bd1d0151a08b55046bc6b68696","subject":"Install mistral-extra so we can test it in gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b3f422b78ed229bd1d0151a08b55046bc6b68696"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-22 15:40:56.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-22 15:40:56.000000000","tz":120},"subject":"Make the action generator an entry point","message":"Make the action generator an entry point\n\nThis way we can decouple the action generator and move\nit to mistral-extra\n\nChange-Id: Iffd819a8934a7973c1d8a9b7a8a9fb1780411203\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8527f472e95e44e4b82eee6fc4f8e86095e6b312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8527f472e95e44e4b82eee6fc4f8e86095e6b312"}]},"branch":"refs/heads/master"},"2d4fea504248ed099442d92e5f99130b092fafbb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-23 09:33:33.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/23/703823/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/23/703823/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/703823/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/703823/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/23/703823/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/23/703823/2"}}},"commit":{"parents":[{"commit":"8c6999079df0eec2e91237cec2b9b95b7eda0c45","subject":"Install mistral-extra so we can test it in gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8c6999079df0eec2e91237cec2b9b95b7eda0c45"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-22 15:40:56.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-23 09:28:38.000000000","tz":120},"subject":"Make the action generator an entry point","message":"Make the action generator an entry point\n\nThis way we can decouple the action generator and move\nit to mistral-extra\n\nChange-Id: Iffd819a8934a7973c1d8a9b7a8a9fb1780411203\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2d4fea504248ed099442d92e5f99130b092fafbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2d4fea504248ed099442d92e5f99130b092fafbb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
