)]}'
{"id":"openstack%2Fmistral~703972","triplet_id":"openstack%2Fmistral~master~Id62fdabe7699e7c3b2977166e253cfc77779e467","project":"openstack/mistral","branch":"master","topic":"workflows","hashtags":[],"change_id":"Id62fdabe7699e7c3b2977166e253cfc77779e467","subject":"Remove OpenStack actions from mistral","status":"MERGED","created":"2020-01-23 11:31:32.000000000","updated":"2020-03-06 09:10:57.000000000","submitted":"2020-03-06 09:08:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703972-1583485704893-99049c42","meta_rev_id":"9347282192a760637b487966d8b83237f954d0be","_number":703972,"virtual_id_number":703972,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-06 09:08:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-06 07:08:08.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":1,"date":"2020-02-26 08:12:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-03-02 10:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2020-02-26 08:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2020-03-02 10:51:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-30 10:09:06.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2020-01-30 10:09:06.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-01-30 10:09:06.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-02-04 08:49:12.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2020-02-25 07:58:55.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-02-25 08:09:39.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-03-02 10:51:19.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-03-06 09:08:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6694c33a22524e9e9da4be283f1d2b717d042f9d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-23 11:31:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1aaf1f1fd4ac7c3e2cfd776661443a10d0bae4c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-23 11:33:14.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d470109472f6617d1979c5019806d81ccce372d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 12:24:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66754048354048818c8cc5a637e04570 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f01540a89f84cf28fe971acf39c6c3a : SUCCESS in 11m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/944ca00d0306467cad0404ccc902c0e4 : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b6061621f0340c2889dd6eca7b07b3d : SUCCESS in 10m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be2fdc2c18984d179ff1ee92d2771529 : SUCCESS in 12m 54s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/adef142073bd466ea9b65e67df6f68f6 : FAILURE in 48m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/518c197dbe5d4987a3a89be92ffce4d5 : FAILURE in 37m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e66d17bbbbd418897774b67ba2c1ebd : FAILURE in 40m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1bca2004980a429cb44609caef8a4fa9 : FAILURE in 34m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2900ad1b06894b66b4a44b872ee6b27a : FAILURE in 43m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1284f5aba581485ab1a851d96b90ad05 : SUCCESS in 32m 57s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d6e974c8e1ca46f2bba3b1029fc1d283 : SUCCESS in 29m 11s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1cab47d5fd6f4b32bb7e9f7f8e5afa07 : FAILURE in 42m 44s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/24b50e0ef0b048a88aae14f3c7263c9c : FAILURE in 31m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae38959535de476b8cb67fb744ce9809 : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":2},{"id":"025c788eecd2050ae3b32730ecd9f3f89dbb0ca0","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-23 12:44:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"75aad7373f83d4c008a9e769bb8753e087952276","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-23 13:56:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b2fecc33012b8b65efaf167966c9126da7c988d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 15:09:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7998d6c94c4e68bf19b0c7bef28795 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/751ba4a7fe9d48d2a213d42c6b71e146 : SUCCESS in 9m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/742a365016324dc69efefa26ea74dc8d : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56509391d2a34be096fbc4b69097d0ac : SUCCESS in 9m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6b2970dc2fc41949dea679eca9761fe : SUCCESS in 10m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e712e93ff3db4c7bb06e5b6d15bc5a76 : SUCCESS in 43m 18s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b10a4b6e397b409fa5e894d80f29c5b6 : SUCCESS in 44m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16bd2c27e3464432b9e09898346fb5e3 : SUCCESS in 56m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9126bd430cee4edbbd85bf0d0c08d4ab : SUCCESS in 50m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/abd96b575b7d404db5a253d6f80bff5f : SUCCESS in 59m 01s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/04d54d406d124beabc5eca02ad25811f : SUCCESS in 31m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7ef5df723e3f482bb8c6bccf3f14703d : SUCCESS in 30m 17s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/624d02a742c049af912e5e30eb135dcd : FAILURE in 26m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ed2d7b6ed9c944ad92de909f92c71aba : SUCCESS in 1h 06m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a58f254107014092bd84d8dae8d5d0d3 : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":3},{"id":"9f6295f0a395f83fde7382ff039945c9fd767201","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-26 07:23:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"23d5e76780720d563ce9249dfcf569cec7e0e539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 08:24:14.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"f727fe2e4989a99d0fc2d3c4f80e0d9d20f04470","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-26 08:24:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"8066ac31c3cb5ac2cfad7d9a7a5adf4ac137bb54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 09:08:16.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe2ddbb262fe4e2c8b7955374d4f5bea : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2502721dfb954961a87a46fe8e167a11 : SUCCESS in 9m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/136015913ef443188c596e2a40e32663 : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a274c451a08545a2a76eeeeb29bf047c : SUCCESS in 9m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d5c4edb36a1423ebb085c2fe4afe2f0 : SUCCESS in 7m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86e08e5a1f3342f5a000d6feb7c0a30c : SUCCESS in 8m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f0fb4847e1a649e6a1c47d8178898e68 : FAILURE in 32m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f04a5694aa1416599aa84b65a814b25 : FAILURE in 36m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d39b3c3b078f41a89412fd9bf9376735 : FAILURE in 32m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/483b299bf6ad48ddb09f887ed78382a3 : FAILURE in 33m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/09f2b56b63d7494e8331bf8137bd3b74 : FAILURE in 33m 24s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/079b5ca359b94cb49573cec19524eb90 : SUCCESS in 29m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/92bfe36b3ba34d58b547731c6016cb72 : SUCCESS in 27m 57s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/da289fbddf9b4367be905d03246d4c73 : FAILURE in 28m 50s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/574f5a51730c49d4812c285bbc1dc1d0 : FAILURE in 34m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffe7bef6bb1d48f6ba909568c9832e23 : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":4},{"id":"5bae4dbcef8978e3c05e3419bee2474c65ada116","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-26 09:12:35.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bdde8f152ce352db4cfefa9cb35af02465be99bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 10:21:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/286c29ee7b584d02bbb9673d8ba680e4 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c74bd8604036418287eb148d443d7da2 : SUCCESS in 8m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0fc504f33e954319acd039b0b3a3a36c : SUCCESS in 9m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eebd561c74f34b76ad1e62ffeb81a1eb : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c6659ba6d4b43b68cd9811a67f4e7de : SUCCESS in 8m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ec47745cbea423ab648aad9eaf37f61 : SUCCESS in 9m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6a3a15bc26e2476cacb78391dca84a79 : SUCCESS in 44m 30s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcbfb4bcde0946329a259d8e3f22706e : SUCCESS in 45m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f949e8d91964658ab80e5d3821542fd : SUCCESS in 44m 41s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0b375ce60ec84647afaa9da1f3ab3677 : SUCCESS in 44m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1036a4c5f7f9477d866641da82c07ae3 : SUCCESS in 44m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bb50eccdb44c4875b64c127f5c2f9687 : SUCCESS in 29m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b4cb3f9b2ba74e40bef7beead823bae9 : SUCCESS in 27m 03s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/37d426e9d4c44c53a0037aab90c1b8d1 : FAILURE in 28m 13s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1c1063b0b5d84aaa938ed5c7cbb62c75 : SUCCESS in 1h 02m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/985ed0d3174f4a6d8a6389954aefb8a5 : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":4},{"id":"0c8107871d58866539998d5936fe23d13439ed5e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-26 13:03:18.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"09d1e2d823bfaa8e3b9cbd09b1e0fb7d6b7602c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 14:07:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2923fc0cdaf4179bc81d6890c6727a5 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9335efd712054def9e2ac04a008975c1 : SUCCESS in 7m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03e80751689944fa9b9fc6d68a8bbd36 : SUCCESS in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c2f6ac509c5427c9532d83e22c59f7e : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1b87b981fe34b59a6b7b0ffb8c23c59 : SUCCESS in 7m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48ceb22e976b43d2acb97d6565f02a51 : SUCCESS in 8m 18s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/250a9632ea684f6198ca51ce912e8557 : SUCCESS in 43m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35bee743bab54d038b559ddc725df28b : SUCCESS in 45m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7effd11f02b44427a43a5f26ea3ff05a : SUCCESS in 43m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/24b4b47c55b548f08f87995a5df087e5 : SUCCESS in 45m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c1c867d06aab43d597d80147853fbae5 : SUCCESS in 41m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9d6d721319f741c09cade24eac900b37 : SUCCESS in 28m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/00a628377b514fc5a6e441183dc89625 : SUCCESS in 26m 47s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e18bcf1280c4438f8fe88ce68911dbf4 : FAILURE in 27m 41s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1c615ff6b1664ba99f6aac05e60c4504 : SUCCESS in 59m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ace8c2a991774c4587f4a9eb147156b5 : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":5},{"id":"41e29629ff38b3d859e375f3cc7c9f37c52fde7c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-29 15:31:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"03845609fa4fe83fa12e7c1a0b068301e287504d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 17:02:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08fe690e0e974ed89e3548b15589b42a : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/886cc0792cdf4eae98090ca92422590d : SUCCESS in 9m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b1560710b6b4809b7cf60e81e1c42dc : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/321600dd9d3a427eb4418f3f157229ab : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4972284869e64a239d54a5926c5a18bc : SUCCESS in 8m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5b684785f7d74afd818aa0b5366529d3 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a92929f254eb48c3a0f4d3d3d0c56aca : SUCCESS in 8m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4451cc23bf1c4454b2a832ab84ec1a28 : SUCCESS in 10m 53s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/53fe31f2828843bbae9c3d1ca0ca7250 : SUCCESS in 48m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3afa3e5c3ac4e4dbeaa82e64af63b74 : SUCCESS in 53m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc4a8f3de13b4e61a91fdbfbd66d6df6 : SUCCESS in 45m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b6e86a9e5f3f48b9aca85f60160290fb : SUCCESS in 46m 24s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ce042e7f1ac149739cc93a771b202e14 : SUCCESS in 42m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0d19fc45347e4caa8868cd4d8a83e32b : SUCCESS in 30m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4f5b9d06b9b543409878fa9dc849fa6c : SUCCESS in 28m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/da323b1abaa049cf867d2f76e1b5cebb : FAILURE in 34m 00s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0314697e3d11467b9a23b5306f4b4711 : SUCCESS in 1h 16m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcbf8672315741c7a5a630b47e5b9557 : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":6},{"id":"fec310e95f025900d12e303bd2e1602031e6aa9f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-30 09:13:46.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"268785d5f6d72c46484af7d25ce31ab789ff73f7","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-30 10:05:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5ede19a072d57c43562c89161d6c5da60e04dbc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 10:05:58.000000000","message":"Patch Set 8: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/mistral on branch master.  The error was:\n\n  while parsing a block collection\n  in \"openstack/mistral/.zuul.yaml@master\", line 76, column 9\nexpected \u003cblock end\u003e, but found \u0027\u003cblock mapping start\u003e\u0027\n  in \"openstack/mistral/.zuul.yaml@master\", line 117, column 14","accounts_in_message":[],"_revision_number":8},{"id":"2bdf1c17c46ee135bbe48fdb59c3a6675d752e1c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-30 10:07:01.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7c89d401f22e456a6eb90c2bc9bd3b3bdb499729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 11:19:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28e59abcedc54ccd88ed0e55fce8f467 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8933ac3a4ee42f381d82334f62e9579 : SUCCESS in 8m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9970e88742254461aa04cfbfcbc584c8 : SUCCESS in 10m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd8e26eda28143339ffafabaa458d69b : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57704270c26d492b9b3a7aa35f8198cc : SUCCESS in 7m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e645cb47707d42618f38d4b2d281358e : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70d07f2a04494bb8bb649fcc378c13bf : SUCCESS in 5m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99df9fe261304402af5b30abfa86dbe6 : SUCCESS in 8m 41s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/55bf0bece5fa4852b16e976ced93e1e2 : SUCCESS in 1h 02m 34s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa2845b81d15475b83f55686d3fb01b1 : SUCCESS in 51m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89ce052627744222b5c98effe39cd2d3 : SUCCESS in 50m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c39471bc030d4600864e497a3b29f210 : SUCCESS in 1h 00m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bcd0cd34da7d41a580b40878929f8f46 : SUCCESS in 47m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/52711b6afab744e1bc32cdab8c5c031d : SUCCESS in 29m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/df5d3f6ae0634f0c854b4e5d83a984be : SUCCESS in 27m 48s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1e572c26f9d7438da60f4de08d622545 : FAILURE in 32m 11s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/05e4ace3e96a4251a2054ef449e5ca13 : SUCCESS in 1h 06m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1bb3ec4a57a4bd595758c6f61f917dd : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":9},{"id":"863640d33c06d5d03b0f0ae55c6779e1748c65da","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-30 14:58:16.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"4567c6594f45b9e806cf2eb085e36d6bdec64df1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 16:16:16.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abee4efd126947aebcd5a22cbbf69ebc : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/634da9e909334025b649a3a34fc40baa : SUCCESS in 11m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c974a520660404283698ee42b8c9552 : SUCCESS in 11m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12017a447d73490184d29ecf0aa721f3 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/592af1c75d5546be8980bab2b9933c98 : SUCCESS in 7m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ad1dce78c0374f6fa84e5d53bffab7d4 : SUCCESS in 3m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/113a9a489c09487fbe168f8280cf6f72 : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54865dde4c31423b8ae9ad64807793e4 : SUCCESS in 10m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/327b7ce016d04462a31f68c09ef20606 : SUCCESS in 45m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09ee08315e5b4e049573764f5814d9b9 : SUCCESS in 53m 34s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f569daab03d42fe828a57b26bb7147d : SUCCESS in 52m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d0654be4fcd84b46969f55fc19c6f7d7 : SUCCESS in 48m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ecab4f19cbe44df093f5e07a1cdb1391 : SUCCESS in 51m 54s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2e0ee29f931f44fa8fe2748d027173f3 : SUCCESS in 30m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/17b69228764348c9911135f4d0b907d8 : SUCCESS in 28m 49s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a04a636e85e042ba8a5eb4a90fe85d5e : FAILURE in 32m 19s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0532704afaf74ccea9aaa74f64768239 : SUCCESS in 1h 08m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d6b87d00cff4b9c9030f35e8f6454f1 : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":10},{"id":"e316fee938a799da1629d988a80698f67881eb67","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-31 09:08:43.000000000","message":"Patch Set 10: Workflow-1\n\nLet\u0027s not merge it for now since there\u0027s an early feedback from some users who are going to make too many changes on their side due to this removal. Let\u0027s discuss it first with them.","accounts_in_message":[],"_revision_number":10},{"id":"f00e33a69ac0bf3171e5a0c51d9815b882d160d4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-04 07:35:31.000000000","message":"Patch Set 10: -Workflow\n\nWe\u0027ve made some more checks, It is ok to proceed.","accounts_in_message":[],"_revision_number":10},{"id":"00e832160e7c5bdf8b324cd6e75fa04002557c98","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2020-02-04 08:49:12.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"459fe148e0e3304fdb6a74bff817d62dd49ed37e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 09:58:16.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbc454252a704395bf297647f232fb74 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/096bee4fbe1f49e8a0bfc723e7f4040d : SUCCESS in 8m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f38c5c7d886b45ccb544556f4e49c204 : SUCCESS in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0404f2fd6a547c2ad0e403a25dd6dc4 : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c39a5d1bfd7a4306ab923acba2b707a1 : SUCCESS in 7m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f7a84dae8b9546cf9c95176de7ce6993 : SUCCESS in 3m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72f54a5d73b24788b22c248b6c0f9a1d : SUCCESS in 4m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfe51100076f4f658f9f3a341d51168d : SUCCESS in 9m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/57b5f5ae8c354c67b37c768bca382352 : SUCCESS in 53m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94f02127678a4211aee5287567337165 : SUCCESS in 46m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b95525df486f4a68bdee08e49a1f02d0 : SUCCESS in 51m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8f554753a0aa4adfb7c71071574d70cd : SUCCESS in 51m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a4e1d38f1ca54f8cb30983151603fe66 : SUCCESS in 47m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ec58890686ba40c0bbd062765b269713 : FAILURE in 29m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6ed318a60e3041359460e17dec2c39a1 : SUCCESS in 29m 25s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/01df24513a6248169910373ce147bf60 : FAILURE in 28m 35s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bcba17e8e881421c8ca24719d43c40fe : SUCCESS in 1h 02m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7fd351f29934acc8835c040f787b525 : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":10},{"id":"099bd697faf1db6eaea9e33df1ad770b6658b749","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-05 04:30:32.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)\n\nLGTM. Just one small question.","accounts_in_message":[],"_revision_number":10},{"id":"bea2ff4f167a4c41b9fd0aaff52e9ec55db2acb2","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-06 06:29:49.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"a8bb57942f6afba78b1aff790e146fb251195cec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 07:43:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd02753e2434d32808db0391968359e : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f69350ce1ed84300a099f32883ef9f56 : SUCCESS in 12m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/459e8f707c494ed79a666ac2a88553ac : SUCCESS in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/471ed63e39554e8487b94ea96a509a50 : SUCCESS in 13m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4eed5ee61d942dcac009b3300b101fc : SUCCESS in 17m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b254e2ce73bb4229bd424639d52c45cb : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9506a4541edb48fc91dc49c171377e41 : SUCCESS in 7m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/068c63a3eaa847669186b83189661fc4 : SUCCESS in 12m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/be3028b1b96e42f7b122c25d68b6b7a3 : SUCCESS in 55m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee0fcf11eac842638689465d9704fda9 : SUCCESS in 44m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c227859d0ee4f4bafb3527ad8b39d7b : SUCCESS in 47m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1ba00feb53cc462ea228d9dc8304a615 : SUCCESS in 48m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fc4d2486c5464e3695b50d6415304084 : SUCCESS in 1h 08m 11s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/18129a495b974bc3a36e0e41a8d67b7f : SUCCESS in 31m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3d781f2a17034db0a2095e064c3a9519 : SUCCESS in 32m 14s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/05ba39cc12a74ca6b51ae1b7a60f117b : FAILURE in 47m 15s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ccc46b0b44e74abfa8617bd436698ace : SUCCESS in 1h 05m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be20180a85e447b5a2da1398db00fcd2 : SUCCESS in 15m 07s","accounts_in_message":[],"_revision_number":10},{"id":"3f16dc2ee0f6c180420b4b5e5e8a4ecc790b25f7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-06 08:03:29.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"3c6e826abcafc5778e4055f191e9fe898dab051d","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-02-10 12:58:22.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"ec647a174d457ed1c5fa06496150b1036889ccf1","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-02-10 15:22:34.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"097ebd1fa13bb18d716cf4fdf7c8f2772b5fe520","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-12 09:54:44.000000000","message":"Patch Set 10: Workflow-1\n\nwaiting for mistral-extra to be packaged as an rpm\nsee https://review.rdoproject.org/r/#/c/14216/","accounts_in_message":[],"_revision_number":10},{"id":"af8f3fa09efb4ebec3aeb9d942a038a0c4f8d88e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-18 14:29:48.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0e88586b8e63b83e8b881136ff84bd60e6418ff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 15:38:59.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7c05b204ea94641a4615d860abd03b9 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b7b611768a34a6bb694f932c308e24c : SUCCESS in 10m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21e10b4aeb6445c786f4f41fa8da6fa0 : SUCCESS in 11m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8d0a2e7e0154653932d596e77d64cf0 : SUCCESS in 13m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d60c17539e5046609988e02f410c2793 : SUCCESS in 11m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a24881a52514c50971be4016cc9ab62 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bafe3dfa704f44938abcb943bf2def38 : SUCCESS in 7m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5274655bc462411badec64cbd5819f5a : SUCCESS in 11m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c97de8f3f75d4e64acd8edb6f78acd6a : SUCCESS in 49m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1da11ceda76442b8b977a93d2939bfeb : SUCCESS in 48m 37s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8006f90489a43b49df792a227e69093 : SUCCESS in 49m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/49158f18ea9448daa2c813f764715790 : SUCCESS in 47m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5130f0b9cb794630ad35cbe18c21042f : SUCCESS in 45m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ab1a991554fd4592b013671fe6f918c9 : SUCCESS in 32m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/51df5d8dbbb34abdac41f3a2f029e818 : SUCCESS in 31m 53s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0db16dc23a1b4188b77f8d3af3214268 : FAILURE in 33m 46s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6d010830e0424b73a32f3b2b6aa8b5c2 : SUCCESS in 1h 04m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc52456d61fe468dac6884e4d80af784 : SUCCESS in 11m 52s","accounts_in_message":[],"_revision_number":11},{"id":"90eaeeff5cc620b29621b13ea37c4849625fbe97","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-19 08:51:02.000000000","message":"Patch Set 11: Workflow-1\n\nwaiting for the rpm spec to be pushed to rdo","accounts_in_message":[],"_revision_number":11},{"id":"53d2b1869e04a2ff9ebab001b8ad800672c8ca5c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-24 06:38:46.000000000","message":"Patch Set 11: -Workflow","accounts_in_message":[],"_revision_number":11},{"id":"f68eccf641fa1f48d109479e861c843046fcc87f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-25 07:58:55.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"9757bf120bf168272086f3deff70c6b184ed146d","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-02-25 08:09:39.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"d2c4d09424d77bc07a77472140fde41b643d3f3a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-26 08:12:59.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"7e3e8e93e8b3fa17bd5f548bccde7371696da5a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 09:18:17.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b2469211c60436bbd45a55f001af9c2 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcb1a0d3133245b3baa3cc4a0adc8b84 : SUCCESS in 9m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1abe3ae263da41039f8836b8c32b9299 : SUCCESS in 9m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdc7aaef412b479ea2b5ad5cc249451e : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac13c5301cf45129a9fea39ab87453e : SUCCESS in 8m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/230e1e6437684b06b3f4d2e90283080c : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edae6dcf3aa144ef9f517aadd6e3f79b : SUCCESS in 6m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0384aae62f940d99cf96b5f795a0486 : SUCCESS in 9m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/559f799bd59a4e088f88bc3efe66856c : SUCCESS in 52m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/421b4c53932041c2ad2c23671c2213d1 : SUCCESS in 44m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5ce18a758094eedb73ba2919ced229a : SUCCESS in 45m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bf0abda5be55449b97b8a44cf43a4683 : SUCCESS in 43m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/401903b77fd14484ac17f279ef2da68c : SUCCESS in 43m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ef13a03d0a3b4708b918656a9f4cffb8 : SUCCESS in 30m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f11370bf6cc54aaba8235af8ecaa59af : SUCCESS in 28m 16s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/649b9c161b2b4a4e9cd97e31ce73a63a : FAILURE in 33m 13s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2e60fa895bab4367a953b6559b9a4e91 : SUCCESS in 59m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fecb44cdcd6d406e98b80bdabc63259a : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":12},{"id":"fcb1fcbb5c14586e54ba7df09e366b30c875a037","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-03-02 10:51:19.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"1399d74c8136637f71eef938c7d1bd58d3cc2ed8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 10:51:39.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"23e7805ff83ad711e63df74ad8289b19997bc1d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 13:12:33.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e032072cd414ea3b2a2c67068e80bb6 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/881966d9e3f7431db55b10e3432e106d : SUCCESS in 10m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f7dc27a9bf74c218837235cb86cd823 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55492be13e064e44b94a2d4162699e32 : SUCCESS in 9m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/32cc48144e324420a904b68f9b86e10a : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cf4d150bce34eb0818e47c0baca0a1f : SUCCESS in 6m 23s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1eaa7a6e6b3c475b961ff93396c3ec81 : SUCCESS in 47m 29s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4aacd03148e149479fcd71ca8398f4b5 : SUCCESS in 42m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d8675f06abe4ba49d4d0676b1a80e92 : SUCCESS in 43m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cd1913b6b99e4126ae93e739e1c1a358 : SUCCESS in 41m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b890130b25d740948a498fc9e145a43c : SUCCESS in 30m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a1f8417252e04d94a72de9379dd056b3 : SUCCESS in 30m 42s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f78c2727a7c944b693bde3fbeb2bb93e : SUCCESS in 47m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39adc2c73aab446d8e9b5ae1c9f452e3 : SUCCESS in 11m 57s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c15e3212f2d7473da89a6834ecb573bc : FAILURE in 26m 47s","accounts_in_message":[],"_revision_number":12},{"id":"9a79bb62e8cfff9050f1b4f40ecf1d9ead74f69b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-03 04:18:10.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"e114acd22a785cff4627344ff309fe8ce055478c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 05:33:37.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96652223d21a4730bb46d0dee66e75f3 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa9853939de6422db0d639febbf911f2 : SUCCESS in 10m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb7acc8d741b46d7bada5ebe0951f028 : SUCCESS in 11m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecd5d61a5e7e4b31b9ae11103f297f32 : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa9381c98220499da60e0f52057fc2df : SUCCESS in 10m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/45c320e03fd74f53bbd21736f58c469e : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e647830c8274908afff8275ac43b258 : SUCCESS in 7m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9634295081c348658d62298a37f59ed1 : SUCCESS in 11m 13s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9671ce9c865c44928408d42595313911 : SUCCESS in 54m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20848556ac1949a28614413aa2f1ba46 : SUCCESS in 48m 01s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31b9e67013c64bfaa2a82a9d5df354e4 : SUCCESS in 48m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/016c430e98c44cd1a97614321cddf4d3 : SUCCESS in 47m 22s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/46fdd2cc58e84e33a85ebd3c3c2e6709 : SUCCESS in 49m 52s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ffa2ee0dd35c4a1b8a54d328990917ea : SUCCESS in 31m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/53d647f78de248478e40c435b22b9596 : SUCCESS in 29m 48s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/352e2a609c4a42e98d650f72a1bc51cf : FAILURE in 32m 41s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a4c46e172da44c51bdd2bf0eeac9c7bc : SUCCESS in 1h 03m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fb7a90f8d7643b498b51d8117d7c093 : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":12},{"id":"d9828e0d1b6feac16c9409a32be380b40089b406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 05:33:52.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"bff2bf09f8e1cfafb307153d82abead14b4cda06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 06:27:29.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54f93f5046804609bfe8953437da0109 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea30d3d56b7a4692b816638cc03a464f : SUCCESS in 9m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/466b3b6d3e1e43d486b80aeadc6cb8e9 : SUCCESS in 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58648061473546978861caab3ed2c7bf : SUCCESS in 9m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/05a4b8ff68ca457badc606f629f09470 : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3450d2f951046279c06ac73a2f4cc36 : SUCCESS in 6m 28s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8fdd74a6ed714d43b14c2ffe7f0db2a9 : SUCCESS in 43m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae72ba820af34170a523c9dd59d3656d : SUCCESS in 51m 01s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4a61e6ec1ef4c239068b789a6d516b1 : SUCCESS in 50m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/acaffffc808946288db703031f0d7a56 : SUCCESS in 51m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/282feaddff4f4b6ab0e0cb591efde60a : SUCCESS in 30m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9c656420a1d24061ab6014b8e31d35fe : FAILURE in 32m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9ea3c17314a04b29a65fd90c271b9159 : SUCCESS in 42m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fea3e5e502064bc49af20b2e117b2935 : SUCCESS in 10m 41s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/5003879f533c4e39aa0e9df98e0901b9 : FAILURE in 28m 12s","accounts_in_message":[],"_revision_number":12},{"id":"7db97b3bcba651c1a730e66e518391a5fb34918b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-03 08:33:55.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"3cda3f648925325fb242146e3412815073625ae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 09:30:37.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6b43429a0dd4e0db00edfa6011fa055 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a3f743cb7f2494eaea37fb72f0fb5c4 : SUCCESS in 10m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a68e01b3795847d9be4cbc5931f1fc24 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c41c5e47716e49b3bcf8515eec42e211 : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ae5918677d45bcb7ed9caf07869074 : SUCCESS in 8m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1a41591a7403460687e0c0dff11498ca : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4686a64b847c4eacb9c7c28d8793d7b2 : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fefc71e4fa04a6c953d137e2cdd1482 : SUCCESS in 10m 32s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7c98511ecb7d4b31a338f7a3605632b9 : SUCCESS in 45m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d414bc57e9140859de4f6440b58e82c : SUCCESS in 43m 21s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c976d04a83d4952bd3071f6128bd9a4 : SUCCESS in 44m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/729a70514a854360a0b0f553531d5454 : SUCCESS in 40m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/df5a962d2ef84762950aa6165a128155 : SUCCESS in 44m 54s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9d779a1898f54c6280ed6c21fea266d5 : SUCCESS in 30m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/bd5e3bc73fd4410f89649bb8c8bdadce : SUCCESS in 29m 04s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/449e76b1ba574665b8b848c8a5aff2a0 : FAILURE in 30m 10s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6e3c8a96b33043cb8cbe2b20a6eecbb6 : SUCCESS in 55m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c0ddebfd1814644acc286a317ffa891 : SUCCESS in 11m 00s","accounts_in_message":[],"_revision_number":12},{"id":"36a271bbf50c4855591776a20dc88ec48154dfef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 09:30:53.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"a747f3172b86821dc6d7834d0eaedc0874f6e1db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 10:29:46.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd02a7c2d12c4f7e9a35a7bb09169f36 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd27e6d0615748279a1dd16152efa4ae : SUCCESS in 10m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8dc94687805146638c24367fa36b5ced : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8b5919209954233be5a6ac5838812dc : SUCCESS in 8m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e5955b6ae66a4dd88eaadfaae224e5c6 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17d0635eac774cc292cff60757cc1aee : SUCCESS in 5m 58s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3b64846476f74a28b0daf40c3b6491c8 : SUCCESS in 51m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bff54b7692ea4551a94989ccafe7d5cb : SUCCESS in 42m 36s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e080c4fb962d4055a5ecdbdd0c642d12 : SUCCESS in 46m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7f4ba4f29c574d888bedde0caabb16e1 : SUCCESS in 44m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a5ac71938cd440ca9da699d46798fb8a : SUCCESS in 30m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c1395d633a464158bd6178f8734cebda : SUCCESS in 28m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f32240a17d484486b36143b94b02150c : SUCCESS in 45m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dde2ac7c8f44e388a62bd8963022d3e : SUCCESS in 10m 44s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d2983290d08f4ecfb04534099fa3354d : FAILURE in 27m 26s","accounts_in_message":[],"_revision_number":12},{"id":"17a898d8f405802b4605a1d32a7fdc93ee873655","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-06 07:08:08.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"a56602a82fa0b761af6acaf7e52267d2d022be0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 08:12:00.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a34cddcc5be4c6995e6b4c501ff3709 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a39343b26ff84abe944c667f7da0f215 : SUCCESS in 10m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2870ff4c88040bd8d4e751a09e6d5d7 : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5302f798bb744142beda56dca31979bc : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f4a7ab16a284e77b8ab94d5eefb89a9 : SUCCESS in 13m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d8655a615e714104b22377e5424e956a : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c395e72f10cc4ad0836c1b79e524f04d : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3922e463c30745f994bb07e15c39b96a : SUCCESS in 9m 11s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/59eb98c4157b402ea0dd011d7bc8d4d9 : SUCCESS in 41m 21s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4111e681738e4540b937017ed00f8843 : SUCCESS in 44m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7090ad2b5bff4300b72f450027629242 : SUCCESS in 43m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cd9123ee8e6643218f6ad4faa40e4d76 : SUCCESS in 44m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8e24e6d2021e48679808d1fccd11e965 : SUCCESS in 41m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a79a06d93b9d4337a673655c21eeb972 : SUCCESS in 30m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/930ac247e26c407db7eefa6d08c79ef0 : SUCCESS in 28m 16s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/75f28a793b464c5bb2e74cf9ac93cbbb : SUCCESS in 1h 01m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa1ac501042641ae8ad2f91c1974a14b : SUCCESS in 9m 41s","accounts_in_message":[],"_revision_number":12},{"id":"e5a77134ac9548c41d5457594bc9f6b7b61b2abb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 08:12:16.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"e2f7993562c13dd5c6a5628e200c56aecd016490","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:08:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"e07d88c8cbbefc3ae39507ef58eb0146d81eb8d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:08:24.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd011b8327d4bb996149704862a9d17 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87d44c3524e242c886e96b7a7f860a50 : SUCCESS in 9m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91991d1a6a7140b5ad45de3cb0a52316 : SUCCESS in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a83d776b13f54ea991174aff2281d441 : SUCCESS in 9m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/082004774c28404a87b6afeb7c5358af : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a069e85c74b4c90a1d88885f4337d3b : SUCCESS in 5m 52s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c4fd41dca6a14dd892fb62750a1f9845 : SUCCESS in 46m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28cb0026ac744f629f70f7b1e9af6916 : SUCCESS in 46m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9e85017529f4245b193163c55d0f87e : SUCCESS in 43m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/14f9eaa7fc1647b091a8701b82261b55 : SUCCESS in 50m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bc4fdcf6246c42db99471b6b0eea2810 : SUCCESS in 31m 36s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7de7942da9d64b2fa0c5638695a887f2 : SUCCESS in 26m 45s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f854818f07794f2a853ffed9337b032a : SUCCESS in 39m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/002bfb8a9ff94489a7edf8a3e44d9f2b : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":12},{"id":"845aa5c35d076944ef81d7e81045632a9d3748b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:10:57.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e75f6063b21841b58e25c149174df5d9 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9aab2728ecd4a3ea3c8deb17fef0b7a : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"8bdf341af7348d3d092c6523a09df118c51c2a53","revisions":{"357bf55e35a7a50234877edb318d5ec3d0e056f3":{"kind":"REWORK","_number":1,"created":"2020-01-23 11:31:32.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/72/703972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/703972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/703972/1"}}},"commit":{"parents":[{"commit":"d5ae8ad1bced6603e4f6b146a4288d92736fd35d","subject":"Merge \"Disable key sorting in workflow definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5ae8ad1bced6603e4f6b146a4288d92736fd35d"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-23 11:31:11.000000000","tz":120},"subject":"Remove openstack actions from mistral","message":"Remove openstack actions from mistral\n\nDepends-on: https://review.opendev.org/#/c/703296/\nChange-Id: Id62fdabe7699e7c3b2977166e253cfc77779e467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/357bf55e35a7a50234877edb318d5ec3d0e056f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/357bf55e35a7a50234877edb318d5ec3d0e056f3"}]},"branch":"refs/heads/master"},"4e4ce2b0b5ab32ec57be48785dd47e87c266fbee":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-23 11:33:14.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/72/703972/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/703972/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/703972/2"}}},"commit":{"parents":[{"commit":"d5ae8ad1bced6603e4f6b146a4288d92736fd35d","subject":"Merge \"Disable key sorting in workflow definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5ae8ad1bced6603e4f6b146a4288d92736fd35d"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-23 11:32:40.000000000","tz":120},"subject":"Remove OpenStack actions from mistral","message":"Remove OpenStack actions from mistral\n\nDepends-on: https://review.opendev.org/#/c/703296/\nChange-Id: Id62fdabe7699e7c3b2977166e253cfc77779e467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e4ce2b0b5ab32ec57be48785dd47e87c266fbee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4e4ce2b0b5ab32ec57be48785dd47e87c266fbee"}]},"branch":"refs/heads/master"},"f110424d29a5bcfe1a0f27b05bef53caf72cd976":{"kind":"REWORK","_number":3,"created":"2020-01-23 13:56:33.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/72/703972/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/703972/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/703972/3"}}},"commit":{"parents":[{"commit":"d5ae8ad1bced6603e4f6b146a4288d92736fd35d","subject":"Merge \"Disable key sorting in workflow definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5ae8ad1bced6603e4f6b146a4288d92736fd35d"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-23 13:56:08.000000000","tz":120},"subject":"Remove OpenStack actions from mistral","message":"Remove OpenStack actions from mistral\n\nDepends-on: https://review.opendev.org/#/c/703296/\nChange-Id: Id62fdabe7699e7c3b2977166e253cfc77779e467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f110424d29a5bcfe1a0f27b05bef53caf72cd976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f110424d29a5bcfe1a0f27b05bef53caf72cd976"}]},"branch":"refs/heads/master"},"454c1f6f37ece220e9ef18953bf250492082ef87":{"kind":"REWORK","_number":4,"created":"2020-01-26 07:23:49.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/72/703972/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/703972/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/703972/4"}}},"commit":{"parents":[{"commit":"d5ae8ad1bced6603e4f6b146a4288d92736fd35d","subject":"Merge \"Disable key sorting in workflow definition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5ae8ad1bced6603e4f6b146a4288d92736fd35d"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 07:23:14.000000000","tz":120},"subject":"Remove OpenStack actions from mistral","message":"Remove OpenStack actions from mistral\n\nDepends-on: https://review.opendev.org/#/c/703296/\nChange-Id: Id62fdabe7699e7c3b2977166e253cfc77779e467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/454c1f6f37ece220e9ef18953bf250492082ef87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/454c1f6f37ece220e9ef18953bf250492082ef87"}]},"branch":"refs/heads/master"},"87475d23d90a20ea5c2c81c34754475f72311d9e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-01-26 13:03:18.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/72/703972/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/703972/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/703972/5"}}},"commit":{"parents":[{"commit":"da93da4ac1da9c9d8e5f4c1d55861804ba1dec01","subject":"Merge \"Add json param to HTTPAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da93da4ac1da9c9d8e5f4c1d55861804ba1dec01"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 13:00:09.000000000","tz":120},"subject":"Remove OpenStack actions from mistral","message":"Remove OpenStack actions from mistral\n\nDepends-on: https://review.opendev.org/#/c/703296/\nChange-Id: Id62fdabe7699e7c3b2977166e253cfc77779e467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87475d23d90a20ea5c2c81c34754475f72311d9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87475d23d90a20ea5c2c81c34754475f72311d9e"}]},"branch":"refs/heads/master"},"9eb498a7fa428728d4beabff8a9d9c131283c0ff":{"kind":"REWORK","_number":6,"created":"2020-01-29 15:31:31.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/72/703972/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/703972/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/703972/6"}}},"commit":{"parents":[{"commit":"da93da4ac1da9c9d8e5f4c1d55861804ba1dec01","subject":"Merge \"Add json param to HTTPAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da93da4ac1da9c9d8e5f4c1d55861804ba1dec01"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-29 15:30:48.000000000","tz":120},"subject":"Remove OpenStack actions from mistral","message":"Remove OpenStack actions from mistral\n\nDepends-on: https://review.opendev.org/#/c/703296/\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: Id62fdabe7699e7c3b2977166e253cfc77779e467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9eb498a7fa428728d4beabff8a9d9c131283c0ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/9eb498a7fa428728d4beabff8a9d9c131283c0ff"}]},"branch":"refs/heads/master"},"4c017c53af0e08bbcb6a0f7766aa6f4752f545f5":{"kind":"REWORK","_number":7,"created":"2020-01-30 09:13:46.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/72/703972/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/703972/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/703972/7"}}},"commit":{"parents":[{"commit":"da93da4ac1da9c9d8e5f4c1d55861804ba1dec01","subject":"Merge \"Add json param to HTTPAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da93da4ac1da9c9d8e5f4c1d55861804ba1dec01"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-30 09:13:27.000000000","tz":120},"subject":"Remove OpenStack actions from mistral","message":"Remove OpenStack actions from mistral\n\nDepends-on: https://review.opendev.org/#/c/703296/\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: Id62fdabe7699e7c3b2977166e253cfc77779e467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4c017c53af0e08bbcb6a0f7766aa6f4752f545f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4c017c53af0e08bbcb6a0f7766aa6f4752f545f5"}]},"branch":"refs/heads/master"},"c151c470c4c3a165a53d9da908daa91d59e17f9e":{"kind":"REWORK","_number":8,"created":"2020-01-30 10:05:33.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/72/703972/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/703972/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/703972/8"}}},"commit":{"parents":[{"commit":"da93da4ac1da9c9d8e5f4c1d55861804ba1dec01","subject":"Merge \"Add json param to HTTPAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da93da4ac1da9c9d8e5f4c1d55861804ba1dec01"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-30 10:05:16.000000000","tz":120},"subject":"Remove OpenStack actions from mistral","message":"Remove OpenStack actions from mistral\n\nDepends-on: https://review.opendev.org/#/c/703296/\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: Id62fdabe7699e7c3b2977166e253cfc77779e467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c151c470c4c3a165a53d9da908daa91d59e17f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c151c470c4c3a165a53d9da908daa91d59e17f9e"}]},"branch":"refs/heads/master"},"604c8106a987b3b28b6db576fcba80250e971603":{"kind":"REWORK","_number":9,"created":"2020-01-30 10:07:01.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/72/703972/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/703972/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/703972/9"}}},"commit":{"parents":[{"commit":"da93da4ac1da9c9d8e5f4c1d55861804ba1dec01","subject":"Merge \"Add json param to HTTPAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da93da4ac1da9c9d8e5f4c1d55861804ba1dec01"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-30 10:06:45.000000000","tz":120},"subject":"Remove OpenStack actions from mistral","message":"Remove OpenStack actions from mistral\n\nDepends-on: https://review.opendev.org/#/c/703296/\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: Id62fdabe7699e7c3b2977166e253cfc77779e467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/604c8106a987b3b28b6db576fcba80250e971603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/604c8106a987b3b28b6db576fcba80250e971603"}]},"branch":"refs/heads/master"},"4675f524749b07c7fd387fdc61076ecb93f69d13":{"kind":"REWORK","_number":10,"created":"2020-01-30 14:58:16.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/72/703972/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/703972/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/703972/10"}}},"commit":{"parents":[{"commit":"f259bbe6b604bbb874951d0168ab52ca40c4407b","subject":"Merge \"Set tempest configuration to support service api\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f259bbe6b604bbb874951d0168ab52ca40c4407b"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-30 14:57:55.000000000","tz":120},"subject":"Remove OpenStack actions from mistral","message":"Remove OpenStack actions from mistral\n\nDepends-on: https://review.opendev.org/#/c/703296/\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: Id62fdabe7699e7c3b2977166e253cfc77779e467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4675f524749b07c7fd387fdc61076ecb93f69d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4675f524749b07c7fd387fdc61076ecb93f69d13"}]},"branch":"refs/heads/master"},"833abf1c181e8fb5b08b51a3cf2462355d368ea2":{"kind":"REWORK","_number":11,"created":"2020-02-18 14:29:48.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/72/703972/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/703972/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/703972/11"}}},"commit":{"parents":[{"commit":"0a3880c22c819b41e8cfc6ea593e3f69112ec7a5","subject":"Make tripleO CI job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0a3880c22c819b41e8cfc6ea593e3f69112ec7a5"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-18 14:29:24.000000000","tz":120},"subject":"Remove OpenStack actions from mistral","message":"Remove OpenStack actions from mistral\n\nDepends-on: https://review.opendev.org/#/c/703296/\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: Id62fdabe7699e7c3b2977166e253cfc77779e467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/833abf1c181e8fb5b08b51a3cf2462355d368ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/833abf1c181e8fb5b08b51a3cf2462355d368ea2"}]},"branch":"refs/heads/master"},"8bdf341af7348d3d092c6523a09df118c51c2a53":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-02-26 08:12:59.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/72/703972/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/703972/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/703972/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/703972/12"}}},"commit":{"parents":[{"commit":"95d9f899dbb971fbbf766023c24a1a538ee65d0e","subject":"Extend task and workflow notification data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/95d9f899dbb971fbbf766023c24a1a538ee65d0e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-21 14:53:34.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-26 08:12:01.000000000","tz":120},"subject":"Remove OpenStack actions from mistral","message":"Remove OpenStack actions from mistral\n\nDepends-on: https://review.opendev.org/#/c/703296/\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: Id62fdabe7699e7c3b2977166e253cfc77779e467\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8bdf341af7348d3d092c6523a09df118c51c2a53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8bdf341af7348d3d092c6523a09df118c51c2a53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
