)]}'
{"id":"openstack%2Fmistral~704056","triplet_id":"openstack%2Fmistral~stable%2Ftrain~I48347fac036fcca4b8815d1ed7dad2b9125211cc","project":"openstack/mistral","branch":"stable/train","hashtags":[],"change_id":"I48347fac036fcca4b8815d1ed7dad2b9125211cc","subject":"DNM: testing mistral jobs keep running on py2","status":"ABANDONED","created":"2020-01-23 19:29:46.000000000","updated":"2021-08-04 16:28:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c78fd2597c9a13f5a25914cab932c3dd417aa647","_number":704056,"virtual_id_number":704056,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-28 10:56:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"date":"2020-02-12 13:19:35.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 10:56:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-12 13:19:35.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"beebd5d28144fc4773128933333bd2033de0fb83","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-23 19:29:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d4cd4a043c2e85075430116d9fa5c0a14a2d167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 21:53:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b772778c83b6493b8d693769f16890ba : SUCCESS in 9m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ae37358aede47fdb3a456292eb4c954 : SUCCESS in 12m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54383e7081a64dc2b4aa01fdd6008d38 : SUCCESS in 10m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/953e2eb4763d4a6eb9060dd6be99465b : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e70053404062431c909ff69425671097 : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c233eddaa37e458caff23ec72b16b8b0 : SUCCESS in 14m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2ad16993ac26400f94aed41fcc81687f : FAILURE in 43m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/518ab6b43c1a489296d333f05a42ad15 : FAILURE in 42m 06s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c4e5c8d5d98f43e7b34c642836e03654 : FAILURE in 41m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7d746d2f69874dcca045cbf97ceb7fce : SUCCESS in 30m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/964eb2f01705449aafdf7f0b5699b8aa : SUCCESS in 31m 03s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/182818e683544ac8ac7c18bc88ae5c80 : SUCCESS in 1h 14m 55s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e752a4ae8dad4a03ae735cb94a48c778 : SUCCESS in 2h 21m 21s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/3cc957dadb494d5aa510331a50497d3a : SUCCESS in 58m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1aecf7f2571949f7a70fd596183d3313 : SUCCESS in 10m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d80587c07d4006c151cea5780a21c0eac0a0458b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-23 21:56:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c910b18e85ed4d918765450098c2157108fc2c71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 00:45:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c678a39dd0c74c44a40be877116c6c56 : SUCCESS in 6m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3571a04e20f24b4793491e42c792a5eb : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2c29abde3dd4b82858355022bcb8fba : SUCCESS in 11m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5615034a49b2421181d377624d5aee65 : SUCCESS in 9m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/961279cbafdb4eb0a2900cc58d698640 : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f18e2ee822546cc97ebbeced72080c1 : SUCCESS in 8m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2aa729d732834d33a952df75037ff35b : FAILURE in 45m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/193e8d2d45dc411da03def74554bbece : FAILURE in 48m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/decf97b3048c4266ac0a51e43e8651dc : FAILURE in 46m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a203339bfe0c4fc39c53b31135584da9 : SUCCESS in 28m 35s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/09d5ebc2ec184b958065c0d444f87d14 : SUCCESS in 27m 52s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/ab041c2809584f99b304d7c56838ad8a : SUCCESS in 1h 19m 48s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a3bba4cd5c254333a3dc86b909372164 : SUCCESS in 2h 43m 23s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2e5bf92a1a594f55ab074532f7bbd6a5 : SUCCESS in 1h 10m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a712ef4f2dc45eab4567e040c62a28b : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4eae508f05e29bccc9f0cd812ca0df7f99937899","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-28 07:49:55.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"58d78a5e56d1f5ae2b740879b9c120d460ab156a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 10:56:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27e1798baeaf482998d5dc80b52d5cc3 : SUCCESS in 8m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8922deb0fb634198974f77c86a47652a : SUCCESS in 10m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b3a317601c044a08c5392be3a455008 : SUCCESS in 11m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f82a123ca4b6400185e1898b49811ea3 : SUCCESS in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e383c1069ec4d2bac2389f1a155b769 : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce639848dceb45659fade1b0770e5096 : SUCCESS in 10m 22s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a94714cb6fd24a03a6b14c587ebaa16d : SUCCESS in 44m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f53139b484b04b49afbaf9eed34e8928 : SUCCESS in 44m 53s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d78fd57f549846fc9abc74d6e28fb389 : SUCCESS in 43m 11s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b2cd9de2e0c84133aea7c681701df71d : SUCCESS in 29m 00s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9d9fd74e2ec1419189baa7d195d8e687 : SUCCESS in 31m 36s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/70b20f9c50834d9c94e7658a48571250 : SUCCESS in 1h 32m 46s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/78fd5b0f64e147e2b30b3f7bd8f0383f : SUCCESS in 2h 57m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/3cd78403bda341458442709379807f08 : SUCCESS in 1h 12m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9de2083edb24a4e8e0f3be278df3b0b : SUCCESS in 13m 00s","accounts_in_message":[],"_revision_number":2},{"id":"2953a46bbde9e829c78c0d2ba980c71726c56fa5","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-12 13:19:35.000000000","message":"Patch Set 2:\n\nDo we still need this ? or this can be aborted ?","accounts_in_message":[],"_revision_number":2},{"id":"c78fd2597c9a13f5a25914cab932c3dd417aa647","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-08-04 16:28:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"df7575b2594b4e3eb06817bbc5abee44fe77b389","revisions":{"0f433ebb27f81ff5263d35e24c48ed40918f3b93":{"kind":"REWORK","_number":1,"created":"2020-01-23 19:29:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/56/704056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/704056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/704056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/704056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/704056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/704056/1"}}},"commit":{"parents":[{"commit":"6651519919df81265ee442b06fe467f17f3c0a3d","subject":"Merge \"Make sure minimum amqp is 2.5.2\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6651519919df81265ee442b06fe467f17f3c0a3d"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 19:28:48.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 19:29:44.000000000","tz":-360},"subject":"DNM: testing mistral jobs keep running on py2","message":"DNM: testing mistral jobs keep running on py2\n\nDepends-On: https://review.opendev.org/#/c/704055/\nChange-Id: I48347fac036fcca4b8815d1ed7dad2b9125211cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0f433ebb27f81ff5263d35e24c48ed40918f3b93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0f433ebb27f81ff5263d35e24c48ed40918f3b93"}]},"branch":"refs/heads/stable/train"},"df7575b2594b4e3eb06817bbc5abee44fe77b389":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-28 07:49:55.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/56/704056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/56/704056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/704056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/704056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/56/704056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/56/704056/2"}}},"commit":{"parents":[{"commit":"6651519919df81265ee442b06fe467f17f3c0a3d","subject":"Merge \"Make sure minimum amqp is 2.5.2\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6651519919df81265ee442b06fe467f17f3c0a3d"}]}],"author":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-01-23 19:28:48.000000000","tz":-360},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-28 07:49:52.000000000","tz":0},"subject":"DNM: testing mistral jobs keep running on py2","message":"DNM: testing mistral jobs keep running on py2\n\nDepends-On: https://review.opendev.org/#/c/704055/\nDepends-On: https://review.opendev.org/#/c/704274/\nChange-Id: I48347fac036fcca4b8815d1ed7dad2b9125211cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/df7575b2594b4e3eb06817bbc5abee44fe77b389"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/df7575b2594b4e3eb06817bbc5abee44fe77b389"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
