)]}'
{"id":"openstack%2Fmistral~704103","triplet_id":"openstack%2Fmistral~master~If2f206a5191d1eecdcf95e1f7b44a87968432876","project":"openstack/mistral","branch":"master","topic":"bp/mistral-restructure-docs","hashtags":[],"change_id":"If2f206a5191d1eecdcf95e1f7b44a87968432876","subject":"Adjust images in the docs and other small fixes","status":"MERGED","created":"2020-01-24 05:38:56.000000000","updated":"2020-01-27 09:46:37.000000000","submitted":"2020-01-27 09:44:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704103-1580118253303-04bc3352","meta_rev_id":"01d2d7a868ee1c472c77a299b415b41292acf650","_number":704103,"virtual_id_number":704103,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 09:44:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2020-01-27 04:46:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-01-27 09:35:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-27 04:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2020-01-27 09:35:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-26 09:12:04.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-01-26 09:13:16.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-01-27 09:35:33.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-01-27 09:44:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"631b92437caf0f73ef0ed651af44518d8d20b693","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-24 05:38:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccdd102a16f7578917131121727051e349ba0bec","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-24 05:41:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"de8615e65f71859ff6c0de9d7ef86b8984769b0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 05:50:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4da4319972bd4cbe9b8cfdc5ba1f9537 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/559404cf247d451bb0987c463e9de0db : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b0164d7591bd652231ca9deb36d47cb4df3549f2","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-26 09:12:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c02cb7d81af6c381aeebe42b64623b217ef17057","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-26 09:13:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7b7ef161aee16ef6b22a6f55457e32d7c942efee","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-27 04:46:42.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"cc16de79f2603eb36588edb09b4bb614a86c511e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 04:54:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32576b69ca274e98a2be2f6e0093bd9b : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124e8b35284f4c06aafee76138b68442 : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":3},{"id":"9c7c73203cce32dce88a1782ccb2827af3fd8a6f","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-27 09:35:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8556c40916779ab1b715a6b5f24f7dce42627746","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 09:35:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c8abd697818a2cdf316ca9c16aa28b611108b2ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 09:44:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a33360ec81c9567516381c0465b33f99ad5981d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 09:44:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2d79b0e409249a79591892c20371409 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d1dc7dcaf9426d84e6a5bbeed65edb : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":3},{"id":"e5a72732334cd1ab4c6a19473625f832702633bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 09:46:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a54bf5da59ee49a2af35cf9db1042013 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0ec5e48012e96e9a2013806b380dbb851f0df04a","revisions":{"a3432e939690ccaae73c5c879f10ff79051d5198":{"kind":"REWORK","_number":1,"created":"2020-01-24 05:38:56.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/03/704103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/704103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/704103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/704103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/704103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/704103/1"}}},"commit":{"parents":[{"commit":"06b44cfd8bc87c8909f153239cd02d867e8024ba","subject":"Move \"Long Running Business Process\" article from Wiki to docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/06b44cfd8bc87c8909f153239cd02d867e8024ba"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-24 05:10:44.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-24 05:35:20.000000000","tz":420},"subject":"Adjust images in the docs and other small fixes","message":"Adjust images in the docs and other small fixes\n\n* Some \".. image:\" RST directives changed to \".. figure:\" where\n  it\u0027s more suitable.\n* All images moved to subfolder where they\u0027re used\n* Fixed indentation in the Workflow Language specification\n* Other minor changes in docs\n\nChange-Id: If2f206a5191d1eecdcf95e1f7b44a87968432876\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a3432e939690ccaae73c5c879f10ff79051d5198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a3432e939690ccaae73c5c879f10ff79051d5198"}]},"branch":"refs/heads/master"},"bc8ff8805bdb15572ed8464862f9cb9cfa816fef":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-24 05:41:01.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/03/704103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/704103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/704103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/704103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/704103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/704103/2"}}},"commit":{"parents":[{"commit":"0047276f1a97f4a00abcd195cf26c00976984f18","subject":"Move \"Long Running Business Process\" article from Wiki to docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0047276f1a97f4a00abcd195cf26c00976984f18"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-24 05:10:44.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-24 05:40:43.000000000","tz":420},"subject":"Adjust images in the docs and other small fixes","message":"Adjust images in the docs and other small fixes\n\n* Some \".. image:\" RST directives changed to \".. figure:\" where\n  it\u0027s more suitable.\n* All images moved to subfolder where they\u0027re used\n* Fixed indentation in the Workflow Language specification\n* Other minor changes in docs\n\nChange-Id: If2f206a5191d1eecdcf95e1f7b44a87968432876\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bc8ff8805bdb15572ed8464862f9cb9cfa816fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bc8ff8805bdb15572ed8464862f9cb9cfa816fef"}]},"branch":"refs/heads/master"},"0ec5e48012e96e9a2013806b380dbb851f0df04a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-27 04:46:42.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/03/704103/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/704103/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/704103/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/704103/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/704103/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/704103/3"}}},"commit":{"parents":[{"commit":"6928615fde51520faf45b89d068b9ba7db99352d","subject":"Move \"Long Running Business Process\" article from Wiki to docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6928615fde51520faf45b89d068b9ba7db99352d"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-24 05:10:44.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-27 04:45:49.000000000","tz":420},"subject":"Adjust images in the docs and other small fixes","message":"Adjust images in the docs and other small fixes\n\n* Some \".. image:\" RST directives changed to \".. figure:\" where\n  it\u0027s more suitable.\n* All images moved to subfolder where they\u0027re used\n* Fixed indentation in the Workflow Language specification\n* Other minor changes in docs\n\nChange-Id: If2f206a5191d1eecdcf95e1f7b44a87968432876\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0ec5e48012e96e9a2013806b380dbb851f0df04a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0ec5e48012e96e9a2013806b380dbb851f0df04a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
