)]}'
{"id":"openstack%2Fmistral~704249","triplet_id":"openstack%2Fmistral~master~I3bc7f9bb6b76953c7030303791a23aae08daf45e","project":"openstack/mistral","branch":"master","topic":"inspect","hashtags":[],"change_id":"I3bc7f9bb6b76953c7030303791a23aae08daf45e","subject":"remove inspect utils it was moved to mistral-lib","status":"ABANDONED","created":"2020-01-26 13:03:18.000000000","updated":"2020-01-29 15:33:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d6a944452f9aaf74e5ad46699b7b914f681b368c","_number":704249,"virtual_id_number":704249,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-27 08:41:33.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-01-29 13:51:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f8abb07e7fdb570661afce754dce381366bdcd9","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-26 13:03:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b4d7ee69d4968469b458fa50ad3ca820649bce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 14:05:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9743a48ff248b08030b0430188a823 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22e83581d9f449669c8fd6e520c1369c : SUCCESS in 11m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a957887c6164c5a8aaa634d679dc75d : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc567c921a8646e182c82dae899e9df9 : FAILURE in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce7982a2f534affaf6c39ebd890c9ac : FAILURE in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f1285999432424d8458d406db76d4fd : SUCCESS in 10m 01s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4c9b137630af4e11b044d101a02ec177 : SUCCESS in 40m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a20879431aa4b1e94874d7a80b61215 : SUCCESS in 47m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54841be3bfb549d3a423c5a8c5bf1dba : SUCCESS in 55m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3acc34ac2ad247e9a62b4e74ca254b95 : SUCCESS in 47m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4b20fc146d40412e8b278d25e800edee : SUCCESS in 59m 10s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c92f2bd569244d308bd40c6701e75103 : SUCCESS in 30m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/2432440a70a249fcb0326112b5acdd48 : SUCCESS in 31m 28s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3f0d8c758bdf4bd7bbaa89ace830a1a5 : FAILURE in 24m 07s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/00759099e8cc49739721f942a513a9cc : FAILURE in 45m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0974c09a790d441e964385a5be28ff9a : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":1},{"id":"58b458a70241db33130273f776d4b5f3e2f55c7f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-26 14:23:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d1826ef9ba7e045012c1574550586aa618a733aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-26 15:22:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86f93c5faa27443d94560b0040d7cdbf : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9359cded1cd429db9a6e6a6b582c464 : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52070423cbf24077a3b95b1bdc2ec270 : SUCCESS in 10m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05869793ec2a4887872f3a08cabe0796 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/541f8218a8e948bbbb8ea81169855b38 : SUCCESS in 8m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47f5b20a550a4398b56da4f4429e4a41 : SUCCESS in 9m 48s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a73b85c23476427dbb3d33fee3b8a056 : SUCCESS in 40m 33s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dff760fc31184a14bf2df3a998ac4199 : SUCCESS in 42m 37s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3312c3f907c8459dbfa42204fe077d67 : SUCCESS in 41m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a231e17a828a46fb8b780f7a8cbd8445 : SUCCESS in 42m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/176d7ed14c7042ecb2efc4ed042d61f3 : SUCCESS in 41m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/166fd252556a472db385209aa9f586ad : SUCCESS in 29m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/953180ff450c4b789c1508bc0c08fad0 : SUCCESS in 28m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/98ca2ae581b545f48c45957a340aadc6 : FAILURE in 24m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2b5b0850023544f79a999e1b5744d725 : SUCCESS in 57m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c45e310c58f44670831b83016acb0f51 : SUCCESS in 10m 15s","accounts_in_message":[],"_revision_number":2},{"id":"8825e258540458fdbe7a03d254dc926abedca535","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-27 07:54:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e241bbd5b4f1915b6d8919022eaab4780c770059","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-01-27 08:41:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b328a97990ac306ed355b1a892eafc016eeadf17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 09:00:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc8332846e642039595113aca9bf4c3 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df1392b83b444970bc165dab08b0b8f9 : SUCCESS in 8m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b535cceaa814c4587a7293f02f48563 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a24ed3e3efb4c808215737c2ece3583 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/681ba1c4c97f4b7cbd48812d690203d9 : SUCCESS in 7m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b82481e4fd5c4aef8eba19fa8dd0f308 : SUCCESS in 9m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bf1cdfc56a6a4ba7871cb738f7e0d364 : FAILURE in 39m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97bba0e4b7f1476ba6c2ba5d637e9280 : FAILURE in 41m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b3bdc6e2f134089b0042a9db80315ce : FAILURE in 42m 55s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a69c1b092b7d4cc0a1a68aff79131969 : FAILURE in 43m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c8f7d30bb8194c7cb3a61b8c9e17129a : FAILURE in 42m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/60fae4e3a3604b96ae0d8169559faaff : SUCCESS in 29m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ed68858dfa894b8fa1bd126a34eeed43 : SUCCESS in 27m 28s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b64fa4bcfd034106916fe5337df8c2ca : FAILURE in 29m 02s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d9de469bfcd44f788965bd63c0ecfb37 : SUCCESS in 1h 05m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce4070bbd12c422a9f11c124e74940f8 : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":3},{"id":"1e9eb40dd8582d7a78e40b1f317bd5b54fdccb30","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-27 09:03:45.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"20270227d95256a1d990503f7196482ef7163537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 10:08:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d044980fa4e74aeea6c4d94ce59f3a12 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cef48504b12d445eafc184d821dc6a64 : SUCCESS in 9m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43939a26bd2b414fbd27b16e9a662d98 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52e77059724a44d0bdbdf5435442f37d : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ec43bd998cb433aa2e4867b3bd3d6c5 : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1221266731624eedba9feb4f83c1b51c : SUCCESS in 9m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2f5c432747ff4668bd3a35e8799a0d62 : SUCCESS in 44m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/719b8f4a5f9b42678ff173c04412d7e4 : SUCCESS in 46m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb658fa30a744b79971d2467f775e18e : SUCCESS in 41m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b707c2f4e0d14381a9a2870c4d1ad05c : SUCCESS in 43m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/920d207a8c6f4bb585848ea4b95f3eda : SUCCESS in 49m 55s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6ba9b94d16404a058de4bd1d9a5ea5e7 : SUCCESS in 29m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ab9ccef9dcd44baca7d9b223d7648869 : SUCCESS in 26m 59s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/db81058dceab45258792fd7a83d66a7f : FAILURE in 31m 57s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/8ffa2a900178487f8d749d1b0b2a80c3 : SUCCESS in 59m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6e1583c12c44f8181a6aad9f0617172 : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":4},{"id":"03d24fc1ae27507763709d44e672a7e44eeaaf51","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-28 07:28:01.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"55517d269eb677f0d73eba2b576d5a5f7aecd63e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 08:50:40.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dadbdb3f42a474db877b9023f670918 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1080753f6b148ec84f503755432484e : SUCCESS in 8m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2df717c7c1c944a7bfaa4ecacfe05877 : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f570920bce814675a30f895ad542834b : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6c113788086488bb3cf4df77d1d55f6 : SUCCESS in 7m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc594dd76ace4cf3ba2b6a64baf5e535 : SUCCESS in 10m 36s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ee1b39ff3597489bb9c7819656a51592 : SUCCESS in 43m 14s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78d8700f3cc843db9662a359391a46f2 : SUCCESS in 44m 19s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/589a23653d5549b1a6e99a76c421b2e2 : SUCCESS in 45m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f671fea9941a4805a150fd130cf38660 : SUCCESS in 42m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/53be2da9a778475497d0a46177460418 : SUCCESS in 41m 37s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a758c4824f00469db99e612a3c2dd1e7 : SUCCESS in 28m 42s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b0046a6047a144888e399a61a7def125 : SUCCESS in 28m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/b9455b0c4b16436ca1f5437f9f7ca26a : FAILURE in 31m 41s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9ae6807256224d53a2f7a7855d34c8ee : SUCCESS in 1h 19m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/205bcb7702a8453e9b63ac80dc1a4ae5 : SUCCESS in 12m 10s","accounts_in_message":[],"_revision_number":4},{"id":"d1510c2db262270c2dabf3c8c204624be2ba1176","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-29 06:37:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"177cda6f9e3b1c88d1250d7a57b166525310bbf1","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-29 07:22:52.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"fab2be30efddddeb105393ba89e4efc3c083568b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 08:33:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a6f5aee9c594e1aaecf7726c82ab6e6 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0aa27a156d964ef49fcb09f35dea95c5 : SUCCESS in 9m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d62d1d271e7f4c878bbed2a72569aeb5 : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f14ecf03cc446e0900aedfe52a918b7 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/961051e37d704f0aa3e894f5686203c2 : SUCCESS in 10m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5e0af75bc317488c9ea87bebc1a5009d : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dea5055c11e48c981892760503d4efa : SUCCESS in 4m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c669ff512808468f8badbd6e4279a393 : SUCCESS in 9m 57s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/89d9890d4bb043e6a492a437b4ea175e : SUCCESS in 44m 17s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/290597a34dd546c6b8806e5781257b72 : SUCCESS in 46m 05s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eebbda3a21c0494ea4a48fca1bf521a9 : SUCCESS in 50m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a11c4a63ef244ebba67835a385b5ac51 : SUCCESS in 44m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dc5fb3492ba043b093ccab3cf64c5491 : SUCCESS in 44m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/54da458ff7864411b90838c42bb854f9 : SUCCESS in 28m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/98e77a90897e4503ab9caadb4c57a378 : SUCCESS in 27m 25s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/971cf1a0062a4019bb7d8c7d4b27ac3e : FAILURE in 30m 02s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1840d6f94f0d471ab126ea91405c82b9 : SUCCESS in 1h 04m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f3f528b2029459dab3410e8c5cb31a6 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":6},{"id":"f0c79a633bef824b7953f2ebb5098dc9fe4324e8","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-29 10:15:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1d0c6490375561bab3700a9acdc5fbb197d2ebca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 10:27:39.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"4ff4ca5b936a95334e7f30a420a752f3abfb87e9","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-29 10:28:02.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"e0613a1c8812268a9001090d01eebdbc239abe48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 11:53:07.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89aad2377f054439805212077af83837 : SUCCESS in 9m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7259c4f989c6480a85c504d1e057f4c5 : SUCCESS in 9m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2dd9f3b1946f4a7bae2bb6f5a8febc45 : SUCCESS in 11m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c1c2666d62647d8bb7dca1ed78eaf7d : SUCCESS in 11m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5baa0ee5fc154778a5f4360c4ee19494 : SUCCESS in 11m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3c798bf96204273a98eef90ad1d2c7a : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a00f8861368a4289bb8783ccf3eaeffd : SUCCESS in 9m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc63922c16b04baeb6340277c1d3eb8b : SUCCESS in 11m 47s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8a30bf5f1dd149a39e14c61251061495 : SUCCESS in 40m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e418c455bf9b440a952980a6212b218a : SUCCESS in 1h 03m 35s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/addfa79422694088ac97432a5a4389d6 : SUCCESS in 1h 01m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7828a8b32eaa41c186f486cde85ffd36 : SUCCESS in 38m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/eeae9452620e4132a7e18e0fb9cfc7b5 : SUCCESS in 1h 02m 09s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/47d04841f78c4ca19db49cd472e41f9c : SUCCESS in 33m 31s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/71ae6a67c1f14dbbba5abd09976f8a99 : SUCCESS in 31m 58s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f26241767bed42148a43e97762da8402 : FAILURE in 24m 52s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/287686b68d754453a92036ecc20d72c5 : SUCCESS in 1h 18m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c41fd0005ea64572a276297239c57320 : SUCCESS in 15m 58s","accounts_in_message":[],"_revision_number":7},{"id":"85e675ef621c8df69420d09b1791c436d728eadb","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-29 12:36:12.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b7a38959189b5caf59606b1f8af7eca53ec523aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 13:51:34.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5aea3d9f951416682b75b7b197c7e69 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8183669c4fbe49649a32c4dbd324f1b6 : SUCCESS in 8m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8265285bbea0459cb367179dc4850f98 : SUCCESS in 12m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/370ec974f8514364a2606060dda3ad8e : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97d908b363274662ab1a6b92e0533e34 : SUCCESS in 10m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/570b1f63eace43ffbe623bca84f5b0ff : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd63ea2247734fee8a6c1544657116c6 : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f720587dd4484a378542fbee910f0b3a : SUCCESS in 12m 25s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6d0c82cf871640b1b18c4d53f7a344a0 : SUCCESS in 42m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e7afc451c984ede8c77f76efaa42dde : SUCCESS in 59m 18s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59d5e7adcd334d09948ad1201d3ba97b : SUCCESS in 50m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/62d9d48ff6ff4550bbe83388bd7fd2ef : SUCCESS in 40m 50s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/305ac9102b144fd29c841e9373245087 : SUCCESS in 58m 14s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/dbc80ca40fc149cfb184b38cac1512f4 : SUCCESS in 32m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/de27dab21eb44fe8a657d0803ff9e014 : SUCCESS in 27m 53s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a8ffb3fb600645688ec4b7b747a9a7bb : FAILURE in 29m 33s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e7b0bd2fe0574b5d93d535f64474c188 : SUCCESS in 1h 12m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae7f538870ee4ff08ef929593338ea61 : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":8},{"id":"8ebd6e646a5aa4e28f2717d82dd599333647cb88","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-29 14:17:37.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2d8c29f20bd11a9ce64c2998b785d3125e404b9f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-29 15:33:46.000000000","message":"Abandoned\n\ndid a squash","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"b50a4394a638180f02d2a2f839ad2e6b182d638a","revisions":{"a2dc38a4467c796c81d60bd35aa6950db5b8f093":{"kind":"REWORK","_number":1,"created":"2020-01-26 13:03:18.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/49/704249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/49/704249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/49/704249/1"}}},"commit":{"parents":[{"commit":"87475d23d90a20ea5c2c81c34754475f72311d9e","subject":"Remove OpenStack actions from mistral","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87475d23d90a20ea5c2c81c34754475f72311d9e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 13:02:09.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 13:02:55.000000000","tz":120},"subject":"remove inspect utils it was moved to mistral-lib","message":"remove inspect utils it was moved to mistral-lib\n\nDepends-On: https://review.opendev.org/#/c/703295/\nChange-Id: I3bc7f9bb6b76953c7030303791a23aae08daf45e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a2dc38a4467c796c81d60bd35aa6950db5b8f093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a2dc38a4467c796c81d60bd35aa6950db5b8f093"}]},"branch":"refs/heads/master"},"99d57deafc5d05b9abda8b21ec2b1e3121668466":{"kind":"REWORK","_number":2,"created":"2020-01-26 14:23:48.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/49/704249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/49/704249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/49/704249/2"}}},"commit":{"parents":[{"commit":"87475d23d90a20ea5c2c81c34754475f72311d9e","subject":"Remove OpenStack actions from mistral","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87475d23d90a20ea5c2c81c34754475f72311d9e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 13:02:09.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 14:23:33.000000000","tz":120},"subject":"remove inspect utils it was moved to mistral-lib","message":"remove inspect utils it was moved to mistral-lib\n\nDepends-On: https://review.opendev.org/#/c/703295/\nChange-Id: I3bc7f9bb6b76953c7030303791a23aae08daf45e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99d57deafc5d05b9abda8b21ec2b1e3121668466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99d57deafc5d05b9abda8b21ec2b1e3121668466"}]},"branch":"refs/heads/master"},"ad796c49979f52bc18b40a6fe21cf711b14755d8":{"kind":"REWORK","_number":3,"created":"2020-01-27 07:54:00.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/49/704249/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/49/704249/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/49/704249/3"}}},"commit":{"parents":[{"commit":"87475d23d90a20ea5c2c81c34754475f72311d9e","subject":"Remove OpenStack actions from mistral","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87475d23d90a20ea5c2c81c34754475f72311d9e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 13:02:09.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-27 07:53:36.000000000","tz":120},"subject":"remove inspect utils it was moved to mistral-lib","message":"remove inspect utils it was moved to mistral-lib\n\nDepends-On: https://review.opendev.org/#/c/703295/\nChange-Id: I3bc7f9bb6b76953c7030303791a23aae08daf45e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad796c49979f52bc18b40a6fe21cf711b14755d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ad796c49979f52bc18b40a6fe21cf711b14755d8"}]},"branch":"refs/heads/master"},"b1790ae64c917c522bd8d93deca5b6188f682cf0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-01-27 09:03:45.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/49/704249/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/49/704249/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/49/704249/4"}}},"commit":{"parents":[{"commit":"87475d23d90a20ea5c2c81c34754475f72311d9e","subject":"Remove OpenStack actions from mistral","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87475d23d90a20ea5c2c81c34754475f72311d9e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 13:02:09.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-27 09:02:57.000000000","tz":120},"subject":"remove inspect utils it was moved to mistral-lib","message":"remove inspect utils it was moved to mistral-lib\n\nDepends-On: https://review.opendev.org/#/c/703295/\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: I3bc7f9bb6b76953c7030303791a23aae08daf45e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b1790ae64c917c522bd8d93deca5b6188f682cf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b1790ae64c917c522bd8d93deca5b6188f682cf0"}]},"branch":"refs/heads/master"},"af68441381083498758e4d53c768dc949060dbb4":{"kind":"REWORK","_number":5,"created":"2020-01-29 06:37:17.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/49/704249/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/49/704249/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/49/704249/5"}}},"commit":{"parents":[{"commit":"87475d23d90a20ea5c2c81c34754475f72311d9e","subject":"Remove OpenStack actions from mistral","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87475d23d90a20ea5c2c81c34754475f72311d9e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 13:02:09.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-29 06:36:46.000000000","tz":120},"subject":"remove inspect utils it was moved to mistral-lib","message":"remove inspect utils it was moved to mistral-lib\n\nDepends-On: https://review.opendev.org/#/c/703295/\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: I3bc7f9bb6b76953c7030303791a23aae08daf45e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/af68441381083498758e4d53c768dc949060dbb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/af68441381083498758e4d53c768dc949060dbb4"}]},"branch":"refs/heads/master"},"54a13413b13fd6ec166b3fd824c096d142cc593f":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-01-29 07:22:52.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/49/704249/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/49/704249/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/49/704249/6"}}},"commit":{"parents":[{"commit":"87475d23d90a20ea5c2c81c34754475f72311d9e","subject":"Remove OpenStack actions from mistral","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87475d23d90a20ea5c2c81c34754475f72311d9e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 13:02:09.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-29 07:22:17.000000000","tz":120},"subject":"remove inspect utils it was moved to mistral-lib","message":"remove inspect utils it was moved to mistral-lib\n\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: I3bc7f9bb6b76953c7030303791a23aae08daf45e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/54a13413b13fd6ec166b3fd824c096d142cc593f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/54a13413b13fd6ec166b3fd824c096d142cc593f"}]},"branch":"refs/heads/master"},"76f34a9b6a1b1ed1095652010b43c3a325b47429":{"kind":"REWORK","_number":7,"created":"2020-01-29 10:15:21.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/49/704249/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/49/704249/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/49/704249/7"}}},"commit":{"parents":[{"commit":"87475d23d90a20ea5c2c81c34754475f72311d9e","subject":"Remove OpenStack actions from mistral","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87475d23d90a20ea5c2c81c34754475f72311d9e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 13:02:09.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-29 10:14:59.000000000","tz":120},"subject":"remove inspect utils it was moved to mistral-lib","message":"remove inspect utils it was moved to mistral-lib\n\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: I3bc7f9bb6b76953c7030303791a23aae08daf45e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/76f34a9b6a1b1ed1095652010b43c3a325b47429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/76f34a9b6a1b1ed1095652010b43c3a325b47429"}]},"branch":"refs/heads/master"},"c408e9bf0445ce2cd9785bd2409620f3383404da":{"kind":"REWORK","_number":8,"created":"2020-01-29 12:36:12.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/49/704249/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/49/704249/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/49/704249/8"}}},"commit":{"parents":[{"commit":"87475d23d90a20ea5c2c81c34754475f72311d9e","subject":"Remove OpenStack actions from mistral","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87475d23d90a20ea5c2c81c34754475f72311d9e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 13:02:09.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-29 12:35:38.000000000","tz":120},"subject":"remove inspect utils it was moved to mistral-lib","message":"remove inspect utils it was moved to mistral-lib\n\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: I3bc7f9bb6b76953c7030303791a23aae08daf45e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c408e9bf0445ce2cd9785bd2409620f3383404da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c408e9bf0445ce2cd9785bd2409620f3383404da"}]},"branch":"refs/heads/master"},"b50a4394a638180f02d2a2f839ad2e6b182d638a":{"kind":"REWORK","_number":9,"created":"2020-01-29 14:17:37.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/49/704249/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/49/704249/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/49/704249/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/49/704249/9"}}},"commit":{"parents":[{"commit":"87475d23d90a20ea5c2c81c34754475f72311d9e","subject":"Remove OpenStack actions from mistral","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/87475d23d90a20ea5c2c81c34754475f72311d9e"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-26 13:02:09.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-29 14:17:12.000000000","tz":120},"subject":"remove inspect utils it was moved to mistral-lib","message":"remove inspect utils it was moved to mistral-lib\n\nDepends-On: https://review.opendev.org/#/c/704280/\nChange-Id: I3bc7f9bb6b76953c7030303791a23aae08daf45e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b50a4394a638180f02d2a2f839ad2e6b182d638a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b50a4394a638180f02d2a2f839ad2e6b182d638a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
