)]}'
{"id":"openstack%2Fmistral~704318","triplet_id":"openstack%2Fmistral~master~I9cc78df13d0f0715538bbdb76c8ccad273bd2033","project":"openstack/mistral","branch":"master","topic":"fake_clients","hashtags":[],"change_id":"I9cc78df13d0f0715538bbdb76c8ccad273bd2033","subject":"Fix fake clients in actions","status":"MERGED","created":"2020-01-27 13:18:08.000000000","updated":"2020-01-30 12:38:04.000000000","submitted":"2020-01-30 12:35:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704318-1580387709914-ff1353d7","meta_rev_id":"be1f2c26fafb3bdc817b838283377ef569639bad","_number":704318,"virtual_id_number":704318,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 12:35:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-27 18:49:22.000000000","_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"recommended":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2020-01-30 07:51:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":2,"date":"2020-01-30 10:34:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2020-01-28 04:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"date":"2020-01-30 07:40:45.000000000","_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-01-30 07:40:45.000000000","_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":1,"date":"2020-01-30 10:34:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2020-01-30 07:40:45.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 04:48:00.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-01-30 07:40:45.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2020-01-30 07:40:45.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2020-01-30 07:40:45.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-01-30 07:51:57.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-01-30 10:34:15.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-01-30 12:35:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bb8b0f34cf9a59afacec613246756441e8e48e8","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-27 13:18:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfba5001629baa0874e238974e802b9c1c586b24","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-27 14:34:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0c029693b7c17d3fc72cd95beefb259e9161602","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-27 14:51:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b64b016c898dea6241846c58d7b3923a368eda0b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-27 15:35:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0106f8ac40648092cec8618336712bc2d373e851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 18:45:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/744949b5c99c42ff8efb5617b0d0c6cd : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56cdb589d2d549a49343c85a98e41878 : SUCCESS in 11m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acf12461cd1c42ffb164613c896e65b6 : SUCCESS in 10m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8069f6c156b64fb9a5fd09fbed16a1f8 : SUCCESS in 12m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb83f34e7bf040569c57ad36ad292b15 : SUCCESS in 7m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3df318d09844020852436a95b4645c7 : SUCCESS in 12m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3dbe572344414f9780b1ec97f3aa0305 : SUCCESS in 51m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d03730be43440cbb9b3c6f75964e745 : SUCCESS in 49m 01s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2e8923f215f403e9edcd521240d154b : SUCCESS in 53m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/17a2555d5fc747148637c0a48e3dfc51 : SUCCESS in 50m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/93ad9968a24c41a39cb8ae5b5b9b6623 : SUCCESS in 46m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1115508f2b754ba191388f3254f103ca : FAILURE in 33m 51s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/39d3b402fe3d4f748e48bd9bd0caf61c : SUCCESS in 28m 34s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0160a1ef03c741618042fdbb5ee17d32 : SUCCESS in 3h 04m 40s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/fd013d71a04c44e89aedf222df1ff168 : SUCCESS in 1h 02m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb8389468f4d49e1be84bfbccadf60bc : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":4},{"id":"9a771fc3d8c630f174aed9e37ff3d9616538aa3d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-27 18:49:22.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d469cb78d1bce3d927b81b531ca49bd297477c4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 21:56:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9268c22a08a846308620437105007dc0 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38a01ce846f44a8ba6b9c9fc4585d88d : SUCCESS in 10m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20a1c39e6d2145f8855995cf5f33cb0c : SUCCESS in 9m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c71869807c344e9c85dd09ff4d9349ca : SUCCESS in 14m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7301821387dc4961bd47e2390f95839a : SUCCESS in 10m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1bbeafd93004bc69b9b7937303fe199 : SUCCESS in 12m 31s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/dee0c5360e3945a8ac0bb8c672cc36f5 : SUCCESS in 41m 06s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b5c8c85abed434792e168b2d6d96cb8 : SUCCESS in 1h 09m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c393e384eafd44b69a512b5b01dbc5e6 : SUCCESS in 1h 06m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/482a179e44f54f6eaae00f7314dab9b7 : SUCCESS in 45m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/be07fe90d8c4497e9a718dde3e439570 : SUCCESS in 46m 53s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bf68868e2c7a49f2ac155e5597f8500e : SUCCESS in 31m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1304b4871bfd49af9ed0eea73ea13d1c : SUCCESS in 29m 54s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/eaeaed0038f34d578aed3e50f0b81d44 : SUCCESS in 2h 59m 11s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d7360b9b35514fdc90dcbfeab727267b : SUCCESS in 57m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a820b0aa0a8467fbaf24f945933b418 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":4},{"id":"ca0836229d637905d6a3e0a3ec47f88d57f725b7","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-28 04:48:00.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ca270318a8a82f515191c060375f098620a4f5e4","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-01-30 07:51:57.000000000","message":"Patch Set 4: Code-Review+1\n\nFormally ok, but I have no experience with designate or ironic.","accounts_in_message":[],"_revision_number":4},{"id":"d1475eda353d9927aaa3a1bcc5fafb7d9d7504d1","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-01-30 10:34:15.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8f2f6458c40c8f85ed95e3bec2c20a64baa0e9a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 10:34:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"76cafcd356d43b096519f37474d8b5251f4cbd42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:35:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66ef6c82baeb46fdb1a1b6b5857bd69e : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f505f76fb6a4cf8a52a48acbdc6c272 : SUCCESS in 12m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e757131b8c59434da04aa89f00c2b77a : SUCCESS in 14m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a54555b6e72b4b61a42da0974ea7bee7 : SUCCESS in 18m 29s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2937368d3148407ea331b64f0c4e8bfe : SUCCESS in 1h 02m 56s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9a435b67b2e47ce87b62cc3a8794f4f : SUCCESS in 1h 06m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a34258c9c30c430d843c6f607e5bf9b7 : SUCCESS in 45m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/29019d3dbe5a435fa2a9ff6e49aa0220 : SUCCESS in 44m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b846c485fb8647879e448c2101dfd4a8 : SUCCESS in 32m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/00062ccf690f4e4fb5ca1bff7f70349b : SUCCESS in 29m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/51c6ea5484aa4329acb60a9cfabc0743 : SUCCESS in 1h 06m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96e61e07e70b46b7a0db76198331f586 : SUCCESS in 10m 49s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7de3b69a22b04c0e8fdfea8f0cde00e1 : SUCCESS in 1h 51m 25s","accounts_in_message":[],"_revision_number":4},{"id":"d44fabee479decb021343d248bbae95018ca1416","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:35:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"cc9cd45560cdc2fb02a8184a1dd346879d770392","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:38:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2261605afbb41e8b934be0352441580 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"eaf0916e312b872f822410fdb732c1a727445543","revisions":{"6f0e861a3039689899354d5cd297677a483da1d0":{"kind":"REWORK","_number":1,"created":"2020-01-27 13:18:08.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/18/704318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/704318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/704318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/704318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/704318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/704318/1"}}},"commit":{"parents":[{"commit":"da93da4ac1da9c9d8e5f4c1d55861804ba1dec01","subject":"Merge \"Add json param to HTTPAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da93da4ac1da9c9d8e5f4c1d55861804ba1dec01"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-27 13:05:27.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-27 13:10:43.000000000","tz":120},"subject":"Fix fake clients in actions","message":"Fix fake clients in actions\n\n* Use inspector_url when creating a fake client for ironic inspector client\n* Add a session and a url for designate fake client\n\nChange-Id: I9cc78df13d0f0715538bbdb76c8ccad273bd2033\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6f0e861a3039689899354d5cd297677a483da1d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6f0e861a3039689899354d5cd297677a483da1d0"}]},"branch":"refs/heads/master"},"48a31e0ccc774dc6ccbdd5b6979271f1f214ae1d":{"kind":"REWORK","_number":2,"created":"2020-01-27 14:34:16.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/18/704318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/704318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/704318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/704318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/704318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/704318/2"}}},"commit":{"parents":[{"commit":"da93da4ac1da9c9d8e5f4c1d55861804ba1dec01","subject":"Merge \"Add json param to HTTPAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da93da4ac1da9c9d8e5f4c1d55861804ba1dec01"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-27 13:05:27.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-27 14:33:57.000000000","tz":120},"subject":"Fix fake clients in actions","message":"Fix fake clients in actions\n\n* Use inspector_url when creating a fake client for ironic inspector client\n* Add a session and a url for designate fake client\n\nChange-Id: I9cc78df13d0f0715538bbdb76c8ccad273bd2033\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/48a31e0ccc774dc6ccbdd5b6979271f1f214ae1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/48a31e0ccc774dc6ccbdd5b6979271f1f214ae1d"}]},"branch":"refs/heads/master"},"bf14e91f04ec1d802b664d28d3bd0322b4dfa677":{"kind":"REWORK","_number":3,"created":"2020-01-27 14:51:57.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/18/704318/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/704318/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/704318/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/704318/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/704318/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/704318/3"}}},"commit":{"parents":[{"commit":"da93da4ac1da9c9d8e5f4c1d55861804ba1dec01","subject":"Merge \"Add json param to HTTPAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da93da4ac1da9c9d8e5f4c1d55861804ba1dec01"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-27 13:05:27.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-27 14:48:04.000000000","tz":120},"subject":"Fix fake clients in actions","message":"Fix fake clients in actions\n\n* Use inspector_url when creating a fake client for ironic inspector client\n* Add a session and a url for designate fake client\n\nChange-Id: I9cc78df13d0f0715538bbdb76c8ccad273bd2033\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bf14e91f04ec1d802b664d28d3bd0322b4dfa677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bf14e91f04ec1d802b664d28d3bd0322b4dfa677"}]},"branch":"refs/heads/master"},"eaf0916e312b872f822410fdb732c1a727445543":{"kind":"REWORK","_number":4,"created":"2020-01-27 15:35:23.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/18/704318/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/18/704318/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/704318/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/704318/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/18/704318/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/18/704318/4"}}},"commit":{"parents":[{"commit":"da93da4ac1da9c9d8e5f4c1d55861804ba1dec01","subject":"Merge \"Add json param to HTTPAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da93da4ac1da9c9d8e5f4c1d55861804ba1dec01"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-27 13:05:27.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-01-27 15:35:06.000000000","tz":120},"subject":"Fix fake clients in actions","message":"Fix fake clients in actions\n\n* Use inspector_url when creating a fake client for ironic inspector client\n* Add a session and a url for designate fake client\n\nChange-Id: I9cc78df13d0f0715538bbdb76c8ccad273bd2033\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eaf0916e312b872f822410fdb732c1a727445543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eaf0916e312b872f822410fdb732c1a727445543"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
