)]}'
{"id":"openstack%2Fmistral~704943","triplet_id":"openstack%2Fmistral~master~I3df3ab96342c456429e20a905615b90bcb94818d","project":"openstack/mistral","branch":"master","topic":"improve_performance","hashtags":[],"change_id":"I3df3ab96342c456429e20a905615b90bcb94818d","subject":"Initialize profiler for scheduler threads","status":"MERGED","created":"2020-01-30 09:00:28.000000000","updated":"2020-01-30 19:12:53.000000000","submitted":"2020-01-30 19:11:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704943-1580411462500-641ca554","meta_rev_id":"755fe71503e6923d4be8d340c1a074d9d176c084","_number":704943,"virtual_id_number":704943,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 19:11:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2020-01-30 09:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-30 11:22:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-30 11:22:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-30 09:04:13.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2020-01-30 09:04:13.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2020-01-30 09:04:13.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-01-30 09:04:13.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-01-30 09:32:16.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-01-30 11:22:58.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-01-30 19:11:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ec27e2b1c360f37a1a19e6983c25ef21bb4e815","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-30 09:00:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b17fe78ca466eeb3d1a81b288c94f4141461891b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-01-30 09:27:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"460c716b93c5252a2d45f03cb94a9b5378aefbab","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-01-30 09:32:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"187c0b2f892399e08bc29c6165cb02aa0fd96246","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-30 11:22:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ca32b75ba4db63c569b7ed34cfd73708f86738b5","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-30 11:22:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"76ff0bb151d5550d050e53e5755de65100558368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:43:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/735a538c9a2543af8d2ac212e990b6e2 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9077563675c54b3d803bcf246b369aab : SUCCESS in 10m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/462e66f1f76e4583802ee6a52484788b : SUCCESS in 14m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8ff9d36ad9346abb0bfc0ee6acb632a : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bf3405ff20748d699e0707491350974 : SUCCESS in 14m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b41b0cc8f7964fc2a45e8a9240c51952 : SUCCESS in 14m 19s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7dbc591b1e19454987fdda7797cd11c4 : SUCCESS in 1h 02m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/518633ea68dd40279e4dfb327683dfcb : SUCCESS in 57m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55c170e81e064a2e8b841751644d0a2f : SUCCESS in 55m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1bfef8c784df4512bcb18ed219805907 : SUCCESS in 1h 03m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e3b970078c2d4dd2b179f84e7444f87e : SUCCESS in 1h 02m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/622bb006b7194e5d8460abc6bbf35198 : FAILURE in 29m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9b27925e67f044109d976361510ada22 : SUCCESS in 33m 15s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/648bb932cfbe4aa2b1fb348ff694b392 : SUCCESS in 2h 57m 58s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/37a797c16146481a8336eb5701d5563d : SUCCESS in 1h 17m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a87a8fe59bb47f9984a4e7ac8941d88 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":2},{"id":"7e40d14374db4e8e301eb4d257377cc66ed797f8","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-30 12:46:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"016d095bdfc041ca9778e2ed23bea31957c54f77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 15:13:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c075a098ff554e91917d98d027dd4e14 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62299538bfb84f2590832e1f63e15055 : SUCCESS in 10m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f99845b3acb40418b6b0b2a91f91a2e : SUCCESS in 10m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c702ca801d4b4af3a1b11a2278df606c : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a36ac08ce0342c1b6cc18de24b6a0df : SUCCESS in 10m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee4ef76b9e4c471e8512ff1ef0a748e2 : SUCCESS in 11m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9454d2c43378465f82051f0bdd74b74e : SUCCESS in 41m 17s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2665b0abbc4a40308d759936b9d05ca8 : SUCCESS in 46m 30s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03f9aa4c432348fbb655bd83601c11c7 : FAILURE in 48m 43s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b6a7fcf928994faea126b74e994f6ee2 : SUCCESS in 56m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/96c9eb56e0bc4af8a592c47c3d77cdc6 : SUCCESS in 41m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2361387ff2e047dcaaa69356d901ede0 : SUCCESS in 31m 43s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8a9ef4d793364fcda4ebdc0e09afb4f8 : SUCCESS in 30m 09s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/dfdfa584a8234ed68b67cef664d505bd : SUCCESS in 2h 24m 48s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7f1dae20d3a54afba6f510ec297287d2 : SUCCESS in 56m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d91a2b236a12472bb3c3b232e1af87ab : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c43f8390f0334bea9557e8074c2688b846eafe10","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-01-30 15:14:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a06d2dd4800ad8c9291220e37d635580f36a9c16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 17:43:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b5483151114d42872d6f710307adb9 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/097e1d48f7dd42f5804be1db15016ed7 : SUCCESS in 10m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9946c7d86c7447a3875853ad5f4edc54 : SUCCESS in 11m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/834e6617d5ae47578923ada6d4f0b2f6 : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d1a38023fd44948363ee0ab9c3bb3d : SUCCESS in 7m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f4855098bc349a7966bb16d74b3ed20 : SUCCESS in 11m 19s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d5c37b4742c94d23ba215886c1d98e16 : SUCCESS in 45m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ef2668dfb2f48c0b016ce832f91478b : SUCCESS in 43m 11s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/504badf9802540a4bc900587e8f20c46 : SUCCESS in 43m 30s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/eff774969dd0411482388e086515d4ab : SUCCESS in 52m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d31ceff558b342188881738595475bcb : SUCCESS in 1h 07m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8f0adae1b4e941eb93fc82b9d16ac548 : SUCCESS in 30m 49s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e9b100c11f1e4c2caafe10cd66202135 : SUCCESS in 29m 11s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9754ee40dc99413d999035a36c13943e : SUCCESS in 2h 26m 04s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0011fa706e724a25b2cf7e817ca71675 : SUCCESS in 1h 03m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/744a21fbd8a04c75b717d4b3bc2598b3 : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ce6e230535a49b7033e11a5510751da6ef1a0cda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 17:43:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e5b24204dcfc5e34a70e6e15bcfa64dc79951e00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 19:11:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3201f38f05874be5b0348b32280d6bfd : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a89379c576304752a454625c5791f456 : SUCCESS in 8m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd9b153ddfca4bda910bf9e499af1412 : SUCCESS in 10m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/909df4ee4d4540c1ad890ccce50e24a8 : SUCCESS in 12m 38s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/97c8f0e227164c8c8534e4ccf720b1be : SUCCESS in 52m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b705adc4db1c4c82acb26b76a9c63c9b : SUCCESS in 46m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dea9ab0dabb4bf59572803bbda67f62 : SUCCESS in 49m 28s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c6460cdc05d74dbfb542cdcd84ab102e : SUCCESS in 45m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c5f2a638bea84aba83a91cae7cbabe19 : SUCCESS in 34m 29s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0a12e2884b8f4268ac4f54296912f544 : SUCCESS in 33m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1ac3b685996b4d6f953f17dcab3bc11e : SUCCESS in 55m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f2e4066e649462da5fd4d1f2e4d6268 : SUCCESS in 10m 53s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c595ceacca624896b6e3490acd8342f4 : SUCCESS in 1h 25m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e95de2bd3c1bc9dc5e69a3c54048348375d400c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 19:11:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7c5db0d6689ee34275d89e8f778e5bb5f6cfdb7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 19:12:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59dfdb0943204c449913a91b1bb0369b : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"511be4f98f27eacda8755ceb4cce360ffbb66b4c","revisions":{"b50739be46da618f295b1925557227298012229e":{"kind":"REWORK","_number":1,"created":"2020-01-30 09:00:28.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/43/704943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/43/704943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/43/704943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/43/704943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/43/704943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/43/704943/1"}}},"commit":{"parents":[{"commit":"da93da4ac1da9c9d8e5f4c1d55861804ba1dec01","subject":"Merge \"Add json param to HTTPAction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da93da4ac1da9c9d8e5f4c1d55861804ba1dec01"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-30 08:57:47.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-30 08:57:47.000000000","tz":420},"subject":"Initialize profiler for scheduler threads","message":"Initialize profiler for scheduler threads\n\n* It turns out that osprofiler wasn\u0027t initialized properly for\n  threads in which scheduler runs its jobs. So profiling simply\n  didn\u0027t work for this threads and lots of important info didn\u0027t\n  get to the profiler log. This patch fixes it.\n* Other minor style fixes according to the Mistral coding guidelines.\n\nChange-Id: I3df3ab96342c456429e20a905615b90bcb94818d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b50739be46da618f295b1925557227298012229e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b50739be46da618f295b1925557227298012229e"}]},"branch":"refs/heads/master"},"511be4f98f27eacda8755ceb4cce360ffbb66b4c":{"kind":"REWORK","_number":2,"created":"2020-01-30 09:27:35.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/43/704943/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/43/704943/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/43/704943/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/43/704943/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/43/704943/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/43/704943/2"}}},"commit":{"parents":[{"commit":"680a9e2e0791ddd1acf8968be641903a12b993f0","subject":"Merge \"wrapped the value of parameters in inputs(in wf get API) with \"\".\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/680a9e2e0791ddd1acf8968be641903a12b993f0"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-30 08:57:47.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-01-30 09:23:56.000000000","tz":420},"subject":"Initialize profiler for scheduler threads","message":"Initialize profiler for scheduler threads\n\n* It turns out that osprofiler wasn\u0027t initialized properly for\n  threads in which scheduler runs its jobs. So profiling simply\n  didn\u0027t work for this threads and lots of important info didn\u0027t\n  get to the profiler log. This patch fixes it.\n* When evaluating a YAQL expression, sometimes we get a huge result\n  value that we always put into the debug log. In practice, it doesn\u0027t\n  make much sense and, moreover, it utilizes lots of CPU and disk\n  space. It\u0027s better shrink it to some reasonable size that would\n  allow to make necessary analysis, if needed.\n* Other minor style fixes according to the Mistral coding guidelines.\n\nChange-Id: I3df3ab96342c456429e20a905615b90bcb94818d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/511be4f98f27eacda8755ceb4cce360ffbb66b4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/511be4f98f27eacda8755ceb4cce360ffbb66b4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
