)]}'
{"id":"openstack%2Fmistral~705635","triplet_id":"openstack%2Fmistral~master~I247b674d8a358795871cfa87bcdf29f4857ca2d8","project":"openstack/mistral","branch":"master","topic":"improve_performance","hashtags":[],"change_id":"I247b674d8a358795871cfa87bcdf29f4857ca2d8","subject":"Set the delayed call \"key\" field to the right value","status":"MERGED","created":"2020-02-04 11:39:33.000000000","updated":"2020-02-05 16:34:53.000000000","submitted":"2020-02-05 16:31:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705635-1580920261397-a46f390a","meta_rev_id":"8174c4a84eaa18b1a93ce61ccdd177d981d430bd","_number":705635,"virtual_id_number":705635,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"date":"2020-02-05 14:29:57.000000000","_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 16:31:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":2,"date":"2020-02-05 11:54:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":2,"date":"2020-02-05 14:16:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-05 09:10:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":1,"date":"2020-02-05 11:54:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-04 12:26:02.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2020-02-04 12:26:02.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2020-02-04 12:26:02.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-02-05 09:10:46.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-02-05 11:54:03.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-02-05 14:16:51.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-02-05 16:31:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"390f8ac5abc3ecc1a0a3a2be8db948f2d1f1bbe8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-04 11:39:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6220b2735c252d6abbc01f7a939ee72c286605e5","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-04 11:39:53.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1e7082f8616dab5d5f0c9c39a6833dfa3965d65b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-04 11:45:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d141fcc82fa2a30a9a2fed4cef26b957ff2f0140","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-02-04 12:54:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"db916bdb621eb290fb449600bdf1e622497155e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 14:03:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90bfb26e40a646c4b9d3b00537f1d9cd : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1bafce3ab7343c1b297d6cdc83c6e92 : SUCCESS in 10m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e0ce1148e454c1b838dcf5ce436d35b : SUCCESS in 12m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6affdeb73dff4540a09b2b79e46c3363 : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d275e075ff07421583741c36f8b4ee92 : SUCCESS in 8m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f89b01e81c1e4324b52f30a6b2393be7 : SUCCESS in 9m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8d8aa068fc1d4fac92042a0f3dcc3bcb : SUCCESS in 50m 38s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06806c7cc0764f1e96425bad9c9753aa : SUCCESS in 59m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63974cf200fe4262b4fd40f1c6ba467c : SUCCESS in 43m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/80d84203ad1644d0987050931fe47580 : SUCCESS in 41m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/20d56ce3c06a44d3975d94b81648e2c4 : SUCCESS in 42m 50s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/03b5042df5fc45ccb8a36432a3ab1ecc : SUCCESS in 31m 08s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a5d394d338bb46d6b30ef07a32ea5196 : SUCCESS in 29m 10s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3c687f428b9449818f38bc24dc531f81 : SUCCESS in 2h 17m 18s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/311383a041294c1eb435b85e088b7124 : SUCCESS in 54m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cc8769ae563422aaa95cf7e95ba97d0 : SUCCESS in 11m 53s","accounts_in_message":[],"_revision_number":2},{"id":"06d0649b8760b8eda109b8fd39da0427b198822c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-05 09:03:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"82a06e882afcbd749673811d169b54b50f2eaea4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 09:03:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"08847f837d199b3e85efc801a6a392941acd1926","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-05 09:08:15.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 05e38c278ccdc2579cc6cb121f0b75e3004c17bf","accounts_in_message":[],"_revision_number":2},{"id":"ef4d7e839460d90bc52cd00c2383ca8cf225ed46","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-05 09:09:54.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"368338c39d2ed264cc3a4f6489809ad62954227b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-05 09:10:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"697731686943232b54f51bf89f0c0dd4145c9f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 11:40:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3427ec1c24ca430f88569bd4105642f7 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5c2905e161f4d619c0b6fb1b1811158 : SUCCESS in 11m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10e78b86f59c4cd597182c42bfadfb78 : SUCCESS in 11m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c28c84055124a14a9231209dcc1121f : SUCCESS in 12m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd1ba2354d745b281860ed77f649c8f : SUCCESS in 8m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6dd8f6b044b74fd4a30570f566500584 : SUCCESS in 11m 05s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fac72d76115d4ccd91efc5a709e7bef8 : FAILURE in 50m 58s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/768bc9facf8a4a28bba83ce9c239e5dd : SUCCESS in 43m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/990a5dc66a234c749b832d42bdcb104a : SUCCESS in 45m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2aefa6e6cfef46d9876cd6df09569f34 : SUCCESS in 55m 14s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a71931c0653841838520d462e4f47dcb : SUCCESS in 40m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/45c19da7aad04e2d867e75f2c49501a0 : SUCCESS in 30m 02s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/589be6c4a22a4348903ffde5a2c2f686 : SUCCESS in 31m 20s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7ed1a7484e394722bb659253dd1e072c : SUCCESS in 2h 24m 18s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d515a7bc65e14dceaa65d1024341b05d : SUCCESS in 1h 08m 17s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4957a65fc4d045639303fdb4a784fdf7 : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":3},{"id":"62ec28f70014604f4888e37fe9a69685a001b1f5","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-02-05 11:54:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"702613c342c4d78c55aac54e83984c42af9f1247","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-02-05 14:16:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"876b0857df65186ed4a6cf4265a1efa44d99cb82","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-05 14:29:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7da4ed2ef76d05de27f0b21c0b6f8b62a7f36967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 14:42:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73eb472b89b4acd9fec82a45adeb46b : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b04343c29918485fb311ad24faf81c4f : SUCCESS in 10m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d80bfb176fa4efe9080b80dfa2b9997 : SUCCESS in 9m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d698d81d93a4f82ba7a6cf7c181c734 : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/412f0b3242614bba879d96a54272c117 : SUCCESS in 8m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/944a90f1f83d4cfe882491615b322751 : SUCCESS in 10m 09s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8de90352d8474312a9cd7e3467fb4020 : SUCCESS in 48m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e40e1653b99841bea348c0ffc07ff297 : SUCCESS in 42m 18s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c158f0d8b2d145bbbcf596495a8eb0b0 : SUCCESS in 44m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/88a107f8f3344c8589316723fd1cd8cf : SUCCESS in 49m 34s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/eda58c7a5bf14bdf94ae3d7df6b3903d : SUCCESS in 49m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a8919f669a9c470b98ca033128e49750 : SUCCESS in 29m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/694fa3d815b9430cae81e61c4fc083c0 : SUCCESS in 28m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6a2bf42633714b53a55f573b55e93593 : SUCCESS in 2h 40m 01s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0f83dd2e8e1b4b59b2e4035a2a4905c0 : SUCCESS in 1h 04m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68626acf7900418bb3469af8b1c247c6 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":3},{"id":"2fea684aff4ac373f8bc7df8690fe62074a86189","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 14:42:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"691d0e02ce83d2780333c6ff52425472e9ba3d42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 16:31:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6848bd9ee79c889f051e1132bec2df69643b18f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 16:31:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a04afa939ee484aa9739ece65813f73 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/310545b4cdd4413dbc052949bb407595 : SUCCESS in 11m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4284a7fc78854b629f561a18d691cdf0 : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d18f1957cb6b40d286fa68bc388ae2f7 : SUCCESS in 10m 12s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1b0ddb82e3914546ae9845d4d42341bc : SUCCESS in 51m 23s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cc81e7c08164a329d81457bee2b0d50 : SUCCESS in 56m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5b2a6d843964711ab280d1a4df72261 : SUCCESS in 46m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5128f721de6e4fa39ddde090d330d520 : SUCCESS in 51m 35s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/43754de2221a4b649a8a7686127a414b : SUCCESS in 31m 08s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e6dd09b4566d4b6883d5ea4b520fc410 : SUCCESS in 29m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/bf6b32c355d04f46aad5aa46faf82afd : SUCCESS in 52m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a9d18bad5614a1aabdc58f3a6a067e3 : SUCCESS in 12m 13s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7710d0d545504e32b597fe50a0d61f16 : SUCCESS in 1h 42m 01s","accounts_in_message":[],"_revision_number":3},{"id":"d4d558520da92bdc78364f598c4d2422b0a6ed7f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 16:34:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dcc78524fec40cd9d603caa8181df14 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5b5576dd04ac596b619ee61e0aa837effd319c49","revisions":{"84a546f13303c25f97de2bfffa4aa8246db2b2bb":{"kind":"REWORK","_number":1,"created":"2020-02-04 11:39:33.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/35/705635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/35/705635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/705635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/705635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/705635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/35/705635/1"}}},"commit":{"parents":[{"commit":"6dd891a173a0b83bb4f1c41e4971fd38f1bdddc8","subject":"Merge \"Initialize profiler for scheduler threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6dd891a173a0b83bb4f1c41e4971fd38f1bdddc8"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-04 11:36:15.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-04 11:39:00.000000000","tz":420},"subject":"Set the delayed call \"key\" field to the right value","message":"Set the delayed call \"key\" field to the right value\n\n* The field \"key\" of the DelayedCall class must be set to the right\n  value. It allows to significantly optimize performance of workflows\n  with \"join\" tasks. However, it was broken during the refactoring\n  in the summer of 2019. This patch fixes it.\n* Added another profiler trace decorator.\n\nChange-Id: I247b674d8a358795871cfa87bcdf29f4857ca2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/84a546f13303c25f97de2bfffa4aa8246db2b2bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/84a546f13303c25f97de2bfffa4aa8246db2b2bb"}]},"branch":"refs/heads/master"},"6e10414dc74fdce75af5c10668f70d5745fca538":{"kind":"REWORK","_number":2,"created":"2020-02-04 11:45:08.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/35/705635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/35/705635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/705635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/705635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/705635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/35/705635/2"}}},"commit":{"parents":[{"commit":"6dd891a173a0b83bb4f1c41e4971fd38f1bdddc8","subject":"Merge \"Initialize profiler for scheduler threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6dd891a173a0b83bb4f1c41e4971fd38f1bdddc8"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-04 11:36:15.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-04 11:44:54.000000000","tz":420},"subject":"Set the delayed call \"key\" field to the right value","message":"Set the delayed call \"key\" field to the right value\n\n* The field \"key\" of the DelayedCall class must be set to the right\n  value. It allows to significantly optimize performance of workflows\n  with \"join\" tasks. However, it was broken during the refactoring\n  in the summer of 2019. This patch fixes it.\n* Added another profiler trace decorator.\n\nChange-Id: I247b674d8a358795871cfa87bcdf29f4857ca2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e10414dc74fdce75af5c10668f70d5745fca538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6e10414dc74fdce75af5c10668f70d5745fca538"}]},"branch":"refs/heads/master"},"5b5576dd04ac596b619ee61e0aa837effd319c49":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-02-05 09:09:54.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/35/705635/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/35/705635/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/705635/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/705635/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/35/705635/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/35/705635/3"}}},"commit":{"parents":[{"commit":"6dd891a173a0b83bb4f1c41e4971fd38f1bdddc8","subject":"Merge \"Initialize profiler for scheduler threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6dd891a173a0b83bb4f1c41e4971fd38f1bdddc8"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-04 11:36:15.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-05 09:09:50.000000000","tz":0},"subject":"Set the delayed call \"key\" field to the right value","message":"Set the delayed call \"key\" field to the right value\n\n* The field \"key\" of the DelayedCall class must be set to the right\n  value. It allows to significantly optimize performance of workflows\n  with \"join\" tasks. However, it was broken during the refactoring\n  in the summer of 2019. This patch fixes it.\n* Added another profiler trace decorator.\n\nCloses-Bug: #1861988\nChange-Id: I247b674d8a358795871cfa87bcdf29f4857ca2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b5576dd04ac596b619ee61e0aa837effd319c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b5576dd04ac596b619ee61e0aa837effd319c49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
