)]}'
{"id":"openstack%2Fmistral~705900","triplet_id":"openstack%2Fmistral~stable%2Ftrain~I247b674d8a358795871cfa87bcdf29f4857ca2d8","project":"openstack/mistral","branch":"stable/train","topic":"improve_performance-stable/train","hashtags":[],"change_id":"I247b674d8a358795871cfa87bcdf29f4857ca2d8","subject":"Set the delayed call \"key\" field to the right value","status":"MERGED","created":"2020-02-05 09:08:15.000000000","updated":"2020-02-10 13:18:26.000000000","submitted":"2020-02-10 13:16:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705900-1581340601731-9ea6cd4b","meta_rev_id":"2d3934a5aca85a61420c96ca683052b283bef831","_number":705900,"virtual_id_number":705900,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 13:16:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-02-06 04:17:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-02-06 04:17:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 13:16:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be12ebc1c75611db2388005de69d13c513c9367e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-05 09:08:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0a787693c1872d9fbe61c56620addc1b1f164fca","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-05 09:09:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0056fbe1e6d58fd7b001d28c50f2e96477bfd93c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 11:37:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eb24bb8ba124c2c83c8c4365b26699f : SUCCESS in 7m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8498fbf286c048859f7199502038757c : SUCCESS in 9m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67038c4c63394c089bdb468615330592 : SUCCESS in 10m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2914b55c0eb40c9bf809fa38fa6a199 : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bdb1255bb904a278908d501f71d7361 : SUCCESS in 5m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27c61e9a10aa4157a756a01c1ed9c78d : SUCCESS in 10m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6b3432a2c0d84675afa9770de37a3984 : SUCCESS in 45m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/316339ab639f406f8521ea9d9b8601cd : SUCCESS in 42m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/96190157e5cb4670bad72f70e8fb8bad : SUCCESS in 43m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6fc8a670bc584f59857199cbc4245b9a : SUCCESS in 30m 16s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8590d599c68748c38ee7886720cf7653 : SUCCESS in 29m 43s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/d96c0d4d8f8547f5905bd9836d57aeff : SUCCESS in 1h 15m 17s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d04b99223ce64ef498b3ec1248967f3d : SUCCESS in 2h 21m 33s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f5024c9cbb854faf9fd993345bf0f194 : SUCCESS in 1h 11m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcdd89d055c24b3999a4d5d3501322da : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":2},{"id":"aa985c070e6cf730e24efa562e5682ca3f92b204","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-06 04:14:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f7155773d17d8de0f8d92ded1c6e96b6332d6b05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 04:14:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7babd03ef0d92098d6f7f51b4613bcde6c77754e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-06 04:17:15.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f315ec9f117f09d98cadc8e7d51eaab8b8f9e5a1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-06 04:17:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6b452963ee5b3e9c5a857313d1a86d5628bbd058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 06:28:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdc7289ae4774ea9b660fff61f87cfc4 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/133788f314e0490282dff6232fa6a0b3 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95491a5ec7d748459fbbf5ee4d3ec67f : SUCCESS in 8m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e906078d2a44954bdd693a92a3ce998 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c7a5e42d780437ea932e1edc70c4c15 : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07280c03f9d847ada1c9c4324a719eff : SUCCESS in 8m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/caece7724db6436d9ba2145a24b146c9 : SUCCESS in 44m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1479682b1024493b82cf03e821dea303 : SUCCESS in 41m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8734c723fc17428f94894f67fede3f99 : SUCCESS in 43m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6f9c1562d3224ba88660c7c45832c941 : FAILURE in 28m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/587106ff55474e23b6f6674cf1264820 : SUCCESS in 28m 56s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/aa85054f12444b81896ef5c3cb073222 : FAILURE in 1h 13m 11s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/719f1e581994499d8e28157e8a688ff3 : FAILURE in 2h 09m 51s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/237429638901427b85ebf4bb0beed278 : SUCCESS in 1h 02m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c544f2b2ca94c83a892336936e2015d : SUCCESS in 10m 11s","accounts_in_message":[],"_revision_number":3},{"id":"70c71036f221d8aeb541ac12be589733d984e2c6","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-06 07:23:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"13fcfa11a40047ebf486349f0dc1ad335c45556d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 09:27:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27c43c8a7cca4416a0ef1f91b9e65b4f : SUCCESS in 8m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/024373757ae0496889ab3f46f625cfd0 : SUCCESS in 9m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62662325fee64c2986178e84d65bf424 : SUCCESS in 9m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d109ce9548634296aeae581ad4786533 : SUCCESS in 9m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5f68ffed16e4e4e9906fbc4efc6688f : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff4dbb06abb049f8ab93f2cf0f604e60 : SUCCESS in 9m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/89e96c9b164e42b6aab6d42b87ccbab3 : SUCCESS in 47m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c164ee9321444ca2845007657d952fd8 : SUCCESS in 46m 52s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/effd1bdd20e84c6db022080384d63a6b : SUCCESS in 48m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2dc6ce9895044ff0922e46794b969d74 : SUCCESS in 29m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1c699a8b1f314dd1994150327574ada4 : SUCCESS in 28m 03s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/5f54bfb366084a92985cb67f4ffa9faf : FAILURE in 1h 19m 20s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1654132a7e0f42a4b3cdd60c493e5650 : FAILURE in 2h 02m 55s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7b180567140347259fe4aeb97caef64d : SUCCESS in 1h 10m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dde809c5f9ec448eb3fb42efd14b0f68 : SUCCESS in 10m 53s","accounts_in_message":[],"_revision_number":3},{"id":"e13b9b730e60a0eb51050fd2b32a954b68c5ce8c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-10 04:19:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5a6d1f546a33f6e672a29d9dd193e1cad7f8360c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 07:17:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83a120981a4441359152a17e190a2128 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/133cde6ab4294a05bd6e36211f27074f : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a61af7bc0a68464db884c8698ff3129c : SUCCESS in 8m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cdcaa66caa53469e9cd89db1e8ace3e1 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43da13856b24450eb41b9a684b3f46f1 : SUCCESS in 4m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f8c32e0fef54b80823f0e456f9392c1 : SUCCESS in 9m 09s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ea6bae912979455ca3f41b2511c24897 : SUCCESS in 44m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/df5a2dd7eafa49a4ab7674e1ecfee694 : SUCCESS in 43m 13s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0ec55fa732f94accbf366db1d42c467e : SUCCESS in 43m 34s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a9bae5872c7f478685b58ae55a0b16b1 : FAILURE in 29m 12s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7fae756422bb42a88503f0eb6c9918ed : SUCCESS in 28m 08s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1f26f89c757649fda7220a82e529a7e8 : SUCCESS in 1h 27m 29s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/6863eae618bf47f1b9e6fb42b4630dc3 : SUCCESS in 2h 53m 09s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ea50f04148714c80addbf7f342fe6930 : SUCCESS in 1h 02m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff4979cff3864a65ab3f4b7beb50c31c : SUCCESS in 11m 05s","accounts_in_message":[],"_revision_number":3},{"id":"47a3269eb3603e8686ac58fe66e240abed6a4f08","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-10 08:33:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d3ae6edc56830605240051f33f76aabaf1914f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 11:35:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1520e63df0db4f54b2e1ca0621ac5391 : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/202d3d8d99f242979c6bd942a9759add : SUCCESS in 11m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28a2e15fff6d45c0b16a935968df4e93 : SUCCESS in 11m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4042f64ba4114cf490d65a2fdc7cb0e5 : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ae51fb3b61d4b30af189761c8b271a5 : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e2d643f582843f59cfed093c13a53b4 : SUCCESS in 10m 32s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d834864db28942bcb05dd0c0e1c489c6 : SUCCESS in 48m 44s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/bec0ab5f06c04943bdb19ea8338aa8c7 : SUCCESS in 47m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4e3ea3b91b5c4f62aeeb285919823b25 : SUCCESS in 45m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2de9dfe28ca1418a928b4067dfc13179 : SUCCESS in 28m 41s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6623b5ea6e0b418c9141181d91294296 : SUCCESS in 29m 36s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/1856dadd50fd484d9ab4db299d58e0ab : SUCCESS in 1h 34m 35s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/813c6bd71a114387b9811a43ddf423fc : SUCCESS in 3h 00m 31s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/048263d7fa74451cbe81951ed3a1c758 : SUCCESS in 1h 03m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c93ac58ee3a4499d9a987230fe6fad5b : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":3},{"id":"1a298323bd5fd4dec915568ae3e3d60066ae534b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 11:36:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"351a93e1f647315d0078e6624422863041661f8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:16:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39b042069f0a42f289a819c767c6c6d8 : SUCCESS in 7m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8982a76d93b248168f25313b564e9278 : SUCCESS in 10m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0770ed63a655475cb333dbb29d6ccdcc : SUCCESS in 10m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a67de0687cd491d92128b310b4ee600 : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad72eb8bc47406ea70781ce4beb9df9 : SUCCESS in 5m 58s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c8cd17bae2e146d2a3b8707974ecc5a3 : SUCCESS in 46m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8d31c46c1f064eb8a4961219277a154e : SUCCESS in 46m 22s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e16f1277dcef4e4695bb23e71373c31f : SUCCESS in 31m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/016d060ef01e4a74962432641c4fecc8 : SUCCESS in 27m 31s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/59550aee0e3242eda03bc9c5fe68e461 : SUCCESS in 1h 21m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/53eca041e1b94a5eb32eb31a5e2aa525 : SUCCESS in 44m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ddf28d5b24f4bd88ff68407b00192f9 : SUCCESS in 11m 51s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b6808f8fe29e4a5fb00c3c6c9659751e : SUCCESS in 1h 35m 17s","accounts_in_message":[],"_revision_number":3},{"id":"1f814248860c72b4db2d67ce61f941b51b99039c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:16:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"94b3f6f1939af857d7952d793b1104986e5dc527","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:18:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c55ae4f0389a4f56bac8f6f28704a215 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7c54c28c4c38ad1a71bf4388ffb1fe70200b373e","revisions":{"05e38c278ccdc2579cc6cb121f0b75e3004c17bf":{"kind":"REWORK","_number":1,"created":"2020-02-05 09:08:15.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/00/705900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/00/705900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/00/705900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/00/705900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/00/705900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/00/705900/1"}}},"commit":{"parents":[{"commit":"6651519919df81265ee442b06fe467f17f3c0a3d","subject":"Merge \"Make sure minimum amqp is 2.5.2\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6651519919df81265ee442b06fe467f17f3c0a3d"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-04 11:36:15.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-05 09:08:15.000000000","tz":0},"subject":"Set the delayed call \"key\" field to the right value","message":"Set the delayed call \"key\" field to the right value\n\n* The field \"key\" of the DelayedCall class must be set to the right\n  value. It allows to significantly optimize performance of workflows\n  with \"join\" tasks. However, it was broken during the refactoring\n  in the summer of 2019. This patch fixes it.\n* Added another profiler trace decorator.\n\nChange-Id: I247b674d8a358795871cfa87bcdf29f4857ca2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/05e38c278ccdc2579cc6cb121f0b75e3004c17bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/05e38c278ccdc2579cc6cb121f0b75e3004c17bf"}]},"branch":"refs/heads/stable/train"},"2ef8bfb56feb232ce7f8890a84a8fc01fe06dc4c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-05 09:09:01.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/00/705900/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/00/705900/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/00/705900/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/00/705900/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/00/705900/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/00/705900/2"}}},"commit":{"parents":[{"commit":"6651519919df81265ee442b06fe467f17f3c0a3d","subject":"Merge \"Make sure minimum amqp is 2.5.2\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6651519919df81265ee442b06fe467f17f3c0a3d"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-04 11:36:15.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-05 09:08:57.000000000","tz":0},"subject":"Set the delayed call \"key\" field to the right value","message":"Set the delayed call \"key\" field to the right value\n\n* The field \"key\" of the DelayedCall class must be set to the right\n  value. It allows to significantly optimize performance of workflows\n  with \"join\" tasks. However, it was broken during the refactoring\n  in the summer of 2019. This patch fixes it.\n* Added another profiler trace decorator.\n\nCloses-Bug: #1861988\nChange-Id: I247b674d8a358795871cfa87bcdf29f4857ca2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ef8bfb56feb232ce7f8890a84a8fc01fe06dc4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/2ef8bfb56feb232ce7f8890a84a8fc01fe06dc4c"}]},"branch":"refs/heads/stable/train"},"7c54c28c4c38ad1a71bf4388ffb1fe70200b373e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-02-06 04:17:15.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/00/705900/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/00/705900/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/00/705900/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/00/705900/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/00/705900/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/00/705900/3"}}},"commit":{"parents":[{"commit":"6651519919df81265ee442b06fe467f17f3c0a3d","subject":"Merge \"Make sure minimum amqp is 2.5.2\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6651519919df81265ee442b06fe467f17f3c0a3d"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-04 11:36:15.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-06 04:17:04.000000000","tz":0},"subject":"Set the delayed call \"key\" field to the right value","message":"Set the delayed call \"key\" field to the right value\n\n* The field \"key\" of the DelayedCall class must be set to the right\n  value. It allows to significantly optimize performance of workflows\n  with \"join\" tasks. However, it was broken during the refactoring\n  in the summer of 2019. This patch fixes it.\n* Added another profiler trace decorator.\n\nCherry-picked from 5b5576dd04ac596b619ee61e0aa837effd319c49\n\nCloses-Bug: #1861988\nChange-Id: I247b674d8a358795871cfa87bcdf29f4857ca2d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7c54c28c4c38ad1a71bf4388ffb1fe70200b373e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7c54c28c4c38ad1a71bf4388ffb1fe70200b373e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
