)]}'
{"id":"openstack%2Fmistral~706234","triplet_id":"openstack%2Fmistral~master~I93c1e9ed166847aea07531f98a9924a728efbab3","project":"openstack/mistral","branch":"master","topic":"extend_notification_objects","hashtags":[],"change_id":"I93c1e9ed166847aea07531f98a9924a728efbab3","subject":"Extend task and workflow notification data","status":"MERGED","created":"2020-02-06 11:12:13.000000000","updated":"2020-02-25 14:41:36.000000000","submitted":"2020-02-25 14:39:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706234-1582641594914-1d1e633c","meta_rev_id":"276bdb66087f18fa766a113720ca3976f4a45dcb","_number":706234,"virtual_id_number":706234,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-25 14:39:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"recommended":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"all":[{"value":1,"date":"2020-02-25 08:52:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2020-02-21 03:52:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-20 14:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2020-02-21 03:52:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 14:11:03.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-02-21 03:52:12.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-02-25 08:52:50.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-02-25 14:39:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c7c86b375b8df1dc1f5a91502fa7c4febcaaeb2","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-02-06 11:12:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05e9ef6e171809ed2bed19d5b76256f02d6751e0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-06 11:17:29.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f88623e629fbd6d07424000d29ff2d69d122c0d","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-02-06 12:12:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1827e9aafb1e14a0236664ef84a86f7b99113220","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-02-06 12:13:07.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\ndone!","accounts_in_message":[],"_revision_number":2},{"id":"bcc162f7246818a1bd3dd22fcaebc27416ad0e0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 14:26:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff7854baeacc45fea59935f135b70a7d : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4787e9e0cb44411cb62f22202ef5f080 : SUCCESS in 9m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/434b4e028486417aa1cb869c8a6e405e : SUCCESS in 11m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1677f380bbb54d1f9f243748c09aee16 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dcc40b3c61f44c7a89abadd97a25530 : SUCCESS in 11m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae2022e5e26f4dc981e1e03131d0bd9d : SUCCESS in 12m 10s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/151f95b1eb7247978ed351bcc4e6a664 : SUCCESS in 42m 23s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/044fa450b4bd4597889f56de826039f4 : SUCCESS in 48m 32s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e1454184d38436b87246ce7100afcfe : SUCCESS in 49m 13s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0270aad5bdee406e8a0ca4b6489f886f : SUCCESS in 46m 27s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/25bab9520e02488686561ae8a1abdc0f : SUCCESS in 46m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fd5dffc67b90481f9a86a61bd449ff6b : SUCCESS in 30m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/fe100b4451914eefb36f1aafaeedbd99 : SUCCESS in 28m 42s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/00642d2a5b2147af8b5620649efa710a : FAILURE in 2h 06m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ac23f09187a14e3f8d5a54ad773a8886 : SUCCESS in 1h 07m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3aa445bc796545e89903d39ce23b5dba : SUCCESS in 14m 51s","accounts_in_message":[],"_revision_number":2},{"id":"1c53187366a5c3e6bfb2052954742e903740723c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-07 04:00:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1541919e92f993dbd3341392815a681150a6a51f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-07 04:01:35.000000000","message":"Patch Set 2:\n\nOleg, is there a test for this that we can modify?","accounts_in_message":[],"_revision_number":2},{"id":"6131948e6695b3140f9ca2bcd26e9cd01cf87f0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 06:50:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3057517486d54b088ee397d9a117e23a : POST_FAILURE in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26df28582d5e4dd78f342c2dca60040f : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f20789e80a264bb38baf79514bae5ae1 : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/829e73663cd6415cad4cc984a0ffc306 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a0331229844f88be73cb8a49c898a6 : SUCCESS in 6m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a052a75b64844bbc9ec811e5be058cc2 : SUCCESS in 8m 36s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7959c915381c425ba8d0bd5588ee5d2b : SUCCESS in 40m 37s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9fc726fb30f4a849e900e38c42bb340 : SUCCESS in 41m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3751dbc7b07f4ff4b6ed53890d4e57d9 : SUCCESS in 51m 09s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/52036d25c0b9482cb48c8bf2e39328dd : SUCCESS in 41m 51s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5e154957be6d40b0b70ad538f5b4f2d0 : SUCCESS in 37m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b4b06ce8d1684538ae6bbbd22fe693d8 : SUCCESS in 28m 22s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/08b71f38ade74a15a016ed755c93a2ac : SUCCESS in 29m 33s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1d699bae674246dfa38d7231fb77c9bf : SUCCESS in 2h 45m 37s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/07a5ec0a9cf3408a8ecc9be558305838 : SUCCESS in 50m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fdedef4ff544da4b86d4984d4e50383 : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":2},{"id":"f25cf5437aa595e150c68f5d322898693a297b03","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-20 10:55:46.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e173f2b2b519377850f0307c454c842c28e1c391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 14:09:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd039529e27d4afc95baa22e472b3318 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba8c80a9f40148a2a54ac849bcc88828 : SUCCESS in 10m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7928dec48e234e6aacd1cfc2189596a2 : SUCCESS in 10m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06c503953d2a49e790b059b5b9ba97cd : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b737dd307e4981b349c21da227c7a9 : SUCCESS in 9m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8826238189e64187a3a58f780e0d4fdb : SUCCESS in 10m 33s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6a1400a72e4043e0a01c8f4fcf0db61c : SUCCESS in 45m 10s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17f2e0d94bad4723adae2f9ca56078d0 : SUCCESS in 45m 47s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c0ed839613943459083efca6d1c5ffc : SUCCESS in 44m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1abbdfa308cb4ea1ab2c6e287170d825 : SUCCESS in 44m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/be21ac9917914279a0181bb7d45da44e : SUCCESS in 45m 31s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b23a5fd7009947a595fb0d46e314e01f : SUCCESS in 30m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d0985f3d13bf4eeda6d9544e1df6e856 : SUCCESS in 29m 10s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f2483cb0a0d74c74a2964a6387feb76b : SUCCESS in 3h 08m 00s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/bf5ba35577794760bd650a490ee35c98 : SUCCESS in 1h 01m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8add14bfe3164868aef600b4ab4f34dd : SUCCESS in 12m 33s","accounts_in_message":[],"_revision_number":3},{"id":"c58cf9dda69e5e1a23997bf025c2157374d88e15","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-20 14:11:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"79be5e8c9963acbdd10dde646a455340e0f0972a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 03:52:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"89155151302fb18b3f0f6bbaf8b28dff1f87f71c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 03:52:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bb3467c6f780ede50e4a7e72cb8f20de4d5316f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 05:33:52.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/312196eec7e3466cbb24dfe8dbb696e6 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40848ffd8afe49d8b0aee9972905cc47 : SUCCESS in 8m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2dda19ead644de9a218263a7d77d33b : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1105e887e00f4d1fae49949da720e95a : SUCCESS in 8m 07s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4b087d73c0564493bd4cdea8c452c6f1 : SUCCESS in 50m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cae6575cdd44f7abf445862d526f159 : SUCCESS in 41m 49s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae6eec0bba874ede9fb26776c65ddebb : SUCCESS in 44m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/788ae56cb1524e60864f14175a26f0f8 : SUCCESS in 41m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a4489e9b240145d3a6296261fb84bbfd : FAILURE in 29m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3627357039ce48f99ae69db60913f826 : SUCCESS in 30m 35s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1ba1f3c4121d4bee9bf867903739f547 : SUCCESS in 45m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05dd3a57086f4102abd3160998f71394 : SUCCESS in 12m 04s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a24bd7b260c94ffca9a9cc5a221b6898 : SUCCESS in 1h 37m 05s","accounts_in_message":[],"_revision_number":3},{"id":"737d743e96871fabd34a1d10d0b2d369ecf6852d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 06:37:38.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ca343a26842c57196dc1df0ee86b059b6aabb277","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 09:30:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d6baf19a2bd4495b27f0d3a637ff723 : SUCCESS in 9m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e35ee6636e05409e922e84f77ab8fdde : SUCCESS in 14m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cfac3a3b10bb48008c954709a0582ae4 : SUCCESS in 12m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9785a8e6c286400fbd2a1685eec10016 : SUCCESS in 11m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eeb88f5217748a7b80b2b965299e06a : SUCCESS in 13m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05a181b94a7c4a7689bcbe6f0b3b60eb : SUCCESS in 13m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3eb9e2a0b8de4b94a7078fc3fde0d292 : SUCCESS in 44m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c147ce1c0ebc4f4a8b17289d5a99ea4a : SUCCESS in 48m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39786b3536734b47b1b16faeeeb6720f : SUCCESS in 47m 34s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dab8af4bbc0d46f388a0612579b8f3f1 : SUCCESS in 46m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/27c2d396da8b445ba3dd136fcf8a230f : SUCCESS in 46m 30s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/5f37716fcf18424f9e005b7a693c687b : FAILURE in 35m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/30dfdab7d3ec40beb3213d93759072b3 : SUCCESS in 33m 19s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/eacb70b8c3f242d48b75fa06adb44ec9 : SUCCESS in 2h 47m 32s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/51be67a911b24f03b7da5161693716d8 : SUCCESS in 1h 21m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59bcdb08e62b4669b3f601e8b4b5d23e : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":3},{"id":"c885bd15dfb54cdff1bce0651004b356624a02e1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 10:13:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cfb7f25f4ca9b6d2a691b49aad88020c73e24b0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 12:50:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0947dc0d26348528252b12a1409052d : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a56bbcc807046ca8bc2568b7e70cb53 : SUCCESS in 10m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8119e5a1075048bd9810763334459e6f : SUCCESS in 11m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82eef1c9906345348c9a8e6a14ee604b : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd7450e3e2e4339a22181bc1d5332a7 : SUCCESS in 9m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d5f2852eede4a02b6bfe336953596a4 : SUCCESS in 10m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bb5149e16c6641c58dd1a86baffb5d3b : SUCCESS in 1h 03m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6df53b5c886d4afebdf19bda2762ad44 : SUCCESS in 45m 24s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09b215f5b40648d8b4e57990081f5b4e : SUCCESS in 1h 03m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ef8ab255b89e461a82a759e5da641f85 : SUCCESS in 51m 29s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/055f6feb732f40e598119e4b0964c309 : SUCCESS in 48m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/32b2817c5344460a9d87fcedb08e3111 : SUCCESS in 31m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0a7e92ffefd348c68fee1d78f9d75278 : SUCCESS in 30m 06s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f3d3352ed7e44713895b6abf743a0372 : SUCCESS in 2h 32m 51s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6e5829c010014b199ef4d280c945da15 : SUCCESS in 1h 01m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9d72c8349654a60939f75bf08108341 : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":3},{"id":"b67875496038da543408c8940cc0e7a8a004d7be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 12:51:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5352bbff24a22aaa5482e5b55c7845b5abc66aed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 14:53:18.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d2a3fd20f94245ab44320d805eae43 : SUCCESS in 9m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c25d4304d08044ef8f65cd67a4af3a1e : SUCCESS in 12m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9748a6b37aa2432c8059d3e315a599e4 : SUCCESS in 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef59fc137d9a4b1c8a64b49f3709b30e : SUCCESS in 8m 58s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1109bf68eb294c4b9c44bdf1e9b34be5 : SUCCESS in 38m 24s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/346beba043cc4579a476cd2478448fa6 : SUCCESS in 54m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86e0c5bcb8224beab4f501d6be0ad838 : SUCCESS in 48m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/28dae6169f2b4db9bd0069f93adfc8a7 : SUCCESS in 42m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c5048d8233854effa6dfc80757cb58d6 : FAILURE in 35m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/37ec49fc62c1410ca80ae2fb7c7854b9 : SUCCESS in 31m 14s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b8abbaec578d4ee5aef2d26614351268 : SUCCESS in 46m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56cffc2828824842a54303bbf517d6f8 : SUCCESS in 13m 57s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8668d65fc5454a36be42c6a6cc1088a3 : SUCCESS in 1h 59m 43s","accounts_in_message":[],"_revision_number":3},{"id":"c59c40830220b80a7f16c270f809d626019a4d5f","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-02-25 08:52:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ccc745f688ec1c9ad70526d244dd090a994c3e44","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-02-25 08:52:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"560ae97c680d5be444ab263f347547beca0f01e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 12:48:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0754d530eb53403fbc32a3586f25846a : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6025c4bbd0fb401ab87a4adbb8f22784 : SUCCESS in 10m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4101e75cba334509a3c01a27265aa8f9 : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66976850eed74e7ba93325bdf3a3001f : SUCCESS in 11m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f53a81512b8c435e8993c1483a358361 : SUCCESS in 10m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d821516af0704704b409558a98e33785 : SUCCESS in 12m 14s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1e7bae8ba0c14d639f70d30f778e3ce6 : SUCCESS in 57m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55fc439c1bd84a12902967e299536c36 : SUCCESS in 56m 08s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec0d3ebaf331499ea6f01bb8862be3fb : SUCCESS in 56m 14s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/df565795a5734455b18ac4f4d7fc2d4d : SUCCESS in 49m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4923d56ca3cd4528a17f8f18e7bdabab : SUCCESS in 48m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b9524f393f7b46ecbbe8798024ad489e : SUCCESS in 30m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/153b423d0cc5467ea322017763380202 : SUCCESS in 29m 51s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/02a17bc3a8504f79a149d4dd02467a10 : SUCCESS in 2h 53m 30s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f076695a5afb468e9bd0c338d4a1030b : SUCCESS in 1h 03m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0aed1a0b8feb40e7a63d4ffce5946d57 : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":3},{"id":"1ae06478037ecd668e2d68a6e043c8bf567d0809","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 12:48:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9f2763f1ed862245a81bff0d270a35e2d3c4b302","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:39:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e1cbdcd625a2975374e215a0d400b8fea2873b87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:39:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811aa9c486584c97918598d39b5eaaee : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0681d606fe443b99e394b5044f58465 : SUCCESS in 9m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e826843d149f4ba2a6bb0285b2f21447 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6327b09429946dba324ff0b7d789dcd : SUCCESS in 12m 04s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/18f11a787a854fc6a1d9f364341cc993 : SUCCESS in 51m 31s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66483b857a614a1c83da8c6da9aa3c0c : SUCCESS in 54m 20s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/668ae4ff5d984591a0ee50361c53a7fc : SUCCESS in 51m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/98d8d0e5df9e42158d1546ff19b170b9 : SUCCESS in 51m 46s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fdbf9a15890b4aa583e4e4a706a23b08 : SUCCESS in 29m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f76919420d474894b36535b8ad95e277 : SUCCESS in 28m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/99846c905b804cb5bc5af2a3432672a9 : SUCCESS in 45m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f6fdb6ed33643ba82a0d82bf985cf43 : SUCCESS in 11m 18s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e08ed3ec0cd84de3a5e63ac1d9cb8ec2 : SUCCESS in 1h 45m 37s","accounts_in_message":[],"_revision_number":3},{"id":"92a12a8a6b7a8190047204a58288591d0e28de8f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:41:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a911f1b4b8d442beb884b7d6de27d2cb : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"95d9f899dbb971fbbf766023c24a1a538ee65d0e","revisions":{"7edcd712ed713b9d4d3a452f4d549726880652d1":{"kind":"REWORK","_number":1,"created":"2020-02-06 11:12:13.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/34/706234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/706234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/706234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/706234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/706234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/706234/1"}}},"commit":{"parents":[{"commit":"5b5576dd04ac596b619ee61e0aa837effd319c49","subject":"Set the delayed call \"key\" field to the right value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b5576dd04ac596b619ee61e0aa837effd319c49"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:09:05.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:10:51.000000000","tz":180},"subject":"Extend task and workflow notification data","message":"Extend task and workflow notification data\n\nChange-Id: I93c1e9ed166847aea07531f98a9924a728efbab3\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7edcd712ed713b9d4d3a452f4d549726880652d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7edcd712ed713b9d4d3a452f4d549726880652d1"}]},"branch":"refs/heads/master"},"b3f7bbe432fc08a49c1fce12368cffa03d3ad2f6":{"kind":"REWORK","_number":2,"created":"2020-02-06 12:12:50.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/34/706234/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/706234/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/706234/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/706234/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/706234/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/706234/2"}}},"commit":{"parents":[{"commit":"5b5576dd04ac596b619ee61e0aa837effd319c49","subject":"Set the delayed call \"key\" field to the right value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b5576dd04ac596b619ee61e0aa837effd319c49"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:09:05.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 12:08:07.000000000","tz":180},"subject":"Extend task and workflow notification data","message":"Extend task and workflow notification data\n\nChange-Id: I93c1e9ed166847aea07531f98a9924a728efbab3\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b3f7bbe432fc08a49c1fce12368cffa03d3ad2f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b3f7bbe432fc08a49c1fce12368cffa03d3ad2f6"}]},"branch":"refs/heads/master"},"95d9f899dbb971fbbf766023c24a1a538ee65d0e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-20 10:55:46.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/34/706234/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/706234/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/706234/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/706234/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/706234/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/706234/3"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:09:05.000000000","tz":180},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-20 10:55:46.000000000","tz":0},"subject":"Extend task and workflow notification data","message":"Extend task and workflow notification data\n\nChange-Id: I93c1e9ed166847aea07531f98a9924a728efbab3\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/95d9f899dbb971fbbf766023c24a1a538ee65d0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/95d9f899dbb971fbbf766023c24a1a538ee65d0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
