)]}'
{"id":"openstack%2Fmistral~706245","triplet_id":"openstack%2Fmistral~master~If9f0c9bea587b486998af1c18e282bedba453499","project":"openstack/mistral","branch":"master","topic":"bug/1862161","hashtags":[],"change_id":"If9f0c9bea587b486998af1c18e282bedba453499","subject":"Fix incorrect in-depth search of affected tasks","status":"MERGED","created":"2020-02-06 11:43:26.000000000","updated":"2020-03-11 08:14:44.000000000","submitted":"2020-03-07 16:51:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706245-1583599877153-c6fdaf72","meta_rev_id":"1cc9fbb0d0acd1721c3ae1bd3f7da2992eaf4ffd","_number":706245,"virtual_id_number":706245,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-07 16:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":2,"date":"2020-03-06 04:09:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"date":"2020-03-06 11:00:54.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-07 12:51:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 04:09:36.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-03-06 11:00:54.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-03-07 12:51:04.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-03-07 16:51:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"265a2f1dd148c90cceb1f13d69221716220be996","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-02-06 11:43:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd97685a7ed7408a076938a8bf62e71034ee6d88","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-02-06 12:30:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e60b2d59d11cf733c0cfb3f56efdcc6653948a6b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-06 13:12:53.000000000","message":"Patch Set 2: Code-Review-1\n\nI would add a unit test for that","accounts_in_message":[],"_revision_number":2},{"id":"edd2e61d82925be33c74e3131111fa58629a79b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 14:48:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15669b6d95e24136ac88facef5ab6ed8 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c73128ff5b9e4cfaaed5415804428309 : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0535bb1e9407452a8d7221ff37642b91 : SUCCESS in 9m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2189a668ef2f410db0b52868e992ce3a : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17886ed95e9846b1bde49de6a26caa4c : SUCCESS in 7m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccc3e4a4f15a489e9b2fb98e7651efa2 : SUCCESS in 9m 15s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a1375f0d292748ab897e5fcbc6280cd1 : SUCCESS in 1h 38m 48s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03df48082dfd4d74832e9f2c23c4aa77 : SUCCESS in 50m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f92fc256132c4a15b0781265e70ef3c4 : SUCCESS in 43m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/591c1cff0e304d62ace4b20481cda398 : SUCCESS in 48m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c17dc6ecb51e472c96df827977b37268 : SUCCESS in 44m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/34eba42f312041b382fb36aeee0c738c : SUCCESS in 29m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7d5d22caae0742468f6edfc9143ffa71 : SUCCESS in 29m 16s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3d009c0ac9dd46b3b0166108a3a3ebeb : FAILURE in 2h 13m 36s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9adc12351a354864a7c9bcbffb5bc011 : SUCCESS in 1h 03m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8bd990a317c43169645699355dbad49 : SUCCESS in 11m 25s","accounts_in_message":[],"_revision_number":2},{"id":"8a8390c30cd9d6a54208987a90b8ddc811018c4a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-07 04:06:12.000000000","message":"Patch Set 2: Code-Review-1\n\nGreat catch, Oleg! Thanks for fixing this. I agree with Eyal that we need a test. \n\nIn general, for this kind of logic that is very important and complicated at the same time we can\u0027t afford landing patches w/o tests. I always try to write a test first that fails and then fix it.","accounts_in_message":[],"_revision_number":2},{"id":"f6b33b4e29ee61edcee8d78d5980af16f19deb01","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-03 04:51:31.000000000","message":"Patch Set 2:\n\nCan you please proceed with this patch?","accounts_in_message":[],"_revision_number":2},{"id":"402db2b131f340532a59a532d0afd053d83de9e8","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-03-03 12:41:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"de8f1a5aaf8c725b0a26bfe36cd89dd679166f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 18:46:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c2aa1b7a8c44b1097eef20963541c0a : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54b9db46f7e040cb8787dd89b0d905c2 : FAILURE in 11m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/593cb6c75329497b8528f315d4c367a3 : FAILURE in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57e94b7c5b3f42b787f85e871942a67c : FAILURE in 12m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe46521f85e483293e2e4c1778a0231 : SUCCESS in 10m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a2236ec616a44a4b4a7ee73a370763c : FAILURE in 12m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8ab0c04f07174cb7955ed43936445d0e : SUCCESS in 56m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e98301c3b6e45c280f630e62eae21d8 : SUCCESS in 53m 28s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efeeeb364d104fdd821070ef1770aeca : SUCCESS in 53m 46s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/797430fe0cb04f509fe4fd3b9e0d8bfe : SUCCESS in 56m 07s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6373170b4c054e78991679de08616c2b : SUCCESS in 54m 24s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e9bef722fdd1480a8b326600b3d32505 : FAILURE in 30m 54s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/92563abc906643a3aeae2c4650f693fd : FAILURE in 30m 29s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c2b70bc8ac034c0385f3edd4f5b1df0f : SUCCESS in 3h 02m 32s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f4125e31827542c7a978d93f51d4070c : SUCCESS in 1h 00m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea50f4c6affc4fa5af8349afa3b55a69 : FAILURE in 14m 08s","accounts_in_message":[],"_revision_number":3},{"id":"6393e091ea0d46221329e299f7bdf289bd1fa1f6","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-03-04 10:56:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"20a535401b0c9aff3324c23bc9b03c3b2e4c78ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 13:37:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3abf941a69224f559d382cdb4cb22d2a : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49e5434acfef47718cc3fbd08885da57 : SUCCESS in 10m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ac5a6dcea3b4cfeada5a74ffd218539 : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc6c991d82494ad5bf51d994c5fe8ea5 : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03d1c0e2bc0b4040af748e5441f71070 : SUCCESS in 8m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37f8e61a28804a24aa6888e1439f8321 : SUCCESS in 10m 41s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fbb7e95bfe8548f491c49f43aacba73a : SUCCESS in 48m 39s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0c3e8afe92c4e76aff42908c637f5e9 : SUCCESS in 51m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6cb397c1f9f41bc80442befff95d3d0 : SUCCESS in 46m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0457203b3038488b9395d46a5c1a5266 : SUCCESS in 49m 20s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4dc2bd16786f43e3a872f5429308c396 : SUCCESS in 48m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4fbd1114e9ca49709f476eb6705ded27 : SUCCESS in 33m 03s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/67c58dfcb8774c598f8c97b913da0995 : SUCCESS in 29m 08s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/21bf88ffd2cc4130abde84e1a9cfb00d : SUCCESS in 2h 36m 57s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b4b16d4f04b64dd98bda782567e21562 : SUCCESS in 1h 04m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ea047dacf6f47f4b1d94c2cc39d44e5 : SUCCESS in 12m 05s","accounts_in_message":[],"_revision_number":4},{"id":"0363d30e3521fcb4c42f9a03ccbe70a0a9b50a43","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-05 05:03:39.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2507ca0a346847cbb7690ccb9665707d3eff684b","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-03-05 11:02:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"96f86939a9cf41e28a58c672d6806b9fbfeef294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 12:12:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d43b805b4cd741d997555b12bb8bee94 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88693142d45d48b298c6deb7c6196ac4 : SUCCESS in 10m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc04b5ee65f34412ae68088df4494b59 : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3605d34e8e814b018446fbdb9647a59e : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4394e41bcb544cd6b03b18d3bad12533 : SUCCESS in 12m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1da817af29914e1db287fa8a15a0bf33 : SUCCESS in 12m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a15b82c59e8e487799bebe3d48afa9c0 : SUCCESS in 41m 52s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b01a5ce28424fbda58bc286b41eaddd : SUCCESS in 49m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85089d1c8dc8453fb787d3e17db3b03f : SUCCESS in 50m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/a895e33ce01b442899931eeb79369652 : SUCCESS in 50m 19s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e5c3d586c46c4cbeb1dc4587d7fa8d52 : SUCCESS in 49m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8b998cbeed8d43808b5f1585f75e99f8 : SUCCESS in 29m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5fdc3835b79647e1a8e6d7af504dca0b : SUCCESS in 28m 46s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b705fc902666431892e4b755946f00cf : SUCCESS in 1h 02m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63fbb9276f8b40938f7fa46a8f3b7798 : SUCCESS in 10m 40s","accounts_in_message":[],"_revision_number":5},{"id":"79d03507e434e7936b764927ad85f0e330686bcb","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-06 04:09:36.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5f9882c1bc1eb0a7252d614e20111b05ce104b15","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-07 12:51:04.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6160b83b21bdce03d31e3c741f00db13cbc13c7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 12:51:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c1ae86f47fc989396c1c3e1b5727a6c684fdd562","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 13:40:01.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffc21c36fada4490a9ceb51212ddedf1 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07c8aa239c8a4e45b6eff200ebee970b : SUCCESS in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e5fb8fe9ad545b6937d29e2f45224df : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17bd1f63a17e491a8cecaef5f0fe0003 : SUCCESS in 5m 56s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a5311edc8f0b46faa7048e44e1bbeca9 : SUCCESS in 41m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39841d5379de46ccb9534b45fc4d7156 : SUCCESS in 40m 39s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d6d8219030c4cdfa2fe08733f9da8ea : SUCCESS in 43m 31s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/89b2ce05cab841a6a90eb7aa0d387a26 : SUCCESS in 41m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/88c221ecc7be42559178b3106d98a275 : FAILURE in 27m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5363414b91c14789a3f500ea62046ca8 : SUCCESS in 26m 43s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/40f2396648a743efb6420085934cb3db : SUCCESS in 37m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac46762583c94690a134d5d13543028e : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":5},{"id":"571d4ba184cfabaca60ba24276aa58b8b3f04c2a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-07 14:58:10.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c0f16d1952b0333f26174bceabb7fe8afd538c7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 16:00:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6804e74319a49468d6bfd4cafade2cc : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e60d64dd4d2b43d8b17691d31d391d2d : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7dcdd8f80914de59dccc850fb6e6644 : SUCCESS in 9m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/340ca46c166f48679e2d13e03e2fcbf6 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b610bde36bc4dcf9a4bc8c8f0b3d394 : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f6d6899cf814640927a452ef2fccf7b : SUCCESS in 11m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/53a5603c958645738078b841804ffae5 : SUCCESS in 40m 44s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ef08923c3c843979d3d64b73bec9093 : SUCCESS in 44m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d688c14d9c4a4c2cb233b04af8eda771 : SUCCESS in 43m 40s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6bc251a3995647ea8266be6546698225 : SUCCESS in 43m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4e78af15a0104e80bef5391d68d14190 : SUCCESS in 43m 27s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f714970e46534803b3da254f32397cd2 : SUCCESS in 27m 37s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9a60ccd6e78f428f9c4de1a5446bd3d9 : SUCCESS in 28m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a41f92cdcc714229a52b578589729932 : SUCCESS in 57m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9cd46dc479c74a789de679d3ab95c142 : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":5},{"id":"3863fa4aab40975233e3d479c7e946789f2f8447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 16:00:59.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"deeef0dee6fcce81e00f6e60c51dba9c2e042c59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 16:51:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89c7df74165b4bca88e011d477df69b2 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7683e1840bd54bb78e5cfea521c2313f : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37dd92729640454ebc9dd8093f7c5bbe : SUCCESS in 8m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c4dab998b240d384b37541c7d5f377 : SUCCESS in 6m 53s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/ee3d553b7ab3450d9cc2876ced38b828 : SUCCESS in 44m 32s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9eb1c492491e47faa4f8a82b4ba0b5f2 : SUCCESS in 44m 12s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9606917261354f2da99a07c901c92453 : SUCCESS in 44m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3eb579fef3684988940a7db0e2226a8a : SUCCESS in 43m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/76f6d388cac3458d91dc57f13a6582ec : SUCCESS in 28m 29s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/593acbc7520046a4a06d80a26dc57651 : SUCCESS in 27m 04s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f0fc898e7c1f43e2b2a256dd9aa338a7 : SUCCESS in 45m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e4e1a527f0e435d94847351af25aeb0 : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":5},{"id":"bb39704a8ab151d68fe98abee0d1c835d32f9489","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 16:51:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"0529d8aa353d62fe973cd04c433183f34853c389","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 16:53:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04b1f16cd16b449da836de6ff96d132f : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":5},{"id":"aba488b649b7c84b7255b93222f11b3697fbe33a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-11 08:14:44.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit c99c345fb59edd16a461d18674809b6f43db7c53","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"de633d5d48f48258bac214089391e4cda5b51f34","revisions":{"a6d694f397c33e15777043c7fc2d93c2565c63bd":{"kind":"REWORK","_number":1,"created":"2020-02-06 11:43:26.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/45/706245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/706245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/706245/1"}}},"commit":{"parents":[{"commit":"5b5576dd04ac596b619ee61e0aa837effd319c49","subject":"Set the delayed call \"key\" field to the right value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b5576dd04ac596b619ee61e0aa837effd319c49"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:43:00.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:43:00.000000000","tz":180},"subject":"Fix incorrect in-depth search of affected tasks","message":"Fix incorrect in-depth search of affected tasks\n\nChange-Id: If9f0c9bea587b486998af1c18e282bedba453499\nCloses-Bug: #1862161\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a6d694f397c33e15777043c7fc2d93c2565c63bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a6d694f397c33e15777043c7fc2d93c2565c63bd"}]},"branch":"refs/heads/master"},"f2ea02d873ee3fcdc6456b555525706398b83b0f":{"kind":"REWORK","_number":2,"created":"2020-02-06 12:30:23.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/45/706245/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/706245/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/706245/2"}}},"commit":{"parents":[{"commit":"5b5576dd04ac596b619ee61e0aa837effd319c49","subject":"Set the delayed call \"key\" field to the right value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b5576dd04ac596b619ee61e0aa837effd319c49"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:43:00.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 12:28:44.000000000","tz":180},"subject":"Fix incorrect in-depth search of affected tasks","message":"Fix incorrect in-depth search of affected tasks\n\nChange-Id: If9f0c9bea587b486998af1c18e282bedba453499\nCloses-Bug: #1862161\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2ea02d873ee3fcdc6456b555525706398b83b0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2ea02d873ee3fcdc6456b555525706398b83b0f"}]},"branch":"refs/heads/master"},"bad1b90a15dee6d5ebb0b1a76d99ea6ddc8dd6b2":{"kind":"REWORK","_number":3,"created":"2020-03-03 12:41:17.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/45/706245/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/706245/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/706245/3"}}},"commit":{"parents":[{"commit":"1217e4ca61c03e2f86a4ff1ea75f516ae2114a22","subject":"Merge \"Add caching for YAQL/Jinja expression functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1217e4ca61c03e2f86a4ff1ea75f516ae2114a22"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:43:00.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-03-03 12:41:00.000000000","tz":180},"subject":"Fix incorrect in-depth search of affected tasks","message":"Fix incorrect in-depth search of affected tasks\n\nChange-Id: If9f0c9bea587b486998af1c18e282bedba453499\nCloses-Bug: #1862161\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bad1b90a15dee6d5ebb0b1a76d99ea6ddc8dd6b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/bad1b90a15dee6d5ebb0b1a76d99ea6ddc8dd6b2"}]},"branch":"refs/heads/master"},"32f5edb70193aa02c78b9951793b5521638ed483":{"kind":"REWORK","_number":4,"created":"2020-03-04 10:56:40.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/45/706245/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/706245/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/706245/4"}}},"commit":{"parents":[{"commit":"1217e4ca61c03e2f86a4ff1ea75f516ae2114a22","subject":"Merge \"Add caching for YAQL/Jinja expression functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1217e4ca61c03e2f86a4ff1ea75f516ae2114a22"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:43:00.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-03-04 10:56:19.000000000","tz":180},"subject":"Fix incorrect in-depth search of affected tasks","message":"Fix incorrect in-depth search of affected tasks\n\nChange-Id: If9f0c9bea587b486998af1c18e282bedba453499\nCloses-Bug: #1862161\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/32f5edb70193aa02c78b9951793b5521638ed483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/32f5edb70193aa02c78b9951793b5521638ed483"}]},"branch":"refs/heads/master"},"de633d5d48f48258bac214089391e4cda5b51f34":{"kind":"REWORK","_number":5,"created":"2020-03-05 11:02:06.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/45/706245/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/45/706245/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/45/706245/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/45/706245/5"}}},"commit":{"parents":[{"commit":"1217e4ca61c03e2f86a4ff1ea75f516ae2114a22","subject":"Merge \"Add caching for YAQL/Jinja expression functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1217e4ca61c03e2f86a4ff1ea75f516ae2114a22"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:43:00.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-03-05 11:00:22.000000000","tz":180},"subject":"Fix incorrect in-depth search of affected tasks","message":"Fix incorrect in-depth search of affected tasks\n\nIn case if the join task is not exist (every upstream task have\nERROR state) workflow will stuck in RUNNING state because of\nthe bug in affected tasks search. This patch fixes this bug.\n\nChange-Id: If9f0c9bea587b486998af1c18e282bedba453499\nCloses-Bug: #1862161\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/de633d5d48f48258bac214089391e4cda5b51f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/de633d5d48f48258bac214089391e4cda5b51f34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
