)]}'
{"id":"openstack%2Fmistral~706247","triplet_id":"openstack%2Fmistral~master~I88c930386f14ef6f17191ff0005bfda235c87ca6","project":"openstack/mistral","branch":"master","topic":"bug/1862164","hashtags":[],"change_id":"I88c930386f14ef6f17191ff0005bfda235c87ca6","subject":"Enable conversion task_result if convert_input_data is turned on","status":"ABANDONED","created":"2020-02-06 11:57:09.000000000","updated":"2026-03-28 19:40:18.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"953bf2b2b5c2275204a3a957cf2bec0e10393f35","_number":706247,"virtual_id_number":706247,"owner":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-06 14:53:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":-1,"date":"2020-02-07 03:58:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 14:53:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-07 03:58:44.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"}],"messages":[{"id":"7bf8e8da435e72cc20d6510d03a07d1829568539","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-02-06 11:57:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"256b06310ac5188b4de2fc6c40ebb46beb4fbf0f","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-02-06 11:58:15.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c35fe9b0bce99fde2c2e5b58645649fcc97d4d38","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-02-06 12:31:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7e74b15e8ea221786835125f16fb756c3e6936a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 14:53:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac2fcfcf084243d68f71d1869a3fdfbd : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/519d4b28263c499fb824fe23fba8d9bb : FAILURE in 12m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4590d84c1d824b0c9e4ec5e7522cb229 : FAILURE in 15m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b8879cf2c0a4b9f9015c64fdfc2f5fa : FAILURE in 15m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d707f9739d0f4384b434b756d39e9ebc : SUCCESS in 15m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ed88d55bf93471b827c94c2ef9b7184 : FAILURE in 10m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8a4d072b46644849ab13e30bc5f84f24 : SUCCESS in 1h 02m 22s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3bfd3545a59451db93a55ca49a07de7 : SUCCESS in 44m 41s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ac6101abbe54769a35539e56cca1751 : SUCCESS in 45m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/6b3cb3e7e0d64dedb8d03068ba1deaec : SUCCESS in 46m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d8d31a704df34e3994fa78027fa9d651 : SUCCESS in 43m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ae8935e7e17c428f823689bc60556d90 : FAILURE in 31m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/df0551fa64fc45c6b05d470ae7e5fbd9 : FAILURE in 28m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/201604d028624016b1840ecbc0bd2d4d : FAILURE in 2h 05m 02s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/afc9fea325aa4324b72b527e6cc97c09 : SUCCESS in 59m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72873650e8794bd58083fb424e661255 : FAILURE in 11m 32s","accounts_in_message":[],"_revision_number":3},{"id":"0eb22efb0a3152ec7751c58ae5a524d87ac6bb8c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-07 03:58:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nOleg, I\u0027d like to see more verbose commit message (or more detailed description of the corresponding bug). What was broken, in what cases it appeared, what example reproduces it, and similar things. And how this patch fixes it.","accounts_in_message":[],"_revision_number":3},{"id":"b3b4798b4c2d45bcd88e8916b7d1fadd1bc14329","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-07 03:59:14.000000000","message":"Patch Set 3:\n\nAlthough the patch is pretty simple, consequences may be serious.","accounts_in_message":[],"_revision_number":3},{"id":"2235e5d796cf841d2626fc7e83a90a339fdbd807","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-03 04:51:07.000000000","message":"Patch Set 3:\n\nCan you please proceed with this patch?","accounts_in_message":[],"_revision_number":3},{"id":"953bf2b2b5c2275204a3a957cf2bec0e10393f35","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-03-28 19:40:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"88f55b4213870282ae2a5788ee4a70af197214e5","revisions":{"8ef82b6cd5471df9b3d6ced851a145a5c48f230b":{"kind":"REWORK","_number":1,"created":"2020-02-06 11:57:09.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/47/706247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/706247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/706247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/706247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/706247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/706247/1"}}},"commit":{"parents":[{"commit":"5b5576dd04ac596b619ee61e0aa837effd319c49","subject":"Set the delayed call \"key\" field to the right value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b5576dd04ac596b619ee61e0aa837effd319c49"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:56:33.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:56:33.000000000","tz":180},"subject":"Enable conversion task_result if yaql.convert_input_data is turned on.","message":"Enable conversion task_result if yaql.convert_input_data\nis turned on.\n\nChange-Id: I88c930386f14ef6f17191ff0005bfda235c87ca6\nCloses-Bug: #1862164\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8ef82b6cd5471df9b3d6ced851a145a5c48f230b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8ef82b6cd5471df9b3d6ced851a145a5c48f230b"}]},"branch":"refs/heads/master"},"b43805f98b6c6744609205f32bc3e27abfeebf99":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-06 11:58:15.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/47/706247/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/706247/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/706247/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/706247/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/706247/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/706247/2"}}},"commit":{"parents":[{"commit":"5b5576dd04ac596b619ee61e0aa837effd319c49","subject":"Set the delayed call \"key\" field to the right value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b5576dd04ac596b619ee61e0aa837effd319c49"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:56:33.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:57:36.000000000","tz":180},"subject":"Enable conversion task_result if convert_input_data is turned on","message":"Enable conversion task_result if convert_input_data is turned on\n\nChange-Id: I88c930386f14ef6f17191ff0005bfda235c87ca6\nCloses-Bug: #1862164\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b43805f98b6c6744609205f32bc3e27abfeebf99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b43805f98b6c6744609205f32bc3e27abfeebf99"}]},"branch":"refs/heads/master"},"88f55b4213870282ae2a5788ee4a70af197214e5":{"kind":"REWORK","_number":3,"created":"2020-02-06 12:31:25.000000000","uploader":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"ref":"refs/changes/47/706247/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/706247/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/706247/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/706247/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/706247/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/706247/3"}}},"commit":{"parents":[{"commit":"5b5576dd04ac596b619ee61e0aa837effd319c49","subject":"Set the delayed call \"key\" field to the right value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5b5576dd04ac596b619ee61e0aa837effd319c49"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:56:33.000000000","tz":180},"committer":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 12:31:12.000000000","tz":180},"subject":"Enable conversion task_result if convert_input_data is turned on","message":"Enable conversion task_result if convert_input_data is turned on\n\nChange-Id: I88c930386f14ef6f17191ff0005bfda235c87ca6\nCloses-Bug: #1862164\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/88f55b4213870282ae2a5788ee4a70af197214e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/88f55b4213870282ae2a5788ee4a70af197214e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
