)]}'
{"id":"openstack%2Fmistral~707798","triplet_id":"openstack%2Fmistral~master~I9270317c12eaa2d42536768e52051ed6cb35b5e4","project":"openstack/mistral","branch":"master","topic":"bug/1862940","hashtags":[],"change_id":"I9270317c12eaa2d42536768e52051ed6cb35b5e4","subject":"Resolving an error with an invalid region in keystone client.","status":"ABANDONED","created":"2020-02-14 09:53:46.000000000","updated":"2026-03-28 19:40:17.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"16bd7e1d7d5cd78863944ccd4ffed59696783380","_number":707798,"virtual_id_number":707798,"owner":{"_account_id":31596,"name":"Ilya Zubov","email":"zubov@selectel.ru","username":"nortlite"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-05 05:46:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-19 06:59:07.000000000","updated_by":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"reviewer":{"_account_id":7700,"name":"Nikolay Mahotkin","email":"nikolay.makhotkin@gmail.com","username":"nmakhotkin"},"state":"REVIEWER"},{"updated":"2020-02-19 06:59:07.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2020-02-19 06:59:07.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-02-19 06:59:07.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-02-19 06:59:07.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-02-19 07:00:12.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-02-19 15:23:59.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2020-02-25 13:53:59.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-03-05 05:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32c6db04389a0922b54544f8812432d8b4f8f653","author":{"_account_id":31596,"name":"Ilya Zubov","email":"zubov@selectel.ru","username":"nortlite"},"date":"2020-02-14 09:53:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cda21db9d39a49e4d7db7d065f830a2f89ba6849","author":{"_account_id":31596,"name":"Ilya Zubov","email":"zubov@selectel.ru","username":"nortlite"},"date":"2020-02-14 09:54:47.000000000","message":"Topic changed from fix_keystoneclient_wrong_region to bug/1862940","accounts_in_message":[],"_revision_number":1},{"id":"b4fd006051f02208a4605792a7cd1d9c42e9a181","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-14 10:01:08.000000000","message":"Patch Set 1:\n\nThanks, when it\u0027s merged to master we can backport it to train and stein.","accounts_in_message":[],"_revision_number":1},{"id":"e49d2f1f6bd1ecb9a4155d53cda16a0f480e19a3","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-02-14 10:13:41.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e6cd611ba2583aaa245a2c6ac3748ff54ed32055","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 12:42:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd7e7f55e2ae48448bd219393216097b : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/823f00d5a2af491d8b1b39dd33e57917 : SUCCESS in 12m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5ac468e2a454e4c8576b69568a9b7aa : SUCCESS in 13m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9591cce592084d02b55d70a2afe292ef : SUCCESS in 10m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c688179b0524487483923d578198116e : FAILURE in 6m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e27586c49564f9d9c347092470f0c97 : SUCCESS in 11m 26s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5713af3f8030487fb762ac873742acb9 : SUCCESS in 43m 41s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccfc57f7bd7744a3a7b3ae4ad4d161a4 : SUCCESS in 43m 26s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/602c041cb4774b0a9067ac76e50d9597 : SUCCESS in 53m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/89d293b78d23402cbbc7ec6fb75ca94e : SUCCESS in 49m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1171b17c2a3b4313b763f3ffd13068f0 : FAILURE in 22m 20s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7fbba448db354a50816c8c89639e10a6 : SUCCESS in 33m 04s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f6c5c1a01e754f4097a4b13a537dc042 : SUCCESS in 30m 06s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d2a8d3760f554877b5dff79929e0a286 : SUCCESS in 2h 42m 45s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/44610651a5534a14b444394af3135ba9 : SUCCESS in 1h 02m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fcff5d5e4b4f41299cc5b0da8d0226c4 : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"af67da9bb3f685afe46dd79856f01763cd721133","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-18 09:46:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"be82da7bdc104dd01aa0dafb517331eb9a2e54c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 12:43:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28c0a9da0ceb4874a2629c41f3b3cd0f : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c160ac3330ac4713a0b4af94dc982ef3 : SUCCESS in 11m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c697785328b4945aa7e36a405186a91 : SUCCESS in 11m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/304ab83004064123823391b484ecc7d5 : SUCCESS in 10m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7df368e1a5c943b1aa2f6cd0ea20adf4 : SUCCESS in 10m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be1a9252d4b2497b9e404135ec9df34e : SUCCESS in 11m 45s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e9c023981d09463f9921c8e4c48a98a2 : SUCCESS in 42m 58s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/608daae76e234f76aef78d287885b1ee : SUCCESS in 41m 16s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1b94871744444b2b36424982f75afbf : SUCCESS in 41m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8fd34505d6e94b0591de48cc988db953 : SUCCESS in 52m 02s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c194da871f0641e0986737ff61782bd2 : SUCCESS in 56m 51s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8be65a15688e4529b36cde4e9dcdbdab : SUCCESS in 33m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b81d9a03b4d54f25b4d8c048d3e36c8a : SUCCESS in 30m 11s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d5561aa30b2446cba27e4d102176ffa6 : SUCCESS in 2h 52m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/630ceaad8c954feb9c85275eef228482 : SUCCESS in 1h 04m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e659bdfbc58e4f40b8fe192e813141e2 : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3fbc4990cab1ef221756e9e3c2b96197d5a90c3c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-19 07:00:12.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, +1 for now, I\u0027d like someone else to review who understands the topic well.","accounts_in_message":[],"_revision_number":1},{"id":"5c4cf4d6d9e9b7f135a99fb078aab8f2db4c6dfc","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-02-19 08:21:42.000000000","message":"Patch Set 1: Code-Review-1\n\nMaybe I misunderstand something but the keystone_authtoken group already contains these parameters. Is it necessary to add these new options? The intention of the keystone_authtoken and keystone sections is a little blurry in this sense.","accounts_in_message":[],"_revision_number":1},{"id":"eb91460ba1825a8e453fa4b77ce3450b5fe872d3","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-02-19 08:53:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6446f77e0f214e7a500345606aa96237d067d0d","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-02-19 09:17:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd96088e37203058e384e661bf9fc1ff328d4525","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2020-02-19 15:23:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0264b584c09cf5d3b8abace6c913d0007a20489","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-02-20 07:02:01.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThank you.","accounts_in_message":[],"_revision_number":1},{"id":"09b83475674eee13820a1884dcc5c376d628b1a5","author":{"_account_id":31596,"name":"Ilya Zubov","email":"zubov@selectel.ru","username":"nortlite"},"date":"2020-02-25 10:07:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3f3120ebcb0400b14da4c98b429dac1d50de237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 10:54:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6d14b2bf18f4eb7b912f9d2699e3690 : FAILURE in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d60f227b142241e9bf4154b7abdeab8c : FAILURE in 10m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea01c2add69840ed9967256d8a920783 : FAILURE in 10m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06a56683ba064a2c9013a2fc12d16c42 : FAILURE in 9m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d09428c9181e49fd90994d338bcaaaa7 : FAILURE in 9m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04eaa1ab1d4a47d99aaa9790e2ac11cc : FAILURE in 7m 46s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/73c0b46d35d34a95acf4b1c2acc77224 : FAILURE in 36m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e611245fff7d49f5b2278bff33b35da6 : FAILURE in 35m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/923e2b856a604efdbcd2dc53e2b4acb7 : FAILURE in 40m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/79b5aeb11a3346dfb768017edd0c48ba : FAILURE in 37m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/de491575134f440c8cc2b864f2697eb7 : FAILURE in 41m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3b8186f9a1d0450d8022fb5b4438563d : FAILURE in 10m 15s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e82b1f87fa39468faa13f03822cf5ca4 : FAILURE in 8m 20s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0e508a87f0f54036b0349c41ac4b5850 : FAILURE in 37m 40s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ec6acf076dd846c18e0868dd12a7ac8c : FAILURE in 39m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f0084a2b6474859b3f4bb44cb84428f : FAILURE in 10m 22s","accounts_in_message":[],"_revision_number":2},{"id":"aefc95b2ea67766d14e29b303f7c2a7e8c769a82","author":{"_account_id":31596,"name":"Ilya Zubov","email":"zubov@selectel.ru","username":"nortlite"},"date":"2020-02-25 11:08:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"016c02eb68b215515a4df0c3dfdf6f42195f7b09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 11:51:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b1b71b896c04564b948585ba9342d7a : FAILURE in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49594a034b524819b8c72af95d948766 : FAILURE in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b353971ece94f0f96ef3f75a162af25 : FAILURE in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49f48bad253548cea018f4352ce19ff4 : FAILURE in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c66ce68852484f4f97404fcc8f08afd6 : FAILURE in 7m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3345fa0134ed42f2b8e25f8ec6e1c9d2 : FAILURE in 7m 34s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a7b67162474c4217b300d8b279ead2d6 : FAILURE in 30m 04s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5f22f4535354abf977761dac09267a1 : FAILURE in 29m 31s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/924b5c021d20427b9b6824958e462cac : FAILURE in 30m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/160a5e82b3664a69827f8d2fbff64656 : FAILURE in 28m 57s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9649e4d434904c229b306b1d2c4a3034 : FAILURE in 35m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2ec3021086cb4cc58be51fdba9f33a00 : FAILURE in 6m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4ada337e5e7b4559997f7b81fa58bfb4 : FAILURE in 6m 48s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4797e4b77053490cb2dd23161aacb15f : FAILURE in 33m 15s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/661b54ca83cf4bb09dabe366e0bb379b : FAILURE in 34m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/810274d03dec4bf982c70627627766d4 : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":3},{"id":"0364cc479b4fc5f6ca8ecae27f343baa221b4d0e","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-02-25 13:53:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"41689e0c755412980595e8ef3b309e80e564984b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-05 05:07:05.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5fc7ee49a8ecaf517e0226d755b220399e11ad9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 05:46:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b32cc4e978743fca87213edc605beb2 : FAILURE in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77fc08fd71c443ea971ae5466625fcb5 : FAILURE in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1afbe5ab944944469352aa431e27e2d4 : FAILURE in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0906a8fd0a624fe2b0211e0ac3850aec : FAILURE in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a867321c60e4dfcb7ac97e3d8b75460 : FAILURE in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5f76d13e2df4f1bbf23c866407d8ef5 : FAILURE in 5m 58s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/29a9c070151047d388506125a986b260 : FAILURE in 38m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa2314dffa1c4898806c39a95f7fcb05 : FAILURE in 37m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04af83cddb7e4bac978c78aaa1c75419 : FAILURE in 28m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/59994f9046904d6fa8397ec5ac2e706d : FAILURE in 36m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/985f77d5616a4401a017bf46c166ef4d : FAILURE in 27m 40s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/36f889ec19944e4480c000c103478440 : FAILURE in 5m 36s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f4a3d3f3fc3e491b8098b4ab1ae27dbd : FAILURE in 4m 10s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/dd9c0f115cc34ea8aa10c4caffd5b401 : FAILURE in 28m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61793b73f26a4440a01fdb880a3cc4c3 : SUCCESS in 13m 22s","accounts_in_message":[],"_revision_number":4},{"id":"16bd7e1d7d5cd78863944ccd4ffed59696783380","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2026-03-28 19:40:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"01efc340bcb6e011bea31e649d8c9398354b22b3","revisions":{"f91a9421687408fa1aa916ab626a21827525a433":{"kind":"REWORK","_number":1,"created":"2020-02-14 09:53:46.000000000","uploader":{"_account_id":31596,"name":"Ilya Zubov","email":"zubov@selectel.ru","username":"nortlite"},"ref":"refs/changes/98/707798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/707798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/707798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/707798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/707798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/707798/1"}}},"commit":{"parents":[{"commit":"8d75784356c948bb6fcabbb6e3e0b6b59a82bcf6","subject":"Add \"convert_output_data\" config property for YAQL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8d75784356c948bb6fcabbb6e3e0b6b59a82bcf6"}]}],"author":{"name":"Zubov Ilya","email":"zubov@selectel.ru","date":"2020-02-14 09:22:08.000000000","tz":180},"committer":{"name":"Zubov Ilya","email":"zubov@selectel.ru","date":"2020-02-14 09:22:08.000000000","tz":180},"subject":"Resolving an error with an invalid region in keystone client.","message":"Resolving an error with an invalid region in keystone client.\n\nChange-Id: I9270317c12eaa2d42536768e52051ed6cb35b5e4\nCloses-Bug: #1862940","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f91a9421687408fa1aa916ab626a21827525a433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f91a9421687408fa1aa916ab626a21827525a433"}]},"branch":"refs/heads/master"},"452f6f3b1378ee6df06d4db98d87ca363af4aeb9":{"kind":"REWORK","_number":2,"created":"2020-02-25 10:07:49.000000000","uploader":{"_account_id":31596,"name":"Ilya Zubov","email":"zubov@selectel.ru","username":"nortlite"},"ref":"refs/changes/98/707798/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/707798/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/707798/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/707798/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/707798/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/707798/2"}}},"commit":{"parents":[{"commit":"8d75784356c948bb6fcabbb6e3e0b6b59a82bcf6","subject":"Add \"convert_output_data\" config property for YAQL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8d75784356c948bb6fcabbb6e3e0b6b59a82bcf6"}]}],"author":{"name":"Zubov Ilya","email":"zubov@selectel.ru","date":"2020-02-14 09:22:08.000000000","tz":180},"committer":{"name":"Zubov Ilya","email":"zubov@selectel.ru","date":"2020-02-25 10:07:09.000000000","tz":180},"subject":"Resolving an error with an invalid region in keystone client.","message":"Resolving an error with an invalid region in keystone client.\n\nChange-Id: I9270317c12eaa2d42536768e52051ed6cb35b5e4\nCloses-Bug: #1862940\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/452f6f3b1378ee6df06d4db98d87ca363af4aeb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/452f6f3b1378ee6df06d4db98d87ca363af4aeb9"}]},"branch":"refs/heads/master"},"60511fc6ee5f874ddb90de1cc98b5ff5646e06cf":{"kind":"REWORK","_number":3,"created":"2020-02-25 11:08:09.000000000","uploader":{"_account_id":31596,"name":"Ilya Zubov","email":"zubov@selectel.ru","username":"nortlite"},"ref":"refs/changes/98/707798/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/707798/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/707798/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/707798/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/707798/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/707798/3"}}},"commit":{"parents":[{"commit":"8d75784356c948bb6fcabbb6e3e0b6b59a82bcf6","subject":"Add \"convert_output_data\" config property for YAQL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8d75784356c948bb6fcabbb6e3e0b6b59a82bcf6"}]}],"author":{"name":"Zubov Ilya","email":"zubov@selectel.ru","date":"2020-02-14 09:22:08.000000000","tz":180},"committer":{"name":"Zubov Ilya","email":"zubov@selectel.ru","date":"2020-02-25 11:07:27.000000000","tz":180},"subject":"Resolving an error with an invalid region in keystone client.","message":"Resolving an error with an invalid region in keystone client.\n\nChange-Id: I9270317c12eaa2d42536768e52051ed6cb35b5e4\nCloses-Bug: #1862940\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/60511fc6ee5f874ddb90de1cc98b5ff5646e06cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/60511fc6ee5f874ddb90de1cc98b5ff5646e06cf"}]},"branch":"refs/heads/master"},"01efc340bcb6e011bea31e649d8c9398354b22b3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-03-05 05:07:05.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/98/707798/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/98/707798/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/707798/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/707798/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/98/707798/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/98/707798/4"}}},"commit":{"parents":[{"commit":"3ca452229ce12a9ae0fa2cb3d3bace507957de2e","subject":"Merge \"Add an article about profiling into the docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3ca452229ce12a9ae0fa2cb3d3bace507957de2e"}]}],"author":{"name":"Zubov Ilya","email":"zubov@selectel.ru","date":"2020-02-14 09:22:08.000000000","tz":180},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-05 05:07:05.000000000","tz":0},"subject":"Resolving an error with an invalid region in keystone client.","message":"Resolving an error with an invalid region in keystone client.\n\nChange-Id: I9270317c12eaa2d42536768e52051ed6cb35b5e4\nCloses-Bug: #1862940\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/01efc340bcb6e011bea31e649d8c9398354b22b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/01efc340bcb6e011bea31e649d8c9398354b22b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
