)]}'
{"id":"openstack%2Fmistral~708127","triplet_id":"openstack%2Fmistral~master~Ib4a13acbc65612e3dd23a8a13baad5856777de99","project":"openstack/mistral","branch":"master","topic":"cli","hashtags":[],"change_id":"Ib4a13acbc65612e3dd23a8a13baad5856777de99","subject":"Move registration of CLI options to launch.py","status":"MERGED","created":"2020-02-17 13:30:58.000000000","updated":"2020-02-19 09:30:28.000000000","submitted":"2020-02-18 09:40:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708127-1582018810687-86a07e3f","meta_rev_id":"b3c1f191da24d3007a5d25625e1bbc19342d5a0e","_number":708127,"virtual_id_number":708127,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 09:40:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-02-18 05:04:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-02-18 05:04:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 05:04:30.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-02-18 09:40:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f68165988ad8dd5598c5b2719bdc1e517144a5e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-17 13:30:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9830b4813810a894a55a601e84e46aefe36f48ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 16:38:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e64294a3ba484848bd49190e49b5ecb6 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7518946056bd43ee9626239e4ef3ed1e : SUCCESS in 10m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6598c4a7e8444f72aa5cfffb06efd1b0 : SUCCESS in 12m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ea886b414784f238e75659bd84252da : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db373dd1ded428388404270b70870e1 : SUCCESS in 9m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d2f4f7bb7b2499c9523acec34791626 : SUCCESS in 12m 36s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d0cb715c22494c61bf0aff6c3a877d7d : SUCCESS in 43m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85acae74c7724c5e91eef6ce4e0e1115 : SUCCESS in 44m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c56354e31ce1465096898cb3a48f3426 : SUCCESS in 48m 46s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/35173a6d49324d5ea8b308bf5910464a : SUCCESS in 46m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/e23e0c9370d0466f802ae5fb182894f3 : SUCCESS in 46m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/25cce2fff5164cf0b945c4ae0fd527fc : SUCCESS in 30m 27s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/afba2a71a11c494987a60d80f41ce418 : SUCCESS in 28m 27s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0c4ac00480df4b88be8fd34df914691a : SUCCESS in 2h 46m 18s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0c57f57eeedc45e79882d0f938c09253 : SUCCESS in 1h 08m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77690d583d4e45aba490055d325bfd70 : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":1},{"id":"699bed690f87b75f47799f17028c85b658235613","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-18 04:24:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2b40e30c6ef1a45ea6489d8fc46a97c86e68fc24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 04:24:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"51099997a4fc531382faa8d875c9f6405d34388f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-18 05:04:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d61595db4c59caf44cf17349a37a99a23f8b480e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-18 05:04:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f9dfef08805075384fdb0804185df8bc5238cb5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 08:04:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e86130a3e8a4411c99ca300ad33314c9 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afdec928b674488095acf70d2a24246e : SUCCESS in 9m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20b4064d0d6b4ada8923971bbd0d773c : SUCCESS in 12m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed59a7031aec4ce2825a1f7ab0cb8526 : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81702f27916843fb9e22532017f52db9 : SUCCESS in 8m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a63471e9f07145af9be03cfb2a0743dd : SUCCESS in 10m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b687e815b39647028f3c956bf9130fb3 : SUCCESS in 45m 19s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f585e7717b2241a38f336b5d20cb9976 : SUCCESS in 42m 49s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d1ea557e22543c884611fd3bcf00ed0 : SUCCESS in 45m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4b25c99138fe4c5fb1ae89d08f1f0041 : SUCCESS in 39m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/daf0a2afbbbb47b4b980ef243726995d : SUCCESS in 42m 04s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3ae5a62085a345cd966256fe862795ce : SUCCESS in 30m 28s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c6d563de4af34da4b886b2004b158cca : SUCCESS in 29m 54s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/9792dc8eeed54d7e813064b6ae3babd4 : SUCCESS in 2h 49m 31s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0fb2ff2de8e246789b749ea83ddbeb22 : SUCCESS in 58m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64743aff82834d159c6a16cfb9b2a0f5 : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":2},{"id":"1bffedaccad371d68ed4cded0313dce2a96bdd24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 08:05:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"13da4734b10b7541616bf980b8138b8429563436","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 09:40:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a28f33358c534b0eb271df7025d35ddb : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61b69e0d48c34a55bd95f71a1c9fbc07 : SUCCESS in 9m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11af39d96ea74dc49b22f3fdd3ea78ce : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3627ed26fab42808c95e618bbf603b1 : SUCCESS in 9m 13s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/639ad682a5124fb19f27b2bbae428729 : SUCCESS in 51m 25s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17daa5b727ce4de09e09f06e10c6121d : SUCCESS in 45m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e89b1c0fc13e478ab6f76dc37fb4d7a8 : SUCCESS in 42m 02s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ba2eefd5268d4c3d8d149d619d6e5ba1 : SUCCESS in 47m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b61df2581c4b4cc7ae28dc097f13e5d2 : SUCCESS in 32m 41s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6b3a18fdbf41416e8c13811b6e8a96fc : SUCCESS in 30m 27s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/09c5017fb5b44c419958af85cd9d50a3 : SUCCESS in 50m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a117832f37849b893fd286a2bc3e9bf : SUCCESS in 12m 17s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6e6ca8371a11482fb93023ad6f35b886 : SUCCESS in 1h 30m 42s","accounts_in_message":[],"_revision_number":2},{"id":"34c3f8c508666fafdf2b74b226afe2d31ec47429","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 09:40:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"44e5088a6d39a86d8d5a6b41ae320ae53dbe3c01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 09:42:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7de53165022414d9ccf92b28b665fe2 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":2},{"id":"ae36925c4e9cf273544004b94953cf6fcdc117ed","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-19 09:30:28.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 3f61f80c610fc0efa6e1ebc140de7b1a5ad0c4dd","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f23d3c07b2d6b12897a2bfd2034c0693d3824cc8","revisions":{"c3b7cd350bf562e8c9f150422d0164e89c7f4fba":{"kind":"REWORK","_number":1,"created":"2020-02-17 13:30:58.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/27/708127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/27/708127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/708127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/708127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/708127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/27/708127/1"}}},"commit":{"parents":[{"commit":"8d75784356c948bb6fcabbb6e3e0b6b59a82bcf6","subject":"Add \"convert_output_data\" config property for YAQL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8d75784356c948bb6fcabbb6e3e0b6b59a82bcf6"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-17 13:30:46.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-17 13:30:46.000000000","tz":120},"subject":"Move register cli to launch.py","message":"Move register cli to launch.py\n\nThis the only place where it is used, should fix the gate\nfailure of the docs\n\nChange-Id: Ib4a13acbc65612e3dd23a8a13baad5856777de99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c3b7cd350bf562e8c9f150422d0164e89c7f4fba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c3b7cd350bf562e8c9f150422d0164e89c7f4fba"}]},"branch":"refs/heads/master"},"f23d3c07b2d6b12897a2bfd2034c0693d3824cc8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-18 05:04:24.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/27/708127/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/27/708127/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/708127/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/708127/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/27/708127/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/27/708127/2"}}},"commit":{"parents":[{"commit":"8d75784356c948bb6fcabbb6e3e0b6b59a82bcf6","subject":"Add \"convert_output_data\" config property for YAQL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8d75784356c948bb6fcabbb6e3e0b6b59a82bcf6"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-17 13:30:46.000000000","tz":120},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-18 05:04:20.000000000","tz":0},"subject":"Move registration of CLI options to launch.py","message":"Move registration of CLI options to launch.py\n\nThis the only place where it is used, should fix the gate\nfailure of the docs.\n\nChange-Id: Ib4a13acbc65612e3dd23a8a13baad5856777de99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f23d3c07b2d6b12897a2bfd2034c0693d3824cc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f23d3c07b2d6b12897a2bfd2034c0693d3824cc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
