)]}'
{"id":"openstack%2Fmistral~708560","triplet_id":"openstack%2Fmistral~master~I79befa91fcaf393ec19498372f1e6c6953151867","project":"openstack/mistral","branch":"master","topic":"add_profiling_doc","hashtags":[],"change_id":"I79befa91fcaf393ec19498372f1e6c6953151867","subject":"Add an article about profiling into the docs","status":"MERGED","created":"2020-02-19 05:11:05.000000000","updated":"2020-03-04 12:05:49.000000000","submitted":"2020-03-04 12:02:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708560-1583323342328-a42b5d9c","meta_rev_id":"e4818f1cadc41503d884157c1a1b307d2b31f8b9","_number":708560,"virtual_id_number":708560,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 12:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-03-04 11:10:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-04 10:32:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2020-03-04 11:08:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 08:57:53.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-03-04 10:32:36.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-03-04 11:08:16.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-03-04 12:02:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62b4023c65d8b6764ac2b4632bf0e076aca5332d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-19 05:11:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfa3db3832328b42eed1e4a9d37f9a383e94f76a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-19 05:11:22.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"062d47fa4c4b4b632b84990ffd0f5e4c0fa36dc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 05:20:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54d4940e8bdf483bb06875878279bc36 : FAILURE in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26def215264841ddbd80325dd20d0534 : SUCCESS in 9m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0f912685715dab3cc9d328f693666d8001257b0d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-19 09:17:37.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"debd8b3e90772247b3bcc0345c24e2baceb93eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 09:42:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c1fe9e05add4cfba62d593d0506a5a5 : FAILURE in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eac8b129b9c4d5e84b0e7fb9fadc161 : SUCCESS in 9m 12s","accounts_in_message":[],"_revision_number":2},{"id":"e147e6cc9b0bdcca00cf4e4a165d65ab58d25d70","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 07:14:04.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"dc5949f5a4122eeb2c59e3a998ff73c52b923499","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 07:16:05.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"21b9edd8cfd9adde6d46297da1340ee4b6c520db","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 07:17:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"aea3be6801764886fa0a41ca5748ea6bd5dea3b9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 07:18:06.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"9e6609287f564aa0d839e0cb50d5570e5c0a6631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 07:29:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1dc80d4729640a9a19a4f6d5e4ed8f8 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489721dc1d0b46deaacff3c33a0a9546 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":4},{"id":"7fcbe8050bb1445cf544cb1a71dee63b0ec7f1cf","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-02-25 08:57:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"20489cbc9d9c0b4756cc4c4e4d1516b9d21f58cc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-02 07:41:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a06712451a4865fdf56be18885a69dd652bac3c1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-02 09:01:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b353a6db5af76096768024aae7b09bd6b33d3e43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 09:14:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91c0c6ac5478429fadb4183d4cfb59e8 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/802e78a14bdb425fbb79733738e760ad : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":5},{"id":"8053e8893cb81d93e457c9a3a539e4dd05f39093","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-02 09:52:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b48243b62a72e30dfb3831ed854d382dd39c2ddb","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-02 09:58:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"40ba18891d3da5179e2e8a9a76d59c2bf1349f0f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-02 10:13:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a2c11a739c28fa6e84a8a4ebecf51d168855fab3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 10:33:04.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c976502594d0428f9ca9bcfa107e5505 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f599ceecc06e4a64ba53da199c4d4634 : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":8},{"id":"a99e9e804d1503c6742d92d01443f1830e39d2ad","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-03-02 10:38:21.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"f2516cc32f87c535057ad5c360d72b8a7e94dd94","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-03 11:25:44.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"8987f9ec678e46e7ab61cbb5fe23517b4eacee8e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-04 06:44:11.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b659e3635eb2e3a83fe258c8f89e7966015f4e4b","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-04 10:30:22.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2286d18d5d2adf22defa98d80647454caab2117e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-04 10:32:36.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"951420d3a040ffd5d16f2d5a5fcb9a4dceb3253e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 10:47:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b7fc1a8117748859001acf9d36a39fb : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/476726a321324e898549fc179139c520 : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":9},{"id":"01b7f7094e0b5f30d0545daf1a5b748f1ae9cc03","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-03-04 11:08:16.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"8f5f990ad5882521a31ffb5f6b8bd50f56695d48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 11:08:29.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"9f36c9deac8e60156df82badc290c347707e28ac","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-03-04 11:10:20.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"9a07218a7e9d6ba209348356de06bc96c4dd7676","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:02:22.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3be284de5f6747fd8ed4c6b50513a4e2 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf4107299a1447591071a333a971b6b : SUCCESS in 9m 08s","accounts_in_message":[],"_revision_number":9},{"id":"05999e44a36319b4c0c3eff36689469f4d44d235","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:02:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"ff8a72fc26353c5649ba8b6d0aa812cd2ab7c312","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:05:49.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b24701279067458b8622ce74a73f420e : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"cc825ebcd32876b0205fc91f399df026d8fbe582","revisions":{"d5f1f802da392ec5112a5b93f6990cebde6e208e":{"kind":"REWORK","_number":1,"created":"2020-02-19 05:11:05.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/60/708560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/708560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/708560/1"}}},"commit":{"parents":[{"commit":"0a3880c22c819b41e8cfc6ea593e3f69112ec7a5","subject":"Make tripleO CI job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0a3880c22c819b41e8cfc6ea593e3f69112ec7a5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-18 09:12:06.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-19 05:09:39.000000000","tz":420},"subject":"Add an article about profiling into the docs","message":"Add an article about profiling into the docs\n\nChange-Id: I79befa91fcaf393ec19498372f1e6c6953151867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5f1f802da392ec5112a5b93f6990cebde6e208e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d5f1f802da392ec5112a5b93f6990cebde6e208e"}]},"branch":"refs/heads/master"},"0e4430205dcfe981b1ce3faa26be83d15b34dd79":{"kind":"NO_CHANGE","_number":2,"created":"2020-02-19 09:17:37.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/60/708560/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/708560/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/708560/2"}}},"commit":{"parents":[{"commit":"0a3880c22c819b41e8cfc6ea593e3f69112ec7a5","subject":"Make tripleO CI job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0a3880c22c819b41e8cfc6ea593e3f69112ec7a5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-18 09:12:06.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-19 09:13:39.000000000","tz":420},"subject":"Add an article about profiling into the docs","message":"Add an article about profiling into the docs\n\nChange-Id: I79befa91fcaf393ec19498372f1e6c6953151867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e4430205dcfe981b1ce3faa26be83d15b34dd79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0e4430205dcfe981b1ce3faa26be83d15b34dd79"}]},"branch":"refs/heads/master"},"08a91b186fe88cb7ffa971a4df599441245a4843":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-21 07:14:04.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/60/708560/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/708560/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/708560/3"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-18 09:12:06.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-20 10:19:31.000000000","tz":420},"subject":"Add an article about profiling into the docs","message":"Add an article about profiling into the docs\n\nChange-Id: I79befa91fcaf393ec19498372f1e6c6953151867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/08a91b186fe88cb7ffa971a4df599441245a4843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/08a91b186fe88cb7ffa971a4df599441245a4843"}]},"branch":"refs/heads/master"},"a15db2293e3c3376da2b9ec064b88f495cf98f7c":{"kind":"REWORK","_number":4,"created":"2020-02-21 07:17:50.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/60/708560/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/708560/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/708560/4"}}},"commit":{"parents":[{"commit":"0a3880c22c819b41e8cfc6ea593e3f69112ec7a5","subject":"Make tripleO CI job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0a3880c22c819b41e8cfc6ea593e3f69112ec7a5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-18 09:12:06.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-20 10:15:25.000000000","tz":420},"subject":"Add an article about profiling into the docs","message":"Add an article about profiling into the docs\n\nChange-Id: I79befa91fcaf393ec19498372f1e6c6953151867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a15db2293e3c3376da2b9ec064b88f495cf98f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a15db2293e3c3376da2b9ec064b88f495cf98f7c"}]},"branch":"refs/heads/master"},"ea8e2b9c08c4cf1208d98240feeb83d80e0ded20":{"kind":"REWORK","_number":5,"created":"2020-03-02 09:01:17.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/60/708560/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/708560/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/708560/5"}}},"commit":{"parents":[{"commit":"0a3880c22c819b41e8cfc6ea593e3f69112ec7a5","subject":"Make tripleO CI job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0a3880c22c819b41e8cfc6ea593e3f69112ec7a5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-18 09:12:06.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-02 09:00:59.000000000","tz":420},"subject":"Add an article about profiling into the docs","message":"Add an article about profiling into the docs\n\nChange-Id: I79befa91fcaf393ec19498372f1e6c6953151867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ea8e2b9c08c4cf1208d98240feeb83d80e0ded20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ea8e2b9c08c4cf1208d98240feeb83d80e0ded20"}]},"branch":"refs/heads/master"},"51c9501502d4a86d31506bff6693cefe3d9f8b68":{"kind":"REWORK","_number":6,"created":"2020-03-02 09:52:04.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/60/708560/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/708560/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/708560/6"}}},"commit":{"parents":[{"commit":"0a3880c22c819b41e8cfc6ea593e3f69112ec7a5","subject":"Make tripleO CI job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0a3880c22c819b41e8cfc6ea593e3f69112ec7a5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-18 09:12:06.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-02 09:51:46.000000000","tz":420},"subject":"Add an article about profiling into the docs","message":"Add an article about profiling into the docs\n\nChange-Id: I79befa91fcaf393ec19498372f1e6c6953151867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/51c9501502d4a86d31506bff6693cefe3d9f8b68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/51c9501502d4a86d31506bff6693cefe3d9f8b68"}]},"branch":"refs/heads/master"},"b9d9f4540544cbc9c7ccda9f4f375d8c8966e308":{"kind":"REWORK","_number":7,"created":"2020-03-02 09:58:24.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/60/708560/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/708560/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/708560/7"}}},"commit":{"parents":[{"commit":"0a3880c22c819b41e8cfc6ea593e3f69112ec7a5","subject":"Make tripleO CI job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0a3880c22c819b41e8cfc6ea593e3f69112ec7a5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-18 09:12:06.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-02 09:58:09.000000000","tz":420},"subject":"Add an article about profiling into the docs","message":"Add an article about profiling into the docs\n\nChange-Id: I79befa91fcaf393ec19498372f1e6c6953151867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b9d9f4540544cbc9c7ccda9f4f375d8c8966e308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b9d9f4540544cbc9c7ccda9f4f375d8c8966e308"}]},"branch":"refs/heads/master"},"3f7ea93a3453d527281a1bdafc36f0b913fc9ed2":{"kind":"REWORK","_number":8,"created":"2020-03-02 10:13:40.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/60/708560/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/708560/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/708560/8"}}},"commit":{"parents":[{"commit":"0a3880c22c819b41e8cfc6ea593e3f69112ec7a5","subject":"Make tripleO CI job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0a3880c22c819b41e8cfc6ea593e3f69112ec7a5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-18 09:12:06.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-02 10:13:23.000000000","tz":420},"subject":"Add an article about profiling into the docs","message":"Add an article about profiling into the docs\n\nChange-Id: I79befa91fcaf393ec19498372f1e6c6953151867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3f7ea93a3453d527281a1bdafc36f0b913fc9ed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3f7ea93a3453d527281a1bdafc36f0b913fc9ed2"}]},"branch":"refs/heads/master"},"cc825ebcd32876b0205fc91f399df026d8fbe582":{"kind":"REWORK","_number":9,"created":"2020-03-04 10:30:22.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/60/708560/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/60/708560/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/60/708560/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/60/708560/9"}}},"commit":{"parents":[{"commit":"0a3880c22c819b41e8cfc6ea593e3f69112ec7a5","subject":"Make tripleO CI job non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0a3880c22c819b41e8cfc6ea593e3f69112ec7a5"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-18 09:12:06.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-04 10:30:05.000000000","tz":420},"subject":"Add an article about profiling into the docs","message":"Add an article about profiling into the docs\n\nChange-Id: I79befa91fcaf393ec19498372f1e6c6953151867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cc825ebcd32876b0205fc91f399df026d8fbe582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cc825ebcd32876b0205fc91f399df026d8fbe582"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
