)]}'
{"id":"openstack%2Fmistral~708582","triplet_id":"openstack%2Fmistral~stable%2Ftrain~Ie3169ec884ec9a0e7e50327dd03cd78dcda0a39b","project":"openstack/mistral","branch":"stable/train","topic":"bug/1863857","hashtags":[],"change_id":"Ie3169ec884ec9a0e7e50327dd03cd78dcda0a39b","subject":"Add \"convert_output_data\" config property for YAQL","status":"MERGED","created":"2020-02-19 08:59:26.000000000","updated":"2020-02-19 21:32:52.000000000","submitted":"2020-02-19 21:30:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708582-1582147840050-d0314534","meta_rev_id":"708dbe4f7274bc201cc7c2272bad95717aa33273","_number":708582,"virtual_id_number":708582,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 21:30:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-02-19 16:30:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-19 16:29:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-02-19 16:30:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-19 09:21:15.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-02-19 21:30:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8475c46401491fa6dbd3f1bde674aade70e0d409","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-19 08:59:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"199b312344cbccaea8856657eb500bca9dd5550b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-19 08:59:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"49469bbc6ba670c181025606ed76efa5288166c4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-19 09:20:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0a5501a9416551536751343a920ae63140b77cc","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-19 09:21:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"900c0fe6ab67bc76204d2f98078f8295a0610f4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 12:25:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab40882ef8c94240aed16fdbc063772f : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac51f6cb33fe435eb46374c9227583ce : SUCCESS in 11m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/373cf64d212d4ab19e137e71d3b063d6 : SUCCESS in 10m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4be849dcb9f74d54af27e6d56657a419 : SUCCESS in 10m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82ffc7e9ace0459da3d8a7d7341e5876 : FAILURE in 6m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bdda917363f4290b210b36a1d33f9a3 : SUCCESS in 11m 25s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fa1812ccb46e447099b48244aeb496e0 : SUCCESS in 43m 00s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3d2661a0dab449b292e721ea36d2cf53 : SUCCESS in 40m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/3522d9944da64797a77e193b70f152f6 : SUCCESS in 48m 42s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/000f88bc4fad47f38744abae39cc6096 : SUCCESS in 29m 23s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0247c5572136417c99f1d34d2ba4ef6c : SUCCESS in 30m 02s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/cfa9cf55324c436fb59bdd4ed2b52832 : SUCCESS in 1h 18m 19s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0db4bb55e7fd43609885065c26470466 : SUCCESS in 2h 58m 54s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/13d57a50ec044762bf6cafcf0349a343 : SUCCESS in 1h 09m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/668ba3bfb033436ca1fb61bedff5817c : SUCCESS in 14m 20s","accounts_in_message":[],"_revision_number":2},{"id":"2a734f42237e1b80c5c5e0129230b65a7914b6a8","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-19 14:12:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f903850f81543c2a7e9a86b35ab87642993e2da1","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-19 16:29:44.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a0d330b81d59163bf93d75349f3aab77274d4cfe","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-19 16:30:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"284db87e4f44b77848d1d4478120f8afe737088b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 19:31:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d45ab4af4d574e01a922ac165a036bd0 : SUCCESS in 8m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30b8d92a01c148d698aeb16b837215ec : SUCCESS in 11m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02cf5e988f744e589108f079bd7ee605 : SUCCESS in 13m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1652ad307364e459fc94b3d7b18eb53 : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbfb50f7872e4d8aa3da8f6b23cb0e05 : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84ea4f1f22dc49fbadb3e782c9516824 : SUCCESS in 11m 29s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/847cdcaead434be5a407e0be4eb536bf : SUCCESS in 56m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8533b3f2a42e4b1c92280b6285c0630b : SUCCESS in 54m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c1caf782f72c4fbf882a4591c4d8ec61 : SUCCESS in 51m 06s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2311676a31814c1e8d28777fc8677a9e : SUCCESS in 30m 47s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ca1d3c8295d747ec930f75aaf96e0de3 : SUCCESS in 31m 13s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/edf418c22e2247ed9e2d08b5a70be88f : SUCCESS in 1h 14m 16s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/93e29496368742018399464c363417b9 : SUCCESS in 2h 53m 28s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b4b003d916f54dfba4cf3258ba71ba79 : SUCCESS in 1h 05m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92c84f30ca6e4dc99e0f3b3b9889b1ee : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":3},{"id":"6e44dd9ab57769d4a0dd967e619b7facac2e3b72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 19:31:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"945889ad663a3969da4a594e64eea8bad38521ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 21:30:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a563165b464c18a03f873352b378e2 : SUCCESS in 7m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2c902250ffe436cae1bd616b63448f5 : SUCCESS in 12m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dffe1a52f494ce7b6772fc13033b1c2 : SUCCESS in 10m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95ea6b6e30864a6fb6f731f587858f7f : SUCCESS in 10m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90edae828f354f318139450445a2010d : SUCCESS in 8m 09s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8cefd12347124c289c238f80fefd5c8a : SUCCESS in 56m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/4c50fdebbfa145c78a3ceec7f51c92df : SUCCESS in 51m 57s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/383e563ed90943e599efc230c6b45228 : SUCCESS in 31m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/95b9a46f0a2c4ec3b2d6543ae99f4cae : SUCCESS in 29m 38s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e4f54d99893e48039a5ecb4064afdd96 : SUCCESS in 1h 38m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b0d5fea801a3434e88fe41086e1b8582 : SUCCESS in 48m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2d93340e1cb4fc8b02ecaa1008ccf28 : SUCCESS in 13m 06s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/18bc92d47fe4473fa6de11c37fcbd9d9 : SUCCESS in 1h 52m 53s","accounts_in_message":[],"_revision_number":3},{"id":"06f977b6d2a7add455fe6d41dee33e4317546510","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 21:30:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"65ab5b0a65cf1b64d1b8af26b4804341b70840a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 21:32:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e07930f1b4374bab83acf520cd92ef99 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d22676ef396e0fccc6732b14f8be12f4625e3c79","revisions":{"f306d80943436ae24e00c5cecc92b99df9ef096b":{"kind":"REWORK","_number":1,"created":"2020-02-19 08:59:26.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/82/708582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/82/708582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/82/708582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/82/708582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/82/708582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/82/708582/1"}}},"commit":{"parents":[{"commit":"7c54c28c4c38ad1a71bf4388ffb1fe70200b373e","subject":"Set the delayed call \"key\" field to the right value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7c54c28c4c38ad1a71bf4388ffb1fe70200b373e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-11 04:18:50.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-19 08:58:14.000000000","tz":420},"subject":"Add \"convert_output_data\" config property for YAQL","message":"Add \"convert_output_data\" config property for YAQL\n\n* Adding \"convert_output_data\" config property gives an opportunity\n  to increase overal performance. If YAQL always converts an expression\n  result, it often takes significant CPU time and overall workflow\n  execution time increases. It is especially important when a workflow\n  publishes lots of data into the context and uses big workflow\n  environments. It\u0027s been tested on a very big workflow (~7k tasks)\n  with a big workflow environment (~2.5mb) that often uses the YAQL\n  function \"\u003c% env() %\u003e\". This function basically just returns the\n  workflow environment.\n* Created all necessary unit tests.\n* Other style fixes.\n\nCloses-Bug: #1863857\nChange-Id: Ie3169ec884ec9a0e7e50327dd03cd78dcda0a39b\nCherry picked from Ie3169ec884ec9a0e7e50327dd03cd78dcda0a39b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f306d80943436ae24e00c5cecc92b99df9ef096b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f306d80943436ae24e00c5cecc92b99df9ef096b"}]},"branch":"refs/heads/stable/train"},"cf9352b5138f37dd2a5a6698a756d846c8518e25":{"kind":"REWORK","_number":2,"created":"2020-02-19 09:20:29.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/82/708582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/82/708582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/82/708582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/82/708582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/82/708582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/82/708582/2"}}},"commit":{"parents":[{"commit":"7c54c28c4c38ad1a71bf4388ffb1fe70200b373e","subject":"Set the delayed call \"key\" field to the right value","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/7c54c28c4c38ad1a71bf4388ffb1fe70200b373e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-11 04:18:50.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-19 09:20:11.000000000","tz":420},"subject":"Add \"convert_output_data\" config property for YAQL","message":"Add \"convert_output_data\" config property for YAQL\n\n* Adding \"convert_output_data\" config property gives an opportunity\n  to increase overal performance. If YAQL always converts an expression\n  result, it often takes significant CPU time and overall workflow\n  execution time increases. It is especially important when a workflow\n  publishes lots of data into the context and uses big workflow\n  environments. It\u0027s been tested on a very big workflow (~7k tasks)\n  with a big workflow environment (~2.5mb) that often uses the YAQL\n  function \"\u003c% env() %\u003e\". This function basically just returns the\n  workflow environment.\n* Created all necessary unit tests.\n* Other style fixes.\n\nCloses-Bug: #1863857\nChange-Id: Ie3169ec884ec9a0e7e50327dd03cd78dcda0a39b\nCherry picked from Ie3169ec884ec9a0e7e50327dd03cd78dcda0a39b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cf9352b5138f37dd2a5a6698a756d846c8518e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/cf9352b5138f37dd2a5a6698a756d846c8518e25"}]},"branch":"refs/heads/stable/train"},"d22676ef396e0fccc6732b14f8be12f4625e3c79":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-19 16:29:44.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/82/708582/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/82/708582/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/82/708582/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/82/708582/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/82/708582/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/82/708582/3"}}},"commit":{"parents":[{"commit":"3f61f80c610fc0efa6e1ebc140de7b1a5ad0c4dd","subject":"Move registration of CLI options to launch.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3f61f80c610fc0efa6e1ebc140de7b1a5ad0c4dd"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-11 04:18:50.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-19 16:29:44.000000000","tz":0},"subject":"Add \"convert_output_data\" config property for YAQL","message":"Add \"convert_output_data\" config property for YAQL\n\n* Adding \"convert_output_data\" config property gives an opportunity\n  to increase overal performance. If YAQL always converts an expression\n  result, it often takes significant CPU time and overall workflow\n  execution time increases. It is especially important when a workflow\n  publishes lots of data into the context and uses big workflow\n  environments. It\u0027s been tested on a very big workflow (~7k tasks)\n  with a big workflow environment (~2.5mb) that often uses the YAQL\n  function \"\u003c% env() %\u003e\". This function basically just returns the\n  workflow environment.\n* Created all necessary unit tests.\n* Other style fixes.\n\nCloses-Bug: #1863857\nChange-Id: Ie3169ec884ec9a0e7e50327dd03cd78dcda0a39b\nCherry picked from Ie3169ec884ec9a0e7e50327dd03cd78dcda0a39b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d22676ef396e0fccc6732b14f8be12f4625e3c79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d22676ef396e0fccc6732b14f8be12f4625e3c79"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
