)]}'
{"id":"openstack%2Fmistral~709057","triplet_id":"openstack%2Fmistral~master~I94f57a6534b9c10e202205dfae4d039296c26407","project":"openstack/mistral","branch":"master","topic":"fix_yaql_engine_options_initialization","hashtags":[],"change_id":"I94f57a6534b9c10e202205dfae4d039296c26407","subject":"Refactor expressions","status":"MERGED","created":"2020-02-21 07:14:04.000000000","updated":"2020-02-26 10:52:00.000000000","submitted":"2020-02-26 10:50:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709057-1582714212813-8d25ec41","meta_rev_id":"85f6cf953911d2411bcdd17fe73de138515abc62","_number":709057,"virtual_id_number":709057,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 10:50:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-02-26 08:44:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-26 08:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":1,"date":"2020-02-26 08:44:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-02-26 05:45:20.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2020-02-26 05:45:20.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 05:45:20.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-02-26 05:45:20.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-02-26 08:23:37.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-02-26 08:44:57.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-02-26 10:50:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"736f0317a1560dd90f42f01cae981374f949f593","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 07:14:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1c55d06d95fbf5cf7f9f0101d29c80b2b97faf2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 07:20:22.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"fcbd058aee3fa26e6439b2c903444225fd812575","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 07:20:41.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"229dab49f1b04457ad7afc5433529fb197f3335f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 07:54:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"61894501f1762f2746b2766ca3958f4f2ff59565","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 08:10:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"835f65041f65890a6ed36bd02e31e0adbdd020f9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 08:11:09.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"d35049de270fa655b1aae86b030bf16735670921","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-02-21 08:16:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e48f5b0e81d1123bf4974d306e57cad457b06401","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-21 08:21:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"393e33adbb23f1a9990d0a044fa9364ed5073ddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 11:04:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/39b654c68e194eebb65fb8d5c1263ce7 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe28300c846d49fbbc52f3c711a1b818 : SUCCESS in 11m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6a160c986ad4b7b99566942321d9a32 : SUCCESS in 10m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72bb16513a19446ca1967a15be363108 : SUCCESS in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/317ae578abdb4d25b5c669fdaea91a7f : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b01a0fa287441fa89899363b7f17e8 : SUCCESS in 9m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd5d72f37a884217b5f0a8fe973720e1 : SUCCESS in 10m 53s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9c6e6caecb8243b8a5f34aceed4d96c2 : SUCCESS in 45m 51s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/842ee175e0ed408fa5be451fcecdbeb2 : SUCCESS in 44m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4eeeaf377084ed7b90a4ecee8f18059 : SUCCESS in 42m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1917792088c7460c977fdb1b3c9b3ecf : SUCCESS in 46m 40s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/efd792206e1b4757826446881f1c9cd9 : SUCCESS in 42m 00s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/522d60a02e9d4bd788dfd32ff8583eaa : FAILURE in 29m 33s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/f7cce285ecd549e2a65799b5b0f7b8d3 : SUCCESS in 29m 07s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ea65387b1ff2430f9169f77732980f91 : SUCCESS in 2h 48m 59s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d12cbaeb00a1467084ce90f48dd5879b : SUCCESS in 56m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76aa00a95ed84bce812b382c0d11dbdc : SUCCESS in 11m 18s","accounts_in_message":[],"_revision_number":4},{"id":"95127e856d19277ee6897054db5a76d9b744d352","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-25 11:26:52.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"a85efeab2153f7e4835ee9ed34ad7b09ef9d1e9a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-25 11:27:13.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9b8d654736334d4810510fb97535bfad69cd199c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:01:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2ba6acc8f1bd4ec3a92ca52a51546d26 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e8a77e47d0416aa04028b5342b0e45 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4c721a8f9b548e286fd0ef013b82afe : SUCCESS in 9m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c334c44f76b448585bf9959c41da8f2 : SUCCESS in 10m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afe5eb11ea7b4e168e0827a79d1eeb37 : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1426707ed3b84029b44a59cd4788e356 : SUCCESS in 8m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1aa35d5370d41619ee86a3d8d638731 : SUCCESS in 13m 15s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/70c00fb1268441a3bec0a843a89ca054 : SUCCESS in 50m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9a47d66df0849539da89161ed7df58e : SUCCESS in 55m 46s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd5e94426a3c46f4a414c381b1c43898 : SUCCESS in 55m 33s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b163b1ed305c4614b02f6f3868013910 : SUCCESS in 44m 23s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1abfecb2ee7d49bb8f89975fa13ab39f : SUCCESS in 48m 13s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/ce0b69b8e9ea4a9fadea3f65101e8cef : SUCCESS in 33m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/73b9d3515f8b44b3a60e6b40cf76486f : SUCCESS in 30m 01s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/dba4c3a3e99147bbb4fe83f1de5fe4d5 : SUCCESS in 2h 32m 02s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/5b54c100a4d94aa385bc604c38696b8a : SUCCESS in 58m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3feacf6b36104e3981a1c715e982212b : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":4},{"id":"2a0b076acc31cdd00fd411fe8293aff695660e80","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-26 05:37:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2db42e8f7def5a49d8705ce5a9f65aa9d6f6b3d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 08:19:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f6f64477c77f4177b1f3820d0f768735 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ca0ffc471684ca99b2f8c39e929e16c : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76acc9d4113f4e04b560743f9a7c6d43 : SUCCESS in 9m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5bb28b2a03ed4cf78bebc804270dd516 : SUCCESS in 10m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52e37ae8bc60494b84f27c36ee5406de : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaaa1c8238944c9dbe4dcfd9831138d4 : SUCCESS in 8m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbf7616ef2d24a628c87edd92de140b2 : SUCCESS in 9m 32s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3393be6394ae4a1ab0e4fbfd7bc295cd : SUCCESS in 48m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf3b4c5a0a0d4c8e922232f2a78083b2 : SUCCESS in 44m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee6b600fb1324b3cbb0e589c82b4a33f : SUCCESS in 45m 51s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3afaadf8b6ed4ab7922d33215e2c889b : SUCCESS in 44m 54s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/7daf684edee14198aec3e905e92b7922 : SUCCESS in 47m 08s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b9cd2180388947d7b3ed9158dd198ffa : SUCCESS in 33m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b8feadb1dcae4586be3869c6a52df2b7 : SUCCESS in 32m 14s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/46e02d18c15840bb8c6790881e121516 : SUCCESS in 2h 40m 26s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/7858983fc7384278b50199df6748a791 : SUCCESS in 59m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f41f25f20a3942188cdd78eda689558e : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":5},{"id":"753c68ab85e4b02c44843b01898d4dbf25f2624c","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-26 08:23:37.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1e0a296861bf208cba4cbf562f4003d2704f5bc8","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-02-26 08:44:57.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThanks for clearing this up.","accounts_in_message":[],"_revision_number":5},{"id":"f17ef562d172ee639577a259f546240ddabf8ae1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 08:45:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2f2528b5da744964035bccaecc22a636ec9825cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 10:50:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"d56d1cc4e88d924cf94c4a41800d2b9be376c1d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 10:50:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ab89e315a14b42ad9457f75859961211 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/970422adfa7d4638be77009fed2abdb4 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ed3ef82f48742e5a9c5fd508c0f13b1 : SUCCESS in 9m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/385d7b6bb50c49d8aa8c74cc16cd344a : SUCCESS in 10m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f63afdbce8a2483b87dc40d48551ba5f : SUCCESS in 8m 16s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a35f8ce8d71c4855ba618a6fb6ee1c66 : SUCCESS in 46m 08s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8285782f9abf44c4b220e750d2abea17 : SUCCESS in 46m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b84f3709ed924e1bb49c78bece23a046 : SUCCESS in 59m 16s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/72ad0d38ef8f4bfb8d2c82c70e89e228 : SUCCESS in 42m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3731e3f439da4189805f7b32bcc8b2e1 : SUCCESS in 30m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/11bb9a7343a64c9a89a6aab140828680 : SUCCESS in 29m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/d02c8b34a5754ba6b411915e6801730a : SUCCESS in 47m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b6960ce6fa54153a0040ceac82aee1d : SUCCESS in 10m 16s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/95d50f403db549d097d27443476ca2e0 : SUCCESS in 2h 04m 24s","accounts_in_message":[],"_revision_number":5},{"id":"245c37f214359a04dac5e88a03c65611b023a4b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 10:52:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad9208c9d7384dcc909682a16a5a8dd5 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"592981f487b1b4923c9d60a78ab4e236638bdf94","revisions":{"f77b959297a53c030f2618e564f7b2ce2abca08f":{"kind":"REWORK","_number":1,"created":"2020-02-21 07:14:04.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/57/709057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/57/709057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/57/709057/1"}}},"commit":{"parents":[{"commit":"08a91b186fe88cb7ffa971a4df599441245a4843","subject":"Add an article about profiling into the docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/08a91b186fe88cb7ffa971a4df599441245a4843"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-20 10:19:13.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-21 07:08:54.000000000","tz":420},"subject":"WIP: Refactor expressions","message":"WIP: Refactor expressions\n\n* This patch moves code related to YAQL and Jinja into their\n  specific modules so that there isn\u0027t any module that works with\n  both. It makes it easier to understand how code related to one\n  of these technologies works.\n* Custome built-in functions for YAQL and Jinja are now in a\n  separate module. It\u0027s easier now to see what\u0027s related with\n  the expression framework now and what\u0027s with integration part,\n  i.e. functions themselves.\n* Renamed the base module of expressions similar to other packages.\n* Other style changes.\n\nChange-Id: I94f57a6534b9c10e202205dfae4d039296c26407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f77b959297a53c030f2618e564f7b2ce2abca08f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f77b959297a53c030f2618e564f7b2ce2abca08f"}]},"branch":"refs/heads/master"},"f2bb8b4fb4be7c59722097a88d27e196eef44299":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-21 07:20:22.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/57/709057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/57/709057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/57/709057/2"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-20 10:19:13.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-21 07:19:59.000000000","tz":420},"subject":"WIP: Refactor expressions","message":"WIP: Refactor expressions\n\n* This patch moves code related to YAQL and Jinja into their\n  specific modules so that there isn\u0027t any module that works with\n  both. It makes it easier to understand how code related to one\n  of these technologies works.\n* Custome built-in functions for YAQL and Jinja are now in a\n  separate module. It\u0027s easier now to see what\u0027s related with\n  the expression framework now and what\u0027s with integration part,\n  i.e. functions themselves.\n* Renamed the base module of expressions similar to other packages.\n* Other style changes.\n\nChange-Id: I94f57a6534b9c10e202205dfae4d039296c26407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2bb8b4fb4be7c59722097a88d27e196eef44299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f2bb8b4fb4be7c59722097a88d27e196eef44299"}]},"branch":"refs/heads/master"},"43aaca3c1e6e7d7788981a2c381f0f8ca3c95174":{"kind":"REWORK","_number":3,"created":"2020-02-21 07:54:41.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/57/709057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/57/709057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/57/709057/3"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-20 10:19:13.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-21 07:54:27.000000000","tz":420},"subject":"WIP: Refactor expressions","message":"WIP: Refactor expressions\n\n* This patch moves code related to YAQL and Jinja into their\n  specific modules so that there isn\u0027t any module that works with\n  both. It makes it easier to understand how code related to one\n  of these technologies works.\n* Custome built-in functions for YAQL and Jinja are now in a\n  separate module. It\u0027s easier now to see what\u0027s related with\n  the expression framework now and what\u0027s with integration part,\n  i.e. functions themselves.\n* Renamed the base module of expressions similar to other packages.\n* Other style changes.\n\nChange-Id: I94f57a6534b9c10e202205dfae4d039296c26407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/43aaca3c1e6e7d7788981a2c381f0f8ca3c95174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/43aaca3c1e6e7d7788981a2c381f0f8ca3c95174"}]},"branch":"refs/heads/master"},"56a8e6b5ba6f17228cb16f952e9195b7640354b3":{"kind":"REWORK","_number":4,"created":"2020-02-21 08:10:38.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/57/709057/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/57/709057/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/57/709057/4"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-20 10:19:13.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-21 08:10:21.000000000","tz":420},"subject":"Refactor expressions","message":"Refactor expressions\n\n* This patch moves code related to YAQL and Jinja into their\n  specific modules so that there isn\u0027t any module that works with\n  both. It makes it easier to understand how code related to one\n  of these technologies works.\n* Custome built-in functions for YAQL and Jinja are now in a\n  separate module. It\u0027s easier now to see what\u0027s related with\n  the expression framework now and what\u0027s with integration part,\n  i.e. functions themselves.\n* Renamed the base module of expressions similar to other packages.\n* Other style changes.\n\nChange-Id: I94f57a6534b9c10e202205dfae4d039296c26407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/56a8e6b5ba6f17228cb16f952e9195b7640354b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/56a8e6b5ba6f17228cb16f952e9195b7640354b3"}]},"branch":"refs/heads/master"},"592981f487b1b4923c9d60a78ab4e236638bdf94":{"kind":"REWORK","_number":5,"created":"2020-02-26 05:37:12.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/57/709057/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/57/709057/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/57/709057/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/57/709057/5"}}},"commit":{"parents":[{"commit":"753f1bc03fc5ae464d6a3c8266e025dce3cceccb","subject":"Fix YAQL engine initialization","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/753f1bc03fc5ae464d6a3c8266e025dce3cceccb"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-20 10:19:13.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-26 05:36:34.000000000","tz":420},"subject":"Refactor expressions","message":"Refactor expressions\n\n* This patch moves code related to YAQL and Jinja into their\n  specific modules so that there isn\u0027t any module that works with\n  both. It makes it easier to understand how code related to one\n  of these technologies works.\n* Custome built-in functions for YAQL and Jinja are now in a\n  separate module. It\u0027s easier now to see what\u0027s related with\n  the expression framework now and what\u0027s with integration part,\n  i.e. functions themselves.\n* Renamed the base module of expressions similar to other packages.\n* Other style changes.\n\nChange-Id: I94f57a6534b9c10e202205dfae4d039296c26407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/592981f487b1b4923c9d60a78ab4e236638bdf94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/592981f487b1b4923c9d60a78ab4e236638bdf94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
