)]}'
{"id":"openstack%2Fmistral~709076","triplet_id":"openstack%2Fmistral~master~Ia7c260494a8a659988db2329bf3504be9b4b67fa","project":"openstack/mistral","branch":"master","topic":"duplication-removal","hashtags":[],"change_id":"Ia7c260494a8a659988db2329bf3504be9b4b67fa","subject":"Removing duplicated descriptions","status":"MERGED","created":"2020-02-21 09:53:10.000000000","updated":"2020-03-02 11:42:36.000000000","submitted":"2020-03-02 11:40:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709076-1583149232204-1d1e633c","meta_rev_id":"415bdff035f092ea01800385d522de077f0408d1","_number":709076,"virtual_id_number":709076,"owner":{"_account_id":22534,"name":"Andrea Visnyei","email":"andrea.visnyei@nokia.com","username":"visnyei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":31290,"name":"Edit Koselak","email":"edit.koselak@nokia.com","username":"koselak"},{"value":0,"_account_id":21723,"name":"Gergely Csatari","email":"gergely.csatari@nokia.com","username":"csatari"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-02 11:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":2,"date":"2020-03-02 10:41:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":31290,"name":"Edit Koselak","email":"edit.koselak@nokia.com","username":"koselak"},{"value":0,"_account_id":21723,"name":"Gergely Csatari","email":"gergely.csatari@nokia.com","username":"csatari"},{"value":2,"date":"2020-03-02 07:18:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":1,"date":"2020-03-02 10:41:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":31290,"name":"Edit Koselak","email":"edit.koselak@nokia.com","username":"koselak"},{"value":0,"_account_id":21723,"name":"Gergely Csatari","email":"gergely.csatari@nokia.com","username":"csatari"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21723,"name":"Gergely Csatari","email":"gergely.csatari@nokia.com","username":"csatari"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":31290,"name":"Edit Koselak","email":"edit.koselak@nokia.com","username":"koselak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-21 09:54:13.000000000","updated_by":{"_account_id":21723,"name":"Gergely Csatari","email":"gergely.csatari@nokia.com","username":"csatari"},"reviewer":{"_account_id":21723,"name":"Gergely Csatari","email":"gergely.csatari@nokia.com","username":"csatari"},"state":"REVIEWER"},{"updated":"2020-02-21 09:54:20.000000000","updated_by":{"_account_id":31290,"name":"Edit Koselak","email":"edit.koselak@nokia.com","username":"koselak"},"reviewer":{"_account_id":31290,"name":"Edit Koselak","email":"edit.koselak@nokia.com","username":"koselak"},"state":"REVIEWER"},{"updated":"2020-03-02 06:58:31.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-03-02 07:18:00.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-03-02 10:41:59.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-03-02 11:40:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae1cbdf7ef9db309382dc9d92c317ea3e7fff6ee","author":{"_account_id":22534,"name":"Andrea Visnyei","email":"andrea.visnyei@nokia.com","username":"visnyei"},"date":"2020-02-21 09:53:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1713dce6d92db7827bbaa6498c544933cf38570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 10:05:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a6886776f5947e78451bd0dd2b198f8 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf6f1569ba740f480e2f3f13853a573 : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3a2a3a3d60a7f7e09f47b72111b3ff0bded58a9a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-02 06:34:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a5a2bf1b9c0d6f406b8f9d1dd3aa52e05cd45a8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-02 06:58:20.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2d83f8f38cdcdd49d33e33cc8a6ba46c2dabbfd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 07:08:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aa626cf5d0b446b8659d08f45240d14 : FAILURE in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e139449079b45a8a88b7958f69df4c5 : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":2},{"id":"09fd71cf783ebcd90d8fee3bccd90da955c40fea","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-02 07:17:53.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"25d0076277018445852ab02d6581ee08583a2156","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-02 07:18:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ac5c08362e4863e957072db0a22e1c2af08d258a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 07:28:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838dda8096174371a1b081531deca752 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/350210cd6d734e7d99a10caf10b4f4b4 : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":3},{"id":"f188d03c2884656537c714d58787c2b652cfe36b","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-03-02 10:41:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"11bb5d102d064cdd6d5d109eb66b4aa5bd30185c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 10:42:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a5863f20322385a1de22b9bc8179d5d846430c6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 11:40:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03c90545bc82476e9616b6cd876e4371 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fff62c306fd473892828a1683760fc9 : SUCCESS in 11m 28s","accounts_in_message":[],"_revision_number":3},{"id":"26511c77ffbfa71715894bbedadab39a21765654","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 11:40:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0f8bb7c32e44caf2f8edb42a9ad4d0be1bd0fc85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 11:42:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e14bad9611b4467690730751af2176b7 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a683984001e41a2e9284f5dbe0711156ad248452","revisions":{"da46554bdcf79aaf77a885b1faac9322bd750416":{"kind":"REWORK","_number":1,"created":"2020-02-21 09:53:10.000000000","uploader":{"_account_id":22534,"name":"Andrea Visnyei","email":"andrea.visnyei@nokia.com","username":"visnyei"},"ref":"refs/changes/76/709076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/709076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/709076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/709076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/709076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/709076/1"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"Andrea Visnyei","email":"andrea.visnyei@nokia.com","date":"2020-02-21 09:51:25.000000000","tz":60},"committer":{"name":"Andrea Visnyei","email":"andrea.visnyei@nokia.com","date":"2020-02-21 09:51:25.000000000","tz":60},"subject":"Removing duplicated descriptions","message":"Removing duplicated descriptions\n\nRemoved dupliocated Mistral overviews and merged content into description on\n index page.\n\nChange-Id: Ia7c260494a8a659988db2329bf3504be9b4b67fa\nSigned-off-by: Andrea Visnyei \u003candrea.visnyei@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da46554bdcf79aaf77a885b1faac9322bd750416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/da46554bdcf79aaf77a885b1faac9322bd750416"}]},"branch":"refs/heads/master"},"f9babf22da53159dec3a02618b7cc52d78a4e7b0":{"kind":"REWORK","_number":2,"created":"2020-03-02 06:58:20.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/76/709076/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/709076/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/709076/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/709076/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/709076/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/709076/2"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"Andrea Visnyei","email":"andrea.visnyei@nokia.com","date":"2020-02-21 09:51:25.000000000","tz":60},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-02 06:58:16.000000000","tz":0},"subject":"Removing duplicated descriptions","message":"Removing duplicated descriptions\n\nRemoved dupliocated Mistral overviews and merged content into description on\n index page.\n\nChange-Id: Ia7c260494a8a659988db2329bf3504be9b4b67fa\nSigned-off-by: Andrea Visnyei \u003candrea.visnyei@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f9babf22da53159dec3a02618b7cc52d78a4e7b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f9babf22da53159dec3a02618b7cc52d78a4e7b0"}]},"branch":"refs/heads/master"},"a683984001e41a2e9284f5dbe0711156ad248452":{"kind":"REWORK","_number":3,"created":"2020-03-02 07:17:53.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/76/709076/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/76/709076/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/709076/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/709076/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/76/709076/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/76/709076/3"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"Andrea Visnyei","email":"andrea.visnyei@nokia.com","date":"2020-02-21 09:51:25.000000000","tz":60},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-02 07:17:48.000000000","tz":0},"subject":"Removing duplicated descriptions","message":"Removing duplicated descriptions\n\nRemoved dupliocated Mistral overviews and merged content into description on\n index page.\n\nChange-Id: Ia7c260494a8a659988db2329bf3504be9b4b67fa\nSigned-off-by: Andrea Visnyei \u003candrea.visnyei@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a683984001e41a2e9284f5dbe0711156ad248452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a683984001e41a2e9284f5dbe0711156ad248452"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
