)]}'
{"id":"openstack%2Fmistral~709144","triplet_id":"openstack%2Fmistral~master~I26bd00fc4babb3e41c5c9cc513b203177931204e","project":"openstack/mistral","branch":"master","topic":"bug/1864106","hashtags":[],"change_id":"I26bd00fc4babb3e41c5c9cc513b203177931204e","subject":"Links in Mistral README.rst are dead","status":"ABANDONED","created":"2020-02-21 13:48:25.000000000","updated":"2020-03-05 05:06:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8a8b64632d790dfa38c4c4abcf74c977960f2f28","_number":709144,"virtual_id_number":709144,"owner":{"_account_id":15279,"name":"Russell Tweed","email":"rtweed@redhat.com","username":"tedlit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-21 14:03:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-21 14:03:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65641f5d8bd10749de96b8c7f2ad527bfb4580e2","author":{"_account_id":15279,"name":"Russell Tweed","email":"rtweed@redhat.com","username":"tedlit"},"date":"2020-02-21 13:48:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8167a7bd430943bde3b3b414f663946444fe56ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 14:03:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e6bfd5be6df42b982c30c0ca662c87e : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b01cbe0f69488fb3515ca829c657bc : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e44278e67104a35a7c68b0dabb6ee84 : SUCCESS in 9m 44s","accounts_in_message":[],"_revision_number":1},{"id":"dfc44fd5bd46255ef7697bd48df5f3cf1caca99a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-05 05:06:05.000000000","message":"Abandoned\n\nSorry, we somehow missed this patch when it was created. And now the issue has already been fixed.\n\nThanks!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f4c5fd17d4ca3a2262b3d89e4efe008126990cb","revisions":{"6f4c5fd17d4ca3a2262b3d89e4efe008126990cb":{"kind":"REWORK","_number":1,"created":"2020-02-21 13:48:25.000000000","uploader":{"_account_id":15279,"name":"Russell Tweed","email":"rtweed@redhat.com","username":"tedlit"},"ref":"refs/changes/44/709144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/44/709144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/709144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/709144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/44/709144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/44/709144/1"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"Russell Tweed","email":"rtweed@redhat.com","date":"2020-02-21 13:47:23.000000000","tz":0},"committer":{"name":"Russell Tweed","email":"rtweed@redhat.com","date":"2020-02-21 13:47:23.000000000","tz":0},"subject":"Links in Mistral README.rst are dead","message":"Links in Mistral README.rst are dead\n\nFix links in README to reflect changes in documentation tree structure\n\nChange-Id: I26bd00fc4babb3e41c5c9cc513b203177931204e\nCloses-Bug: 1864106\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6f4c5fd17d4ca3a2262b3d89e4efe008126990cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6f4c5fd17d4ca3a2262b3d89e4efe008126990cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
