)]}'
{"id":"openstack%2Fmistral~709547","triplet_id":"openstack%2Fmistral~master~Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738","project":"openstack/mistral","branch":"master","topic":"rally-scenario","hashtags":[],"change_id":"Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738","subject":"Added two new Rally scenarios","status":"MERGED","created":"2020-02-24 15:04:11.000000000","updated":"2020-04-02 12:36:48.000000000","submitted":"2020-04-02 12:31:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709547-1585830716677-d25266a9","meta_rev_id":"6545077d37e236a3ff449d4a2d03924abb06a526","_number":709547,"virtual_id_number":709547,"owner":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 12:31:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-04-02 11:30:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-02 11:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"all":[{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-02 11:31:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 11:30:08.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-04-02 11:31:51.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-02 12:31:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"427f227aec95be3cc3dece55027a7ccb0714b515","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-02-24 15:04:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f021ee76e6a7ef88450a4d4134f1fbcb37da333","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-02-24 15:09:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b58f727bec2c291f271276a97a5b15cab9fec0c2","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-02-24 15:09:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0102dd612216552bd3f9197e05e0858738697d56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 18:18:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/064b5207802244209fcc44871a397c59 : FAILURE in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9be6dcfcc34430d8a0f65edcb6c49a3 : SUCCESS in 15m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f1075aa9f8b40a584a4859949bdd11f : SUCCESS in 14m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3287d93843d45e69e5e66b408af076c : SUCCESS in 14m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1a5fc62a4843b688d01130309387a8 : SUCCESS in 14m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c995fa1dd23542d5a91ea217104da2a2 : SUCCESS in 13m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3de8de9dd94f4d85a5ff1e18502c8ae7 : FAILURE in 38m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/234d68ce021048d4bd658841034a8566 : FAILURE in 39m 15s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4812057a03154087b660ea5cc9fd134e : FAILURE in 38m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1b659e09c893482daec881d72ea9cb9c : FAILURE in 39m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/eec54a7630784f51a956d84f635c1533 : FAILURE in 43m 26s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9f4652ce920741e0a28ae2425fb8d981 : FAILURE in 31m 21s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1ed11523f222468982e707cd4fa34657 : FAILURE in 31m 39s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/178a209ad03f4ba987198f0a1128990d : SUCCESS in 3h 02m 01s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a73b7e7c914e44d3bee5cdd9849d4249 : FAILURE in 40m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e84a10a3d5674e7c8f754b55c13dd5dc : SUCCESS in 14m 28s","accounts_in_message":[],"_revision_number":3},{"id":"2bef38958e28294d87cda3d502babd450c645d36","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-02-25 06:51:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"41026eee46dd1f94c8b8c9f05ede98cddf017139","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-02-25 08:50:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"eb949a1dec0c3e51a2a7633bd61732db16b9f3c7","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-02-25 10:07:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"539c43fe5887093d689d360e98962c65266e7fb0","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-02-25 12:23:45.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2329a8bd0ffd10432073e187ad616bd79cd5748b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 15:23:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e31752162f354ccd8844952cc400e429 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8555153e8fe74b5ebb1100500e3d392e : SUCCESS in 10m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21c02b96954042ccae76dbd22f812d34 : SUCCESS in 10m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59d2e643fbda4c9e9b76432d20b89982 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7558f05d25a3401a87783aa9199ce560 : SUCCESS in 9m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3195d39360434d75b028566bd3028175 : SUCCESS in 10m 37s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/26a1600d18f94aaf962e61f356595d13 : SUCCESS in 43m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cd1e6b88e404fd6a3bfebb865cbce7f : SUCCESS in 53m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6232a469aac34f41951fbd5e75d2366d : SUCCESS in 58m 47s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dcc8050ed5964779a26b581cab58a026 : SUCCESS in 44m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/6534ca178ca54e379fa95a6085a41b78 : SUCCESS in 41m 59s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/8befb576b48f480cba274e7e81f83c69 : SUCCESS in 31m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/60e233cf33274da385d7f168504eb7cd : SUCCESS in 28m 45s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/197360ac35144f009ec5a225e7b03264 : SUCCESS in 2h 51m 24s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c85ece0d24f849f69e780f2a867c2697 : FAILURE in 1h 00m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4346031d5f04ff9ac7938c4dd550cf9 : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":7},{"id":"e707aeccadc41d00b1588642243bd5b1d5ef4aba","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-03-16 08:17:35.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"aa88bdee9863109a2184b7acb4611a022727694a","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-03-16 08:27:45.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1e3fe42172fd9d76398abbf1447aefe45cd25013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 09:47:10.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e06d41ce155a46e6bfa63fa3f6737ba9 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d983ffd7042940a691bdeb6b378b10ed : SUCCESS in 8m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d6eb63ce626440cafcf3d7713a44be5 : SUCCESS in 11m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d06d2eef17f4f79a2ea1f7cdcc2c287 : SUCCESS in 10m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6513b6aac764476aae97ded0d8e9fc78 : SUCCESS in 9m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b7771ad2cd24f0697ffc704d9cd0aaa : SUCCESS in 11m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4418aa5fbe184094b428ea4f1d258470 : SUCCESS in 43m 49s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d379b134a38447d6abe245ed4dacc462 : SUCCESS in 52m 25s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b57a02776801491ab086deaded8a09d0 : SUCCESS in 58m 27s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d9f60e916dfc400cacdf377f0bd021d6 : SUCCESS in 51m 01s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/722a0f9da40142d3b7ca442c9e5bba83 : SUCCESS in 59m 29s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c248a12302f94135928727db79db9178 : SUCCESS in 31m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/546c0da0addf48edbe550f958456646f : SUCCESS in 27m 41s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/03ca28dbb3ee49388fda9849e6899e35 : SUCCESS in 1h 14m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d9a7adf4dbb484c8adaec305fb47530 : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":9},{"id":"2101ba4018d4feea9f9d5ee942e8f3463eebe0e8","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-03-19 08:17:22.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d55505491cb5ff14319dcc351603a0575105fa33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 09:28:52.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b8bdb29d934e2595ffcd222edde821 : FAILURE in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60c5377ace294f5ea84729a96a31593a : SUCCESS in 8m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87d76e481933465eb6674cc53e16a9a1 : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d20e19eab0f345e3b0ff85b073e519b0 : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d78ee059828450683983a1c9aafcf14 : SUCCESS in 10m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08ecead9344b45259e106024006254ec : SUCCESS in 8m 55s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2f7538452b8448aa9504fec3fdfff90b : SUCCESS in 46m 15s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d510de92a3744a28d8e2f18b87f8641 : SUCCESS in 47m 17s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9303695a747a4adba18b0b46b483146a : SUCCESS in 48m 15s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1b70deea051048719e73c9dcd0c22e34 : SUCCESS in 48m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9732b2cc6a954dd7aa96bfba2909eea8 : SUCCESS in 43m 01s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3037dd0e3541437786022e0f33f938c0 : FAILURE in 29m 01s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/076b5bf3980f4f8b8d560aa98de4ea01 : FAILURE in 27m 23s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/0ee9fc2ab6be42e8b65472dcac3a6ca1 : SUCCESS in 1h 09m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa9d64cd6a2a4ebf9deb5dcfb000fcb8 : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":10},{"id":"5b2d7fef112815ce70b051cf9298cf38285d132a","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-03-19 10:59:10.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"b3d153b5cbbf42c317bedce0e5b988abed15287c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 12:21:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e92161723b934793a97338c41496d2d2 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e3a6bf20349401498f1acbeafed2a47 : SUCCESS in 8m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ff57f9ece0a4f80b8e7938735c83fc6 : SUCCESS in 9m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b42fce17b5d49d19607c8dcb2cb2b10 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0214862817941acb013ae8407e51060 : SUCCESS in 8m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8862a68bdb7c453f90452caa2e10c793 : SUCCESS in 8m 52s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a389ecb8f0024f1faeeaada0955591b4 : SUCCESS in 43m 53s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c3d3f4a287a47b09d61973dffb7729d : SUCCESS in 49m 49s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8cfacf1fdf84980a74c5fc37c6ae14a : SUCCESS in 43m 11s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c87035207ec14cc29460488f2fa8438e : SUCCESS in 47m 19s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c082df0ce0bd4f7d8fb29e1c600371a1 : SUCCESS in 43m 36s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bf39f4b3dde04704bf1416a5de359065 : SUCCESS in 30m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d16c42d85838412f8938f8d9fe5df4fe : SUCCESS in 28m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/fd4beba82b1444f2af7c3a85b582df11 : SUCCESS in 1h 12m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4cef7e31e654ec5b21887b03373cec7 : SUCCESS in 11m 20s","accounts_in_message":[],"_revision_number":11},{"id":"379b28eeb1c9cf797c070994608decf2a8e34db5","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-19 12:28:47.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6cbbc9bd105be116e3f884e43c38ae0e8f81f252","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-03-22 08:11:06.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"10efb2280238ffdf4c5e8ccbd91d7ecc01f3b80a","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-03-22 08:14:53.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"944e678b050708147793baca1b3aed438aa86030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 09:16:16.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645e441cc36844429c3bd6bb6ac7406a : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd779691181645b18b72d88c4d7d1f20 : SUCCESS in 9m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a26ef7307e5441e48ef682b3a7595cbb : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e4ce9a7244d42889aefe9ce34d7e294 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5987d6e7654742e5a660ba056133dfe5 : SUCCESS in 6m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65d820cd66c84b349e44be7da83302c6 : SUCCESS in 10m 23s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e598471638684beb84c83f7fc6b41d8e : SUCCESS in 41m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b3bafc9135e4c3e828019f98a27f140 : SUCCESS in 38m 29s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62ee3479d4da452982953a6a28bbf3aa : SUCCESS in 44m 19s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2316d953842b4c46b108df2d64a2f43d : FAILURE in 46m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/31a46b021f3d4230ac207d339eadc66d : SUCCESS in 41m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/837778a5788f409bb25d607f54c697a1 : SUCCESS in 29m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1c940743ef7542b1a8e526684e6fed0c : SUCCESS in 28m 33s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1659004b9cdb429699e74a7cb867edb3 : SUCCESS in 1h 00m 44s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de24779862174ef888cfd80b44979249 : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":13},{"id":"80d4c0fc7c448cc077eb8eaeaa26ce2aa0e729ac","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-03-22 11:59:40.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"941042029e544d88ca098e1e019b32b62bb94a0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-22 13:10:39.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bb0eabec3a4430a9830dea277660b1a : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d3887c9e045436e84f0a6648d9b5ba4 : SUCCESS in 9m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/302142a9f9734d6faeef3cfea7ecb59e : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c30b8e1397c43e5b7313fdcbfba806c : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d037e590c8417fb3a3b34242f11e6a : SUCCESS in 6m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccc991bf519f42ca8fa0f77108cedbec : SUCCESS in 8m 40s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8a054fb528074227b880d4b8c55cc047 : SUCCESS in 46m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04bbadbb838249179789e42bad50420a : SUCCESS in 40m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b05afb2a7bf34e6885990a7bb3dca239 : SUCCESS in 45m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/cbd7a3767dc44dc5b1ca8681097448dd : SUCCESS in 44m 00s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/210936c523494d4eaad11d8492eb1ce6 : SUCCESS in 51m 03s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e04bfd05ed7940c2b09cccdf9c5647ad : SUCCESS in 28m 17s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ad89fbdb6f9446948623d78fdd62da1c : SUCCESS in 26m 31s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/e172b8312cdf49c08997182a52d3c471 : SUCCESS in 1h 05m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7cee01d8a14424794ff6c461beb2a43 : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":13},{"id":"7140f5d7c0aaca300d4d5d2168ef39873eaea5e3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-24 15:20:45.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"edd00948471beb2ba2d5a45bee3083037d0e81a4","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-25 09:13:58.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"a6df40527c804cbac352a0420248986445599e88","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-26 04:29:20.000000000","message":"Patch Set 13: Code-Review-1\n\nAli, please clarify the config option default value question.","accounts_in_message":[],"_revision_number":13},{"id":"a9891bb53700f3672f64fc5710ec44949fadba89","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-03-29 06:51:14.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"47d9bb9357ea972d3bcfc61d5800221d9f21572a","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-03-29 06:51:25.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"aac201da6a2c4b03b231daa3b69cc467a6de213a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 08:02:08.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bca29e21d9b42fcbcd6d1444eb1465f : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d19676f5a24427e92f23a33abdf21a2 : SUCCESS in 9m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/513dd48215d54b018307ad9c685a3ebe : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a155f96579ff49bd86363a8ce3f14291 : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d9a4d91f764dc1a4b150ee56df70c2 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09a47761b32d413689679fd28c38c629 : SUCCESS in 9m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1349d51fce504e8faf27b18f391264c5 : SUCCESS in 43m 01s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bac2cadba534f6db0c87b5ac93866e4 : SUCCESS in 45m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f582a765bd94a19b29dc89768c29983 : SUCCESS in 38m 23s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/fe0426337230474e86dfc273e39d0300 : SUCCESS in 42m 31s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/46f1b5b4cfd145329fd8b9421d009f44 : SUCCESS in 42m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0ece8fb56d2f4e9ab2d3e8880941eaea : SUCCESS in 28m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/9d823ba884c84a299e9c1ca1e85c3408 : SUCCESS in 27m 19s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/b172f68e74e24e7fa327015d8571938a : SUCCESS in 1h 05m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47cd497fbe84431c8a3b553ad8d72cfd : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":14},{"id":"903f4422fbfd7796835dad47bd752b8d4eec7dbf","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-04-01 07:11:42.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"2d9aa4f2440be49aa2b5f8add940493bed1dcb16","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-04-01 07:18:47.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"54dda70786e4b70f20f2d152af824d8f3b8aff5e","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-04-01 07:19:57.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"947bec883514060aeac83a18237a88ad1fe32c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 08:21:19.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/437b858216434af9a54eb4cd112d9b8a : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56685f514fde4a85bd0250ca39f17184 : SUCCESS in 10m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a769b6a8e0b42ab8e8ba8f8f6949a4c : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebc6d8605e49475c990c68bf55e324c6 : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de56d300f7824912b4ecfdd42cf088be : SUCCESS in 8m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/476727c3210542769ae07302bfd8c9aa : SUCCESS in 11m 42s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9eca665913ef46c699c8b66757a9e499 : SUCCESS in 42m 47s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/284294b5f44648b3b90aafc7b435aaff : SUCCESS in 47m 31s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04379982d80e4fdba3361bb124e6e09f : SUCCESS in 45m 35s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/14f551344371409eb52ff6b6f7b4734b : SUCCESS in 39m 18s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0066188795c4402e9cb06ee1eb467556 : SUCCESS in 40m 25s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3ed76d60866b411b9803a56ada64fcfa : SUCCESS in 31m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/4991d3d0073d44f69d7ba18dc6c880ab : SUCCESS in 30m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a469568ee7214fc58a8a052c0330be95 : FAILURE in 57m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f69b37bbe77c42daac3e4d9400894e1e : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":17},{"id":"d55c6013978afddb8360786a0920166c98af3bf5","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-04-01 08:24:16.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"ee0b9aee80f3109d336b595f0d277861ad440f67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 09:26:04.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71911d519e074caf877cd4aa394f2e45 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96fe238847e34ffbb519251404b4d3fe : SUCCESS in 10m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a0cce11e2cc4a8bad07488a9650909d : SUCCESS in 9m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cad747e3e2a490eae2647a872812bf6 : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/678462319e1e4d84a2af033673ea6f36 : SUCCESS in 9m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71323b1f5eba4cfe8641d7b0fc6430d3 : SUCCESS in 14m 51s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b13e7caf7449453994972927e5706e6a : SUCCESS in 48m 03s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b40af7c0886468ca985f19510004b1d : SUCCESS in 44m 38s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/866f3a4b09a640dd9ebdb0d60e02d9ff : SUCCESS in 45m 05s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b6af2857c156450789a7cfded7eb6759 : SUCCESS in 43m 16s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/06ec64db0d3c4dacb136aaf6aeb54eb7 : FAILURE in 44m 28s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b9b2615f6c6041afba2b02f502c8234b : SUCCESS in 31m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/beeed015580b4202b6b6dd1fe493af42 : SUCCESS in 30m 30s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d5019a3f0365445191dee8f6ff099f43 : FAILURE in 59m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ec5a51b05c243e88c63961cf46a07bb : SUCCESS in 10m 23s","accounts_in_message":[],"_revision_number":18},{"id":"f7cfb5ac49728d33b9d37519c98fbddd8c3a0c6b","author":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"date":"2020-04-02 08:36:47.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"1b9ad9671ec3586786f9002415355127978b96a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 09:51:49.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac8669611f44745a5783f61377f3d84 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e989866ea5440beabfddd604ab8e52b : SUCCESS in 9m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccda9b0627a8473daed1a5979b369cba : SUCCESS in 10m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd9134122ba24cc3b290660c90a9dbce : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/def61e6942f84a1ca504940e13185f99 : SUCCESS in 9m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3c467995e4c47eb81801390a60a0be1 : SUCCESS in 9m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4ba30dfd88404868ba2506ac52404e64 : SUCCESS in 47m 35s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1adfb5d3caba471eb8cf49ccbed22d48 : SUCCESS in 49m 52s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9df55b43d6b40d78bf1cb1fa72efb97 : SUCCESS in 43m 56s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/60337f15344d4d028e4cfc53fac6858c : SUCCESS in 47m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/f31149ebebd84e819c51843fed95a31c : SUCCESS in 45m 28s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3db91e83f36043b1828701ba5840d837 : SUCCESS in 33m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/a55f667401e541e7aa925ef0fe3f7865 : SUCCESS in 28m 36s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/1af1b466c07846da928fcbbeb5c87276 : FAILURE in 1h 08m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3d6ae901ea34e749d7c67e66438483a : SUCCESS in 9m 53s","accounts_in_message":[],"_revision_number":19},{"id":"05a90375be3e61cf546870da94b0e7a3fd785852","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-02 10:08:00.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"5c3c2988d2b91d53f9a2a3e24b9e84dccafc6167","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 11:21:04.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99db167066224bf9a724874915231b8f : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f71fefe0701147e3aea664f54ef6369b : SUCCESS in 9m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97ef8b54767d4664a8ffe5206b401ecf : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f8889ef18774cf69f43b98804c24390 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cbf5b93baf74b729fb438b763615f86 : SUCCESS in 8m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8efad612c7224775ae9111b10efe5fb8 : SUCCESS in 10m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7ecf80ee1e0a4b0e84d6f3dc7225a6ec : SUCCESS in 38m 18s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0488759e987640eea939e5e950617ea3 : SUCCESS in 38m 33s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81b32200153a4ca7ae6a2c9e60382c0d : SUCCESS in 43m 21s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3c9909b9437a4ab3b3ce1ae32b43b8d9 : SUCCESS in 39m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ce9c86f1da324a6f895fd72588578e15 : SUCCESS in 41m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fcb983f4132249968f2ca9473668aa0c : SUCCESS in 31m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1da0850f1afb4b5894ba550b1971861b : SUCCESS in 28m 54s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a45405c0fa064423972346341859e0ab : SUCCESS in 1h 03m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4ee037adf6340749500f73f8d33ca7a : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":20},{"id":"2d2147e9605b7a3b6fced44062b459ef3d0acb03","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-02 11:30:08.000000000","message":"Patch Set 20: Code-Review+2","accounts_in_message":[],"_revision_number":20},{"id":"61fbe0f3452a01910afdfa6a608a5d5e2cb22935","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-02 11:31:51.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"1c26e9ce812e2723e166cedf118b22398eddc412","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 11:32:14.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"8547b4c8706d0dbd9f94ad93886660bf27f53045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 12:31:56.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8ded630bcd04cbca151715d9fb27636 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/040edeee6cd343f4ad1f07a234a7f22b : SUCCESS in 10m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56fc333fd4a4408286c41522c35fa2ab : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037852172824414da19f6752cef88409 : SUCCESS in 11m 34s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d5ca2edb723b4bcd98a7626ff14af38f : SUCCESS in 39m 55s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/533150cbd0c546d6b7dd8ef8167be25c : SUCCESS in 56m 58s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0772e14476de42a4b158370431fcd6ff : SUCCESS in 53m 04s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3b02a544b254474db11e0640a610e002 : SUCCESS in 46m 21s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/4e88d1059c7e4da083304d2f4a821983 : SUCCESS in 30m 45s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ee463c6a5f51477daad6fc8cf0136092 : SUCCESS in 31m 50s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9d4d043046e0441c86616b24c5fba7b7 : SUCCESS in 49m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8e5376e750645698384717f30d7ca9d : SUCCESS in 12m 30s","accounts_in_message":[],"_revision_number":20},{"id":"42631c9743371ab4205ec96c7fb5c739087f3a35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 12:31:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20},{"id":"22db8b4e952d2f50239ca295b2334210a67e29c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 12:36:48.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86659892d75a4db9bc690992a9c49eec : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"75d01bfbd325b9fccdede30b57e3e9c738ae40e5","revisions":{"eebb312f664fab99e85e114110269105cb51bebb":{"kind":"REWORK","_number":1,"created":"2020-02-24 15:04:11.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/1"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eebb312f664fab99e85e114110269105cb51bebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eebb312f664fab99e85e114110269105cb51bebb"}]},"branch":"refs/heads/master"},"79a00900921c5845a6d9bd0a4820bde69cd9b15a":{"kind":"REWORK","_number":2,"created":"2020-02-24 15:09:12.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/2"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:08:56.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/79a00900921c5845a6d9bd0a4820bde69cd9b15a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/79a00900921c5845a6d9bd0a4820bde69cd9b15a"}]},"branch":"refs/heads/master"},"07f3bd97a7e94ac0429c4053232e36b43a355275":{"kind":"REWORK","_number":3,"created":"2020-02-24 15:09:53.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/3"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:09:38.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/07f3bd97a7e94ac0429c4053232e36b43a355275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/07f3bd97a7e94ac0429c4053232e36b43a355275"}]},"branch":"refs/heads/master"},"eb6482b91afcdefc921a6e0c3b9cdce1218e52cf":{"kind":"REWORK","_number":4,"created":"2020-02-25 06:51:26.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/4"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-25 06:51:14.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eb6482b91afcdefc921a6e0c3b9cdce1218e52cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/eb6482b91afcdefc921a6e0c3b9cdce1218e52cf"}]},"branch":"refs/heads/master"},"8e0b408d872f1e9cebe0d5cd43667bbddfe91ada":{"kind":"REWORK","_number":5,"created":"2020-02-25 08:50:45.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/5"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-25 08:50:32.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8e0b408d872f1e9cebe0d5cd43667bbddfe91ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8e0b408d872f1e9cebe0d5cd43667bbddfe91ada"}]},"branch":"refs/heads/master"},"c3cc58c62d693bc240cc8c5517ce6e8c3d575a1e":{"kind":"REWORK","_number":6,"created":"2020-02-25 10:07:41.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/6"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-25 10:07:26.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c3cc58c62d693bc240cc8c5517ce6e8c3d575a1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c3cc58c62d693bc240cc8c5517ce6e8c3d575a1e"}]},"branch":"refs/heads/master"},"047e4babf28f46860cf43a3e81c55fbbe1f00847":{"kind":"REWORK","_number":7,"created":"2020-02-25 12:23:45.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/7"}}},"commit":{"parents":[{"commit":"99e004b3d636f1e72258819483401eb5069723a4","subject":"Merge \"Fix adhoc actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99e004b3d636f1e72258819483401eb5069723a4"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-25 12:23:29.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/047e4babf28f46860cf43a3e81c55fbbe1f00847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/047e4babf28f46860cf43a3e81c55fbbe1f00847"}]},"branch":"refs/heads/master"},"79cb350e9f383be1598b411f5ceacd88084dca0d":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-03-16 08:17:35.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/8"}}},"commit":{"parents":[{"commit":"dbfc0bea227e45ba36756bc5e91aba36f83c2f09","subject":"Merge \"Fix incorrect in-depth search of affected tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfc0bea227e45ba36756bc5e91aba36f83c2f09"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali abdelal","email":"ali.abdelal@nokia.com","date":"2020-03-16 08:17:35.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/79cb350e9f383be1598b411f5ceacd88084dca0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/79cb350e9f383be1598b411f5ceacd88084dca0d"}]},"branch":"refs/heads/master"},"b40986654268dbedf1f4d5f02821a0a38f768593":{"kind":"REWORK","_number":9,"created":"2020-03-16 08:27:45.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/9"}}},"commit":{"parents":[{"commit":"dbfc0bea227e45ba36756bc5e91aba36f83c2f09","subject":"Merge \"Fix incorrect in-depth search of affected tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfc0bea227e45ba36756bc5e91aba36f83c2f09"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-03-16 08:27:32.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b40986654268dbedf1f4d5f02821a0a38f768593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b40986654268dbedf1f4d5f02821a0a38f768593"}]},"branch":"refs/heads/master"},"a9a33c3b5b9366544ac369520322565cfb000df3":{"kind":"REWORK","_number":10,"created":"2020-03-19 08:17:22.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/10"}}},"commit":{"parents":[{"commit":"dbfc0bea227e45ba36756bc5e91aba36f83c2f09","subject":"Merge \"Fix incorrect in-depth search of affected tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfc0bea227e45ba36756bc5e91aba36f83c2f09"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-03-19 08:16:55.000000000","tz":0},"subject":"Added two new Rally scenarios","message":"Added two new Rally scenarios\n\nthe new scenarios are a workflow that contains Jinja,\nYaql expressions on a relatively big json object\n(4mb in size).\n\nthe input for these two Scenarios is :\n 1. the number of tasks in the workflow\n    - 30 in rally task\n 2. the size of the param object in mb\n    - 4mb in rally task\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a9a33c3b5b9366544ac369520322565cfb000df3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a9a33c3b5b9366544ac369520322565cfb000df3"}]},"branch":"refs/heads/master"},"356cac15423d364d1ebee71367f3ea0b3ad10cea":{"kind":"REWORK","_number":11,"created":"2020-03-19 10:59:10.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/11"}}},"commit":{"parents":[{"commit":"dbfc0bea227e45ba36756bc5e91aba36f83c2f09","subject":"Merge \"Fix incorrect in-depth search of affected tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfc0bea227e45ba36756bc5e91aba36f83c2f09"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-03-19 10:57:47.000000000","tz":0},"subject":"Added two new Rally scenarios","message":"Added two new Rally scenarios\n\nthe new scenarios are a workflow that contains Jinja,\nYaql expressions on a relatively big json object\n(4mb in size).\n\nthe input for these two Scenarios is :\n 1. the number of tasks in the workflow\n    - 30 in rally task\n 2. the size of the param object in mb\n    - 4mb in rally task\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/356cac15423d364d1ebee71367f3ea0b3ad10cea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/356cac15423d364d1ebee71367f3ea0b3ad10cea"}]},"branch":"refs/heads/master"},"02d5517fdcacb202f2500046feb1c8d644a3ef4b":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-03-22 08:11:06.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/12"}}},"commit":{"parents":[{"commit":"8a5d35ac6f7a8677b33354d8a8fe8ea6a3738ea6","subject":"Merge \"Keep openstack mapping_path option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8a5d35ac6f7a8677b33354d8a8fe8ea6a3738ea6"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali abdelal","email":"ali.abdelal@nokia.com","date":"2020-03-22 08:11:06.000000000","tz":0},"subject":"Added two new Rally scenarios","message":"Added two new Rally scenarios\n\nthe new scenarios are a workflow that contains Jinja,\nYaql expressions on a relatively big json object\n(4mb in size).\n\nthe input for these two Scenarios is :\n 1. the number of tasks in the workflow\n    - 30 in rally task\n 2. the size of the param object in mb\n    - 4mb in rally task\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/02d5517fdcacb202f2500046feb1c8d644a3ef4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/02d5517fdcacb202f2500046feb1c8d644a3ef4b"}]},"branch":"refs/heads/master"},"e4257d7a313bb5f933954e01c6dc99e35de7851b":{"kind":"REWORK","_number":13,"created":"2020-03-22 08:14:53.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/13"}}},"commit":{"parents":[{"commit":"8a5d35ac6f7a8677b33354d8a8fe8ea6a3738ea6","subject":"Merge \"Keep openstack mapping_path option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8a5d35ac6f7a8677b33354d8a8fe8ea6a3738ea6"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-03-22 08:14:29.000000000","tz":0},"subject":"Added two new Rally scenarios","message":"Added two new Rally scenarios\n\nthe new scenarios are a workflow that contains Jinja,\nYaql expressions on a relatively big json object\n(4mb in size).\n\nthe input for these two Scenarios is :\n 1. the number of tasks in the workflow\n    - 30 in rally task\n 2. the size of the param object in mb\n    - 4mb in rally task\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4257d7a313bb5f933954e01c6dc99e35de7851b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/e4257d7a313bb5f933954e01c6dc99e35de7851b"}]},"branch":"refs/heads/master"},"dba4c8c4167fd1b9a7042891bf776496a04168ec":{"kind":"REWORK","_number":14,"created":"2020-03-29 06:51:14.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/14"}}},"commit":{"parents":[{"commit":"8a5d35ac6f7a8677b33354d8a8fe8ea6a3738ea6","subject":"Merge \"Keep openstack mapping_path option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/8a5d35ac6f7a8677b33354d8a8fe8ea6a3738ea6"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali abdelal","email":"ali.abdelal@nokia.com","date":"2020-03-29 06:51:08.000000000","tz":0},"subject":"Added two new Rally scenarios","message":"Added two new Rally scenarios\n\nthe new scenarios are a workflow that contains Jinja,\nYaql expressions on a relatively big json object\n(4mb in size).\n\nthe input for these two Scenarios is :\n 1. the number of tasks in the workflow\n    - 30 in rally task\n 2. the size of the param object in mb\n    - 4mb in rally task\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dba4c8c4167fd1b9a7042891bf776496a04168ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dba4c8c4167fd1b9a7042891bf776496a04168ec"}]},"branch":"refs/heads/master"},"1227944593f71949e1b2aeb62a9dfb4e06b8dd48":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-04-01 07:11:42.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/15"}}},"commit":{"parents":[{"commit":"869cac9def302aa88142901c1cd98ec4d88b5b65","subject":"Merge \"Fix serialization of structures that might contain YAQL types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/869cac9def302aa88142901c1cd98ec4d88b5b65"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali abdelal","email":"ali.abdelal@nokia.com","date":"2020-04-01 07:11:42.000000000","tz":0},"subject":"Added two new Rally scenarios","message":"Added two new Rally scenarios\n\nthe new scenarios are a workflow that contains Jinja,\nYaql expressions on a relatively big json object\n(4mb in size).\n\nthe input for these two Scenarios is :\n 1. the number of tasks in the workflow\n    - 30 in rally task\n 2. the size of the param object in mb\n    - 4mb in rally task\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1227944593f71949e1b2aeb62a9dfb4e06b8dd48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1227944593f71949e1b2aeb62a9dfb4e06b8dd48"}]},"branch":"refs/heads/master"},"ecc1cadd08bbea79202d24d530bd53943ce4c475":{"kind":"REWORK","_number":16,"created":"2020-04-01 07:18:47.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/16"}}},"commit":{"parents":[{"commit":"869cac9def302aa88142901c1cd98ec4d88b5b65","subject":"Merge \"Fix serialization of structures that might contain YAQL types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/869cac9def302aa88142901c1cd98ec4d88b5b65"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-04-01 07:18:31.000000000","tz":0},"subject":"Added two new Rally scenarios","message":"Added two new Rally scenarios\n\nthe new scenarios are a workflow that contains Jinja,\nYaql expressions on a relatively big json object\n(4mb in size).\n\nthe input for these two Scenarios is :\n 1. the number of tasks in the workflow\n    - 30 in rally task\n 2. the size of the param object in mb\n    - 4mb in rally task\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ecc1cadd08bbea79202d24d530bd53943ce4c475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ecc1cadd08bbea79202d24d530bd53943ce4c475"}]},"branch":"refs/heads/master"},"4894c2cb68221db1fd6fda5dacb92f6c9a44b31f":{"kind":"REWORK","_number":17,"created":"2020-04-01 07:19:57.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/17"}}},"commit":{"parents":[{"commit":"869cac9def302aa88142901c1cd98ec4d88b5b65","subject":"Merge \"Fix serialization of structures that might contain YAQL types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/869cac9def302aa88142901c1cd98ec4d88b5b65"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-04-01 07:19:47.000000000","tz":0},"subject":"Added two new Rally scenarios","message":"Added two new Rally scenarios\n\nthe new scenarios are a workflow that contains Jinja,\nYaql expressions on a relatively big json object\n(4mb in size).\n\nthe input for these two Scenarios is :\n 1. the number of tasks in the workflow\n    - 30 in rally task\n 2. the size of the param object in mb\n    - 4mb in rally task\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4894c2cb68221db1fd6fda5dacb92f6c9a44b31f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4894c2cb68221db1fd6fda5dacb92f6c9a44b31f"}]},"branch":"refs/heads/master"},"ee5186eae6feb49fd602a012266c0b160d07cdda":{"kind":"REWORK","_number":18,"created":"2020-04-01 08:24:16.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/18"}}},"commit":{"parents":[{"commit":"869cac9def302aa88142901c1cd98ec4d88b5b65","subject":"Merge \"Fix serialization of structures that might contain YAQL types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/869cac9def302aa88142901c1cd98ec4d88b5b65"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali abdelal","email":"ali.abdelal@nokia.com","date":"2020-04-01 08:24:10.000000000","tz":0},"subject":"Added two new Rally scenarios","message":"Added two new Rally scenarios\n\nthe new scenarios are a workflow that contains Jinja,\nYaql expressions on a relatively big json object\n(4mb in size).\n\nthe input for these two Scenarios is :\n 1. the number of tasks in the workflow\n    - 30 in rally task\n 2. the size of the param object in mb\n    - 4mb in rally task\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee5186eae6feb49fd602a012266c0b160d07cdda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ee5186eae6feb49fd602a012266c0b160d07cdda"}]},"branch":"refs/heads/master"},"39bd8b656b7ed2d60b351af4c696a0826050795c":{"kind":"REWORK","_number":19,"created":"2020-04-02 08:36:47.000000000","uploader":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"ref":"refs/changes/47/709547/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/19"}}},"commit":{"parents":[{"commit":"869cac9def302aa88142901c1cd98ec4d88b5b65","subject":"Merge \"Fix serialization of structures that might contain YAQL types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/869cac9def302aa88142901c1cd98ec4d88b5b65"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"ali abdelal","email":"ali.abdelal@nokia.com","date":"2020-04-02 08:36:31.000000000","tz":0},"subject":"Added two new Rally scenarios","message":"Added two new Rally scenarios\n\nthe new scenarios are a workflow that contains Jinja,\nYaql expressions on a relatively big json object\n(4mb in size).\n\nthe input for these two Scenarios is :\n 1. the number of tasks in the workflow\n    - 30 in rally task\n 2. the size of the param object in mb\n    - 4mb in rally task\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/39bd8b656b7ed2d60b351af4c696a0826050795c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/39bd8b656b7ed2d60b351af4c696a0826050795c"}]},"branch":"refs/heads/master"},"75d01bfbd325b9fccdede30b57e3e9c738ae40e5":{"kind":"REWORK","_number":20,"created":"2020-04-02 10:08:00.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/47/709547/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/47/709547/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/47/709547/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/47/709547/20"}}},"commit":{"parents":[{"commit":"869cac9def302aa88142901c1cd98ec4d88b5b65","subject":"Merge \"Fix serialization of structures that might contain YAQL types\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/869cac9def302aa88142901c1cd98ec4d88b5b65"}]}],"author":{"name":"ali","email":"ali.abdelal@nokia.com","date":"2020-02-24 15:03:20.000000000","tz":0},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-04-02 10:07:55.000000000","tz":0},"subject":"Added two new Rally scenarios","message":"Added two new Rally scenarios\n\nthe new scenarios are a workflow that contains Jinja,\nYaql expressions on a relatively big json object\n(4mb in size).\n\nthe input for these two Scenarios is :\n 1. the number of tasks in the workflow\n    - 30 in rally task\n 2. the size of the param object in mb\n    - 4mb in rally task\n\nChange-Id: Ib33f26573b4b46a628f7d1d28fd18dedcb9c4738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/75d01bfbd325b9fccdede30b57e3e9c738ae40e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/75d01bfbd325b9fccdede30b57e3e9c738ae40e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
