)]}'
{"id":"openstack%2Fmistral~710170","triplet_id":"openstack%2Fmistral~stable%2Frocky~I4fa15994a7359a5f90a0a4671d47b19fe928cf33","project":"openstack/mistral","branch":"stable/rocky","topic":"bug/1837532-stable/rocky","hashtags":[],"change_id":"I4fa15994a7359a5f90a0a4671d47b19fe928cf33","subject":"Retry a DB transaction on \"Too many connections\" error","status":"MERGED","created":"2020-02-27 02:09:34.000000000","updated":"2020-03-07 12:48:11.000000000","submitted":"2020-03-07 12:48:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710170-1583585291930-9c168db8","meta_rev_id":"013bd0921d7a72a68630080e959a1998a30a46fb","_number":710170,"virtual_id_number":710170,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-07 12:48:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-03-05 16:00:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-06 04:04:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-07 10:08:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 12:28:27.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-02-27 13:51:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-03-05 16:00:04.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-07 12:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6bb2d9b0f9728f4f46fa03eb1c6534b666d6cc3f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-27 02:09:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7958dcff37710e97a716de22726aded76a00da4a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-27 02:09:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2175f33a2d1135eb46e981b120865639503eb5de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 02:58:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14b16c35176745ffb6b5f7d7015e7b66 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b67226ee74af4e2cb7281710d5b2a944 : SUCCESS in 9m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ea29305806ce450eb3eb8e7d413a6ba5 : SUCCESS in 8m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/aa092e1f113a498bb8cea8be2e58778e : FAILURE in 6m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9b3b4222ef464bc28467439a23962346 : FAILURE in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1aae776816c46cba2a6ae62eb0424c2 : SUCCESS in 10m 06s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/a9b954197eab45cfb4d1a2f8b591ee7e : FAILURE in 31m 54s (non-voting)\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/47d773e3be6845eeba347c72aedfa3bd : FAILURE in 32m 59s (non-voting)\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/441021741ca54cb8b52406e215ba893f : FAILURE in 34m 17s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/2042a7eae25746b9bf3d98bed95d6bbf : SUCCESS in 31m 30s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/076de1244aef464693edeec8f17b7d37 : SUCCESS in 26m 52s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b2fce9c480fc409b8076d2dc397d5691 : FAILURE in 30m 58s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a6b9ad4cd5e54996b23b653cb4a60951 : FAILURE in 32m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf23aee045724ea387c99d41471192e9 : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"100f0c7255b5256ef25a4ecc73bcae025734b9a9","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-02-27 06:53:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6f3aa63089b070637e20014cc326c75ce4944aac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 08:19:59.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d56687b634cf4420974ad15e5f659f1f : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e13a36f35d294d4b82b36bbfebf130c8 : SUCCESS in 9m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/815bcfcadf6f419db4687d6e0b6a3f20 : SUCCESS in 9m 28s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b7a5b271e4e946d3a8c15d7db6bb8f90 : FAILURE in 5m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a46b5c9ac9a4779a3ea14ae6b4f6a72 : FAILURE in 4m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80f41dfe62a848c4a9de032542cf130d : SUCCESS in 11m 04s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e21f595fd1c54a4c9ad9434af775942a : FAILURE in 34m 23s (non-voting)\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ec41356326794c6295b02f442125f083 : FAILURE in 43m 15s (non-voting)\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8753e35570964b669292f62d52be63a4 : FAILURE in 37m 43s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1bdee3194fbb4984bd9c6549c68aaefa : SUCCESS in 27m 18s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/21ccf512ccc240a4ac384742a7f0c715 : SUCCESS in 26m 05s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/47c9c88905f34fda946eee31666b2895 : FAILURE in 35m 12s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a03b5eb856474a8e9eeb3d244d90cbe0 : FAILURE in 32m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4602275f0fba4494942619b766ec8a52 : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d47f1970cb4926eaed022a42fc513d35d889f495","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 12:25:41.000000000","message":"Patch Set 1:\n\nIs the tenacity bump actually needed?","accounts_in_message":[],"_revision_number":1},{"id":"e079aa3338d587039cfcdc9519354c8c040c175a","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-27 12:28:27.000000000","message":"Patch Set 1:\n\n\u003e Is the tenacity bump actually needed?\n\nyes the method \u0027retry_if_exception_message\u0027 exists only in a later version of tenacity","accounts_in_message":[],"_revision_number":1},{"id":"7a7ee1409e6251d52b19f51672289651440a878f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 12:45:36.000000000","message":"Patch Set 1:\n\n\u003e \u003e Is the tenacity bump actually needed?\n \u003e \n \u003e yes the method \u0027retry_if_exception_message\u0027 exists only in a later\n \u003e version of tenacity\n\nI don\u0027t think this can be backported then as it will break stable policy for raising that requirement.","accounts_in_message":[],"_revision_number":1},{"id":"3678ed6fc52d5ef32065fdd12d326e65d159ef22","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-04 20:47:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6325cea3e3385709272d3ed2879029ce1f9406f","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-05 11:16:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3a983b393f109c8c6daae46a5615c203782ffd58","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-05 12:45:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f024af73dd3c2122b15bd32a59a55d1022b0b1b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 12:45:29.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/mistral on branch stable/rocky.  The error was:\n\n  extra keys not allowed @ data[\u0027check\u0027][\u0027jobs\u0027][3][\u0027mistral-devstack\u0027][\u0027devstack_localrc\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python-jobs\n        - openstack-python35-jobs\n        - publish-openstack-sphinx-docs\n        - check-requirements\n        - release-notes-jobs\n      check:\n        jobs:\n          - openstack-tox-cover:\n              voting: false\n              irrelevant-files:\n                - ^.zuul.yaml$\n              required-projects:\n                - openstack/mistral-lib\n          - openstack-tox-py35:\n              required-projects:\n                - openstack/mistral-lib\n          - openstack-tox-py27:\n              required-projects:\n                - openstack/mistral-lib\n          - mistral-devstack:\n              voting: false\n              devstack_localrc:\n                USE_PYTHON3: true\n          - mistral-devstack-non-apache:\n              voting: false\n              branches: ^(?!stable/(newton|ocata)).*$\n              devstack_localrc:\n                USE_PYTHON3: true\n          - mistral-devstack-kombu:\n              voting: false\n              branches: ^(?!stable/newton).*$\n              devstack_localrc:\n                USE_PYTHON3: true\n          - mistral-tox-unit-mysql\n          - mistral-tox-unit-postgresql\n          # TripleO jobs that deploy Mistral.\n          # Note we don\u0027t use a project-template here, so it\u0027s easier\n          # to disable voting on one specific job if things go wrong.\n          # tripleo-ci-centos-7-scenario003-multinode-oooq-container will\n          # run in Pike and beyond.\n          # If you need any support to debug these jobs in case of\n          # failures, please reach us on #tripleo IRC channel.\n          - tripleo-ci-centos-7-scenario003-multinode-oooq-container\n          - mistral-rally-task:\n              voting: false\n          - openstack-tox-lower-constraints:\n              required-projects:\n                - openstack/mistral-lib\n      gate:\n        queue: mistral\n        jobs:\n          - mistral-tox-unit-mysql\n          - mistral-tox-unit-postgresql\n          - tripleo-ci-centos-7-scenario003-multinode-oooq-container\n          - openstack-tox-lower-constraints\n      post:\n        jobs:\n          - mistral-docker-buildimage:\n              branches: master\n      experimental:\n        jobs:\n          - mistral-docker-buildimage-test:\n              branches: master\n          # This job doesn\u0027t work yet.\n          # - mistral-ha:\n          #     voting: false\n\n  in \"openstack/mistral/.zuul.yaml@stable/rocky\", line 83, column 3","accounts_in_message":[],"_revision_number":3},{"id":"9232f19b0beeadfde8ca38a033fca5a4d2e6e5ca","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-05 12:58:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e13fa9f028ea74a1f27ef4320d9d77015a0a7d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 12:59:11.000000000","message":"Patch Set 4: Verified-1\n\n(4 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/mistral on branch stable/rocky.  The error was:\n\n  extra keys not allowed @ data[\u0027devstack_localrc\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: mistral-devstack-py3\n      parent: mistral-devstack\n      devstack_localrc:\n        USE_PYTHON3: true\n\n  in \"openstack/mistral/.zuul.yaml@stable/rocky\", line 73, column 3","accounts_in_message":[],"_revision_number":4},{"id":"c491dae403b6d5edcbb9471bbc54da4d1184a4d8","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-05 13:07:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9a6f0abd332668f0e930acc5ea2f57d6f32cbe81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 15:57:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4edd9f721dbf406b9d557e9809cbd2bb : SUCCESS in 6m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8031beba88a45da838295ef350fb241 : SUCCESS in 9m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ef1b0ad5becb4036bc70cb8751735c02 : SUCCESS in 9m 13s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/af4cbfafd3b14a0b9882d62224b31101 : SUCCESS in 6m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb3247c668194240be1a8ae880d0498a : SUCCESS in 9m 49s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c32133e9f44d4f809567c21ffac2b397 : SUCCESS in 47m 27s (non-voting)\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7aa73bf58fbb47f4ac6228266459ec98 : SUCCESS in 48m 12s (non-voting)\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/ce5974645242425fa0f2b6e1f96e240f : SUCCESS in 44m 32s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d65224de730d47f8b5e59211763055e2 : SUCCESS in 26m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/598e25b962374ebfa6879e684175d94d : SUCCESS in 26m 19s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8c511fb741894ea08cf459bd3a42027f : SUCCESS in 2h 42m 00s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d162f252d27e4ebb9fc06c83ab6a4813 : SUCCESS in 1h 19m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/adb604cda0654754b7deb069b675255b : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":5},{"id":"f804380e703be528bbf3bfff10acd1031e4c771e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-05 16:00:04.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nLooks good. Going to assume zuul will be happy with this.","accounts_in_message":[],"_revision_number":5},{"id":"6611221423aa6c3544c4e5cbbd6027509bb8a3ee","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-06 04:04:49.000000000","message":"Patch Set 5: Code-Review+2\n\nLooks ok to me. It\u0027s a bit awkward to copy past this tenacity code but I believe it\u0027s a lesser evil in this case.\n\nEyal, are you planning to do anything else with this patch? Insert a cherry-pick info etc.?","accounts_in_message":[],"_revision_number":5},{"id":"1f5560e922258e4a188985e536cd5d988f86b33f","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-07 10:08:43.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0a649ae41e0e7a5593f3fe0f2f55aa6dbc6f7337","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 10:08:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1de37898b4313e402488983e5a4e04a69b3a6ce3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 12:48:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"3ccc3d88d7d0411ad5e1c3f6c2b42beb76b2b30b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 12:48:11.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5300af0595d3484ea896b7b5957f3a6e : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a78cd81e5984c2a984ee0dd940dc569 : SUCCESS in 8m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e92fdf45074f472a8dd833a41cef7dbf : SUCCESS in 8m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/39096737fb0f4cee80939b779bd4646a : SUCCESS in 4m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/d9010cd3bf844fe9a8fd6f1abe4f74a6 : SUCCESS in 26m 44s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e92dd91fb1274708af507c208db5883c : SUCCESS in 25m 41s\n- tripleo-ci-centos-7-scenario003-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9127631ddb524828ae155b67ca8611e1 : SUCCESS in 2h 37m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ef770cc80664d768bf106913b49a2b9 : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4ad66f140bc7cce9c1c7805a22d9d3c5600642d6","revisions":{"f8fb0922c928ff68d3cc004706e3c930d58dbffb":{"kind":"REWORK","_number":1,"created":"2020-02-27 02:09:34.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/70/710170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/710170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/710170/1"}}},"commit":{"parents":[{"commit":"4d3ef8887fb765c9edd36f62cc66904aff4c1b0e","subject":"Mask sensitive data when logging action results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d3ef8887fb765c9edd36f62cc66904aff4c1b0e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 09:59:57.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-02-27 02:09:34.000000000","tz":0},"subject":"Retry a DB transaction on \"Too many connections\" error","message":"Retry a DB transaction on \"Too many connections\" error\n\n* Writing a unit test is very problematic but the fix has been\n  tested manually.\n\nCloses-Bug: #1837532\n\nChange-Id: I4fa15994a7359a5f90a0a4671d47b19fe928cf33\n(cherry picked from commit 4a9d55a1b041007fc1b2340c5a143ffa2eb3f31d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f8fb0922c928ff68d3cc004706e3c930d58dbffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f8fb0922c928ff68d3cc004706e3c930d58dbffb"}]},"branch":"refs/heads/stable/rocky"},"774b6582d1ff27f990ba0b9efab3fbb6f822f150":{"kind":"REWORK","_number":2,"created":"2020-03-05 11:16:26.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/70/710170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/710170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/710170/2"}}},"commit":{"parents":[{"commit":"4d3ef8887fb765c9edd36f62cc66904aff4c1b0e","subject":"Mask sensitive data when logging action results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d3ef8887fb765c9edd36f62cc66904aff4c1b0e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 09:59:57.000000000","tz":420},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-03-05 11:14:47.000000000","tz":120},"subject":"Retry a DB transaction on \"Too many connections\" error","message":"Retry a DB transaction on \"Too many connections\" error\n\n* Writing a unit test is very problematic but the fix has been\n  tested manually.\n\nCloses-Bug: #1837532\n\nChange-Id: I4fa15994a7359a5f90a0a4671d47b19fe928cf33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/774b6582d1ff27f990ba0b9efab3fbb6f822f150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/774b6582d1ff27f990ba0b9efab3fbb6f822f150"}]},"branch":"refs/heads/stable/rocky"},"c2639dad0480c1a5444d899f6f1c7df69cb44285":{"kind":"REWORK","_number":3,"created":"2020-03-05 12:45:13.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/70/710170/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/710170/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/710170/3"}}},"commit":{"parents":[{"commit":"4d3ef8887fb765c9edd36f62cc66904aff4c1b0e","subject":"Mask sensitive data when logging action results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d3ef8887fb765c9edd36f62cc66904aff4c1b0e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 09:59:57.000000000","tz":420},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-03-05 12:44:49.000000000","tz":120},"subject":"Retry a DB transaction on \"Too many connections\" error","message":"Retry a DB transaction on \"Too many connections\" error\n\n* Writing a unit test is very problematic but the fix has been\n  tested manually.\n\nCloses-Bug: #1837532\n\nChange-Id: I4fa15994a7359a5f90a0a4671d47b19fe928cf33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c2639dad0480c1a5444d899f6f1c7df69cb44285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c2639dad0480c1a5444d899f6f1c7df69cb44285"}]},"branch":"refs/heads/stable/rocky"},"a99ffddd2d8a0b6406ab995fb9f69c94dc45d93a":{"kind":"REWORK","_number":4,"created":"2020-03-05 12:58:56.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/70/710170/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/710170/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/710170/4"}}},"commit":{"parents":[{"commit":"4d3ef8887fb765c9edd36f62cc66904aff4c1b0e","subject":"Mask sensitive data when logging action results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d3ef8887fb765c9edd36f62cc66904aff4c1b0e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 09:59:57.000000000","tz":420},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-03-05 12:58:35.000000000","tz":120},"subject":"Retry a DB transaction on \"Too many connections\" error","message":"Retry a DB transaction on \"Too many connections\" error\n\n* Writing a unit test is very problematic but the fix has been\n  tested manually.\n\nCloses-Bug: #1837532\n\nChange-Id: I4fa15994a7359a5f90a0a4671d47b19fe928cf33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a99ffddd2d8a0b6406ab995fb9f69c94dc45d93a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a99ffddd2d8a0b6406ab995fb9f69c94dc45d93a"}]},"branch":"refs/heads/stable/rocky"},"4ad66f140bc7cce9c1c7805a22d9d3c5600642d6":{"kind":"REWORK","_number":5,"created":"2020-03-05 13:07:14.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/70/710170/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/70/710170/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/70/710170/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/70/710170/5"}}},"commit":{"parents":[{"commit":"4d3ef8887fb765c9edd36f62cc66904aff4c1b0e","subject":"Mask sensitive data when logging action results","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4d3ef8887fb765c9edd36f62cc66904aff4c1b0e"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-07-23 09:59:57.000000000","tz":420},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-03-05 13:06:54.000000000","tz":120},"subject":"Retry a DB transaction on \"Too many connections\" error","message":"Retry a DB transaction on \"Too many connections\" error\n\n* Writing a unit test is very problematic but the fix has been\n  tested manually.\n\nCloses-Bug: #1837532\n\nChange-Id: I4fa15994a7359a5f90a0a4671d47b19fe928cf33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4ad66f140bc7cce9c1c7805a22d9d3c5600642d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4ad66f140bc7cce9c1c7805a22d9d3c5600642d6"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
