)]}'
{"id":"openstack%2Fmistral~710903","triplet_id":"openstack%2Fmistral~master~I9cd58684714066e229d6774148119aea03d22efb","project":"openstack/mistral","branch":"master","topic":"(HEAD","hashtags":[],"change_id":"I9cd58684714066e229d6774148119aea03d22efb","subject":"Remove TripleO gate job","status":"MERGED","created":"2020-03-03 05:10:15.000000000","updated":"2020-08-26 07:20:38.000000000","submitted":"2020-08-26 07:15:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710903-1598426109291-210ac784","meta_rev_id":"c55f0c3c542bbfae419e8815768d4475829d805a","_number":710903,"virtual_id_number":710903,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 07:15:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2020-08-26 05:27:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2020-08-26 05:27:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 06:57:56.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2020-03-03 07:00:26.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-08-25 12:49:50.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-08-26 07:15:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5200804fa4b95a14af57680534c220e7cee48df","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-03 05:10:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b60a6d429436d7ff6b46a011fa7a9966d4aa707c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-03 06:58:44.000000000","message":"Patch Set 1:\n\nDougal, Adriano, do you think it\u0027s too early to merge this patch or we can proceed?","accounts_in_message":[],"_revision_number":1},{"id":"e5a64df80690b15e92bd2a1e0676b52f467468e3","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-03 07:00:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"684fde02534168bb794c91be83f93a236fc9d516","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 08:29:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de0a55a08c3468ca1f0e2731ffac29f : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b6b349a90c546bf9792c24098b8677f : SUCCESS in 12m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c455b60866f43f59ff713e10c5ba3c6 : SUCCESS in 12m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f3903d799374586995fcbb761a5e692 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddf01a8c951042f98b03192565831959 : SUCCESS in 9m 12s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/3f1c742931a242fb9fe5d106c9185249 : SUCCESS in 41m 12s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7687f8c5745746929b1b7948dfcb2a0c : SUCCESS in 37m 45s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8af294f636d48b6a7e4b4159fa2d469 : SUCCESS in 44m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c9430615ea534c268a7588ecf2bbd4c3 : SUCCESS in 45m 17s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/99ac427f438e47fda99145f69db030b9 : SUCCESS in 44m 07s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/aaaa9bf8deaf4ed5920a2a39af0f360a : SUCCESS in 30m 09s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/01863cd4a4114f69a66f8229d5daaeec : SUCCESS in 30m 02s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f0c845e9490e403e9d880413483afb80 : SUCCESS in 3h 14m 16s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/4f638a56aa0743a88887b3fba209f68a : SUCCESS in 1h 02m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df4f62eb6b6c43c48928b6104f27e5f1 : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a6d1591e55c9eaf6fe957feda3fe78a7e86f4c24","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-03 08:32:46.000000000","message":"Patch Set 1: Workflow-1\n\nI\u0027ll make it WIP for now. I want to make sure it\u0027s safe to merge first.","accounts_in_message":[],"_revision_number":1},{"id":"e2d4a6547668bdc3fa987f4e3a31a722f8df8e43","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-03-03 09:02:24.000000000","message":"Patch Set 1: Code-Review-1\n\nA little to early, still.","accounts_in_message":[],"_revision_number":1},{"id":"70027e242e514652983f3b73165dec542b77ce3d","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-03 09:04:18.000000000","message":"Patch Set 1:\n\n\u003e A little to early, still.\n\nThe problem is that https://review.opendev.org/#/c/703972/\ncannot pass becuase of mapping.json was moved to mistral-extra","accounts_in_message":[],"_revision_number":1},{"id":"d31fbdafeb2d0f5c2c55c3d38fd48f6fb58e8d46","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-04 07:47:33.000000000","message":"Patch Set 1:\n\nyes","accounts_in_message":[],"_revision_number":1},{"id":"a372817494e5c6c4e717f07726360b8468446578","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-08-25 04:45:45.000000000","message":"Patch Set 1:\n\nAdriano, do you think it is time to land this patch?","accounts_in_message":[],"_revision_number":1},{"id":"d666e74162be7e831d5c21c337faa00e57bf978b","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-08-25 12:49:50.000000000","message":"Patch Set 1: Code-Review+2\n\nyeah we can merge this now.","accounts_in_message":[],"_revision_number":1},{"id":"fdbdda9937c37c13c8d5e3dbaf681c64c43e7c88","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-08-26 04:22:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"10f88f203fe27cf2214c433b82b52b5af9a53a48","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-08-26 04:30:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fc4a66467213747c6d90446b06e46954129b35cf","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-08-26 05:27:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4bf5566f025a69909f83d1eaeea7b9f223722d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 06:16:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9eefceca8c489aaea8ad039794c9c1 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa364cf03dd643b4a5f3976d6ce524ff : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee068ef838e54cf4830aacde4d2ac24b : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67362019f887432f8965b5402f1705a7 : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c42978589394963a27599ccea8b24a4 : SUCCESS in 8m 21s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/d19ce359b2274bc18bc9618e545ef7c4 : SUCCESS in 55m 57s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e47827fa5c34ad28da0dd4b854c90df : SUCCESS in 45m 48s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59b85142aaef4e4587d48ae861a6a5f4 : SUCCESS in 42m 22s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/dfbf3a019fc8466689e612b1a729fb12 : SUCCESS in 29m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/25671dcae8a849579c35c6df4d7bb38f : SUCCESS in 48m 11s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/fbf2acd856504084ab77406c58fa1207 : SUCCESS in 28m 55s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/babdc0fc36094e6d9b8879338042c76a : SUCCESS in 26m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/395b04255690418eab581dbc32feb779 : SUCCESS in 1h 44m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"93ca5024a8be244484498566ba828489b0c2f9ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 06:16:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f708c5d716fe600fcd31fb1aa355603e6b7e1ea1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 07:15:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60fa66d295554f0ca7cc5ed4fd71216f : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e6a39f9af8848e7b8e024e1330c08bd : SUCCESS in 10m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0c50de264494203912246de5b4b7454 : SUCCESS in 9m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2569e5f2e40a428f9ac263e1111124cf : SUCCESS in 8m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ea9d3823370425fb40242a22ed391b4 : SUCCESS in 10m 45s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/73bcb7b7e51f4abda2079b0e7fbed62e : SUCCESS in 45m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb9ed57ef201406db235b11ed5010946 : SUCCESS in 47m 57s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a1c4f4ac4be4ab7adfe49d85d9976c3 : SUCCESS in 52m 32s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/5256105ea92e412e96fa9a58c91101cc : SUCCESS in 49m 21s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/4e1fa74b3f3d43a98a607691a6caed7b : SUCCESS in 49m 31s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/25b6c3bbd5a04aeca587a78054622387 : SUCCESS in 30m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b0e6a7c07ff14fb095a2144f7c444670 : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":3},{"id":"33059e07f814f1a343758c0ec57cd14e1b719d1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 07:15:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0ca8128c5e893c39d1ed0fc05e75f30c20698a50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 07:20:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/991d99787ade47ae8adaeafcda59519c : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"63225d41e41dbedc341f2ef0724901f62dcd5d8b","revisions":{"580548d208d4aab57c9b2e14c7b876c3d58578da":{"kind":"REWORK","_number":1,"created":"2020-03-03 05:10:15.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/03/710903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/710903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/710903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/710903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/710903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/710903/1"}}},"commit":{"parents":[{"commit":"d21fbc963642894ba91aa6be4608e4778d654751","subject":"Merge \"Removing duplicated descriptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d21fbc963642894ba91aa6be4608e4778d654751"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-03 05:02:14.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-03 05:02:14.000000000","tz":420},"subject":"Remove TripleO gate job","message":"Remove TripleO gate job\n\n* TripleO is dropping Mistral now and its jobs keep failing too\n  often because of various internal activities like moving from\n  Py2 to Py3. This patch needs to be merged once we are confident\n  that it will be safe.\n\nChange-Id: I9cd58684714066e229d6774148119aea03d22efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/580548d208d4aab57c9b2e14c7b876c3d58578da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/580548d208d4aab57c9b2e14c7b876c3d58578da"}]},"branch":"refs/heads/master"},"566b563abe28aa511cc6f17a5de15ca5ded2ed3d":{"kind":"REWORK","_number":2,"created":"2020-08-26 04:22:59.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/03/710903/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/710903/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/710903/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/710903/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/710903/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/710903/2"}}},"commit":{"parents":[{"commit":"4733e084aa22508a115e1ea1333258fa076006b2","subject":"Change python-nose package name in devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4733e084aa22508a115e1ea1333258fa076006b2"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-03 05:02:14.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-08-26 04:22:38.000000000","tz":420},"subject":"Remove TripleO gate job","message":"Remove TripleO gate job\n\n* TripleO is dropping Mistral now and its jobs keep failing too\n  often because of various internal activities like moving from\n  Py2 to Py3. This patch needs to be merged once we are confident\n  that it will be safe.\n\nChange-Id: I9cd58684714066e229d6774148119aea03d22efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/566b563abe28aa511cc6f17a5de15ca5ded2ed3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/566b563abe28aa511cc6f17a5de15ca5ded2ed3d"}]},"branch":"refs/heads/master"},"63225d41e41dbedc341f2ef0724901f62dcd5d8b":{"kind":"REWORK","_number":3,"created":"2020-08-26 04:30:02.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/03/710903/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/03/710903/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/710903/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/710903/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/03/710903/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/03/710903/3"}}},"commit":{"parents":[{"commit":"4733e084aa22508a115e1ea1333258fa076006b2","subject":"Change python-nose package name in devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4733e084aa22508a115e1ea1333258fa076006b2"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-03 05:02:14.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-08-26 04:29:48.000000000","tz":420},"subject":"Remove TripleO gate job","message":"Remove TripleO gate job\n\n* TripleO is dropping Mistral now and its jobs keep failing too\n  often because of various internal activities like moving from\n  Py2 to Py3. This patch needs to be merged once we are confident\n  that it will be safe.\n\nChange-Id: I9cd58684714066e229d6774148119aea03d22efb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/63225d41e41dbedc341f2ef0724901f62dcd5d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/63225d41e41dbedc341f2ef0724901f62dcd5d8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
