)]}'
{"id":"openstack%2Fmistral~711172","triplet_id":"openstack%2Fmistral~master~I441310539f7949aea569d85ed998deea4fa0d0b0","project":"openstack/mistral","branch":"master","topic":"bug/1864823","hashtags":[],"change_id":"I441310539f7949aea569d85ed998deea4fa0d0b0","subject":"Add a unit test for @tx_cached and fix a bug in it","status":"MERGED","created":"2020-03-04 07:31:45.000000000","updated":"2020-03-04 12:04:09.000000000","submitted":"2020-03-04 12:02:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711172-1583323341486-30632db2","meta_rev_id":"9d4c3b71dbc35f877a488ec68f2a196da735de5c","_number":711172,"virtual_id_number":711172,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 12:02:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-03-04 08:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":2,"date":"2020-03-04 08:10:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-03-04 07:44:02.000000000","_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},{"value":1,"date":"2020-03-04 08:11:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"date":"2020-03-04 07:49:12.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-04 07:43:49.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 07:43:49.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-03-04 07:44:02.000000000","updated_by":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"reviewer":{"_account_id":30755,"name":"ali","email":"ali.abdelal@nokia.com","username":"abdelal"},"state":"REVIEWER"},{"updated":"2020-03-04 07:49:12.000000000","updated_by":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2020-03-04 08:02:11.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-03-04 08:11:29.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-03-04 12:02:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea35e0819a6662c2ca7b75a35502e8c324ec3ea2","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-04 07:31:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4db1f20b52219302fb30de2999d12bb29ad3f70c","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-03-04 08:02:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b8b35fb308c3f514ee3c473bb902e432137d9011","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-03-04 08:10:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"339b2aae06f3976bd358b2cd9bf7075e91f3ac5f","author":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"date":"2020-03-04 08:11:29.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c38e698c281bd1f44642c368123bdf9431510de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 10:07:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb00ec64726640e2be90e07d75d76fe5 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eadf43e12f1a4010ade59528fed8294d : SUCCESS in 9m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a83088acc6fb4f9c8e96dfbb692bd981 : SUCCESS in 9m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e447e145ab3e480da1ff779783edcbfb : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e269b2c20195451ab88f8d8166c3b742 : SUCCESS in 8m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42d9b0173ab24aef8f1ab3b880bda6ee : SUCCESS in 10m 56s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/30372a008678482cb4ad22741be8b97f : SUCCESS in 45m 09s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/255fab42edca4aa58ba6a255bdd87b09 : SUCCESS in 41m 23s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bbb8750034441b9a9badd46513bac6f : SUCCESS in 43m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e9c93e78312c4a639da9056facc68576 : SUCCESS in 44m 41s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5b6ebf896eec4f02819565b10cb7bb95 : SUCCESS in 41m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/91aa14c824774e6a978bc6c2519f16ab : SUCCESS in 30m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/3cd6c7f432f84000b5eceef869b5597d : SUCCESS in 29m 13s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4fe9d7e8b5c64bb6a3401d8359ae4829 : SUCCESS in 2h 34m 13s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/c7aa61991f4148e08cc9234861b28f79 : SUCCESS in 55m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8aefb0a61a014b64a8d0341744125b6d : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1c431b6163b59e73aeaf9f4715a547ca9890d6f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 10:07:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9bf84f2480a5a4c5b1639f96e5a3e32dc1eca393","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:02:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"16ca7dc4115fc8b98b409bce783f9e44dde1039b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:02:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c5e30ab68cc442c92afe95d255f0005 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6fff9178a6a4eac9f4477e981b988b8 : SUCCESS in 10m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de41a5ff953e41888da0e5b2f0daed86 : SUCCESS in 9m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53ce6e953ca435f921d0c77803c7f66 : SUCCESS in 9m 40s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/b39faffe9937498dbe2d013cf8058444 : SUCCESS in 43m 07s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac0d5d5e5c1c4d43a2097b06b1944b79 : SUCCESS in 44m 43s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c6d1e4107634a698e2bd343b02ad2cc : SUCCESS in 41m 25s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/d7f3191237ef4009bfe4033d1fd90674 : SUCCESS in 45m 45s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/009c3803407d49539d91f3a917fb4304 : SUCCESS in 31m 39s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/c4caddd86bbd4decbb7b2ba4e49b6ba2 : SUCCESS in 30m 46s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/b02bd51e590543e09b4c1a450a2b2a92 : SUCCESS in 49m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/355db7f7fd6f4350a4e4e417ed4d4f93 : SUCCESS in 13m 21s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/971aed52dd584663ad07cd6f7697107d : SUCCESS in 1h 49m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2171137172c4a153dc8f7f05a75d92314f574a40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:04:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e8af582ea84b5389f8385280d46738 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77b450d78553d131c70b8c94a269b250d39fc956","revisions":{"77b450d78553d131c70b8c94a269b250d39fc956":{"kind":"REWORK","_number":1,"created":"2020-03-04 07:31:45.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/72/711172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/72/711172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/711172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/711172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/72/711172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/72/711172/1"}}},"commit":{"parents":[{"commit":"1217e4ca61c03e2f86a4ff1ea75f516ae2114a22","subject":"Merge \"Add caching for YAQL/Jinja expression functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1217e4ca61c03e2f86a4ff1ea75f516ae2114a22"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-04 07:28:49.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-04 07:31:15.000000000","tz":420},"subject":"Add a unit test for @tx_cached and fix a bug in it","message":"Add a unit test for @tx_cached and fix a bug in it\n\n* Added a unit tests for the decorator @tx_cached. The tests\n  helped find a bug in the implementation. One of the conditions\n  in the decorator was wrong.\n\nCloses-Bug: #1864823\nChange-Id: I441310539f7949aea569d85ed998deea4fa0d0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/77b450d78553d131c70b8c94a269b250d39fc956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/77b450d78553d131c70b8c94a269b250d39fc956"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
