)]}'
{"id":"openstack%2Fmistral~712251","triplet_id":"openstack%2Fmistral~stable%2Ftrain~If9f0c9bea587b486998af1c18e282bedba453499","project":"openstack/mistral","branch":"stable/train","topic":"bug/1862161-stable/train","hashtags":[],"change_id":"If9f0c9bea587b486998af1c18e282bedba453499","subject":"Fix incorrect in-depth search of affected tasks","status":"MERGED","created":"2020-03-11 08:14:44.000000000","updated":"2020-03-11 13:21:56.000000000","submitted":"2020-03-11 13:19:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712251-1583932745994-30632db2","meta_rev_id":"565152e22ae72a82983c80d0dda63208e6a90509","_number":712251,"virtual_id_number":712251,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-11 13:19:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":2,"date":"2020-03-11 08:14:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"date":"2020-03-11 08:14:44.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2020-03-11 08:14:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-11 08:14:44.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-03-11 13:19:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67d90eda99a717d7ef39356ff8e63cec579365b8","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-11 08:14:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"164599812c49c828fafc85cd7e8e85acdd3d1f36","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-11 08:14:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e3a9c996ef095bcc365d32ab6957b5ded8291cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 11:18:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b75cf1edbe6c49e2a8bec60f879e38f4 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5240a21aeb6148beacb88aef4e44852e : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd3e4e0bca9b47a2be4e5331e4991624 : SUCCESS in 9m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8a6332f128a4fd6b49922da810ee57d : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c21aae2d7b73409ab0504d2181b18a17 : SUCCESS in 6m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e234446b65c4425bdb5ccf2709e2978 : SUCCESS in 10m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/492778f3aaf143a3badab50f20681b9d : SUCCESS in 45m 36s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/37584464ff37490bbbb67ac3ed899629 : SUCCESS in 47m 09s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dc338a78260c4971ae3e1d42a2636642 : SUCCESS in 49m 06s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/7ce37f6481964df98797e1d143c8a3db : SUCCESS in 31m 50s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/5fa1a0534ba1457ab76bf471aa9270fc : SUCCESS in 30m 21s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b73df6b78e4f424f94fb29e11b80ce45 : SUCCESS in 1h 41m 09s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3cfb773d835b4f4eb35ee3a5cb0fc189 : SUCCESS in 2h 59m 09s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/77ad787caed5466cbaeacf5820febdee : SUCCESS in 57m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51e259c395e34af69e1283507d837459 : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0c3c11a1ff4c0fe84f0a3997f2f882fdd050445e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 11:18:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a204f651e0517235be35f216b573106dcd1a126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 13:19:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea29d2b06df6418693d1fc8c17afacc1 : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ed21ba543ec442cb29e60ff89a3919f : SUCCESS in 10m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/507d4296840943f7b728363932952386 : SUCCESS in 8m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4144411e29a44dfc9f7e494c4f089aba : SUCCESS in 11m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96c56ebd1ec4095aad7319bf5b63809 : SUCCESS in 9m 11s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bc4241d5abca49a3965d1d7c9ce87d5f : SUCCESS in 41m 03s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/2ad3631b0877478d9986c7173916396b : SUCCESS in 40m 19s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f041956866cc428aad7d6e797a76fd44 : SUCCESS in 32m 29s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7b7b0a3cc8384af6bb508a9347043a9d : SUCCESS in 29m 07s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/35d293ab2f71430ebde1a5108c2b8512 : SUCCESS in 1h 20m 05s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1d2f8a8e9bcc48ae8a6e368e8c7440d4 : SUCCESS in 47m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ebb31bc338f49c9ba58dbf695659f0d : SUCCESS in 10m 34s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d62172f932f842328e4b9201b1db077b : SUCCESS in 1h 55m 45s","accounts_in_message":[],"_revision_number":1},{"id":"fdc2d0d0e1ed63b44d5241ee474611d69a31dd10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 13:19:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"331696ab888e49ad0f36c2f5fb16be534c3b32fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 13:21:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a153767e14d64449a1c3716fbbc64821 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c99c345fb59edd16a461d18674809b6f43db7c53","revisions":{"c99c345fb59edd16a461d18674809b6f43db7c53":{"kind":"REWORK","_number":1,"created":"2020-03-11 08:14:44.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/51/712251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/51/712251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/712251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/712251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/51/712251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/51/712251/1"}}},"commit":{"parents":[{"commit":"3df8162946144cda694ec6a750812a58e0ccfdf0","subject":"Fix YAQL engine initialization","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3df8162946144cda694ec6a750812a58e0ccfdf0"}]}],"author":{"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","date":"2020-02-06 11:43:00.000000000","tz":180},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-11 08:14:44.000000000","tz":0},"subject":"Fix incorrect in-depth search of affected tasks","message":"Fix incorrect in-depth search of affected tasks\n\nIn case if the join task is not exist (every upstream task have\nERROR state) workflow will stuck in RUNNING state because of\nthe bug in affected tasks search. This patch fixes this bug.\n\nChange-Id: If9f0c9bea587b486998af1c18e282bedba453499\nCloses-Bug: #1862161\nSigned-off-by: Oleg Ovcharuk \u003cvgvoleg@gmail.com\u003e\n(cherry picked from commit de633d5d48f48258bac214089391e4cda5b51f34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c99c345fb59edd16a461d18674809b6f43db7c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/c99c345fb59edd16a461d18674809b6f43db7c53"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
