)]}'
{"id":"openstack%2Fmistral~712934","triplet_id":"openstack%2Fmistral~master~I5399cee2b95ec66574507f08b8baa26057f4bde9","project":"openstack/mistral","branch":"master","hashtags":[],"change_id":"I5399cee2b95ec66574507f08b8baa26057f4bde9","subject":"[DNM] Testing OOO Mistral tests.","status":"ABANDONED","created":"2020-03-13 12:23:54.000000000","updated":"2020-04-17 05:50:58.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b53a4c1537a45bb9032959ef266e92fcb3d601f5","_number":712934,"virtual_id_number":712934,"owner":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-16 12:37:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 12:25:07.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-03-16 11:02:20.000000000","updated_by":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"reviewer":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"state":"REVIEWER"},{"updated":"2020-03-16 12:37:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d6c13c245c97d4bcc7178fe11284582049ace33","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2020-03-13 12:23:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ba9ea73dccfd20c248291ae87360922e7fbc9d5","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2020-03-13 12:25:49.000000000","message":"Patch Set 1:\n\n[DNM]","accounts_in_message":[],"_revision_number":1},{"id":"e5cf36250a73434109a239aed9ee7607aa1fa5e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 13:30:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8618c89db21a46a68c7c095cc6d9d57f : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32cf0ee06a50484fb7c0b2371a8ac767 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/daf81a7f9ead47a89ee973887e38547a : SUCCESS in 9m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d5b0eecf87e4da9ae6a81ea68409e98 : SUCCESS in 9m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/897f495c36a04e27a86a4c0bfeab9b06 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39613301d8af4620bbe0acbfb12ed0f1 : SUCCESS in 7m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39143acc0ee94a1fad20c569f6362fab : SUCCESS in 10m 08s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/18b2dd83535b4346ae71695ca63380bf : SUCCESS in 41m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cbc8cde8b2e4d5f90e4177e98fc499a : SUCCESS in 52m 04s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a30e19c888eb4f7f8ab6880c2e618845 : SUCCESS in 57m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/9253041713ae4724a98ca1677a4a44ab : SUCCESS in 44m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/aa2b67b7a19d414983a8021eb565d2e5 : SUCCESS in 42m 43s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/23b3db2cdc7b4fe092acb23042905689 : SUCCESS in 29m 14s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8fea27c01a1943dcaaa96f96ce380282 : SUCCESS in 28m 08s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/ec4892162fe44f8c8430a859c682481d : SUCCESS in 1h 00m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/333aa178fef74aa79a436d0690aba365 : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":1},{"id":"39543cd503ea3b068e508cca51fc1ec5d8f968f6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-03-13 13:44:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"719935cdd0023ff584748fa928b182c1853cbed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 14:58:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5da744f1d4b24f8eae8254f50b188fb6 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88027831b4054826a1e5cdddec89f0eb : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f696ded48f604832bcee0456ed9ac5ef : SUCCESS in 9m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d3c5f6b2efc4768918413d967c998cb : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fde7ae0380664139807eedbb20da54ec : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/531f7c6723c746e2917a9b57984e6e1d : SUCCESS in 7m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad71c8dfa3994a2784713bf7003d1a79 : SUCCESS in 11m 27s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/bb4ab8264c2e46f2ae66dec912cdfff7 : SUCCESS in 47m 50s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea5475f0af4346c5962862124ecdd240 : SUCCESS in 49m 07s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ffa1833b0734b97b5cfdef2d39b5407 : SUCCESS in 43m 57s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/97a9f561c4dc4e288dc934bb050bb0fd : SUCCESS in 46m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dee5ad55b28649ac88e5dc241666bf5c : SUCCESS in 44m 44s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e6a43b9403774d8ab2436ab1b6316da7 : SUCCESS in 32m 59s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ae6feb34da7646e989e89bbda5fb4c70 : SUCCESS in 29m 14s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f68baf1656fd44feb5e8c900b89d283a : FAILURE in 1h 08m 33s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/6686922b14b54d40a186a03003dc695d : SUCCESS in 1h 03m 22s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c33653aafdc24c1c9f002a629aa7fecb : SUCCESS in 10m 33s","accounts_in_message":[],"_revision_number":2},{"id":"00000ad47a969c3bb5deb2c65796c5c8b6ed5712","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2020-03-16 11:02:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0a7fc1f5ddc49fa45db0cef01e4b2088a5dabf6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-03-16 11:20:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dc7a0610e1a4c5464f3a697b6afd618cbc4879dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 12:37:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/44ffe38113684b1abc16b792c12e49b1 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aa9b22d6b224f2289e4aa7a280b752b : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06ca1aaba3f049d88deda816a4d4cb5b : SUCCESS in 9m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98386e7e8f6d4219bb13ceb38c463e50 : SUCCESS in 12m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79ae05814b124d6a862d4fd6d08a508e : SUCCESS in 10m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e05ae2be20d417e9ffa7f38d7bfdd01 : SUCCESS in 11m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc9e178f45f24ebeb978630518506111 : SUCCESS in 9m 43s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/912ae2e25eee4ee78a8357ae7de8cd14 : SUCCESS in 45m 13s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2893160da47b4668b3568ca012e98276 : SUCCESS in 55m 45s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e42ee90f3eeb49fea47b86fb00d781e2 : SUCCESS in 54m 08s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/c4b1850f48fa4870a0c7b8388166b132 : SUCCESS in 1h 12m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fc3b4fa7264548aeb82f8efa01093726 : SUCCESS in 53m 58s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/14c8a44e869e4c1d8da53b81a948d23c : SUCCESS in 30m 46s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7b2da2ca66b94e25a7c29b2d4abe45a3 : SUCCESS in 28m 13s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/82c2528128ef4b9ea5df39ada9382480 : FAILURE in 1h 04m 43s (non-voting)\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/76ceafe593b54240a025e3068ca6017a : SUCCESS in 1h 12m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79d7395c89014c939988c25e518265ec : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":3},{"id":"b53a4c1537a45bb9032959ef266e92fcb3d601f5","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2020-04-17 05:50:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"876ce0869c562097159c72c47a9d14fbc4b5f583","revisions":{"b018b6a43b2a3f04e7f64978cb284361334125ec":{"kind":"REWORK","_number":1,"created":"2020-03-13 12:23:54.000000000","uploader":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"ref":"refs/changes/34/712934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/712934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/712934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/712934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/712934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/712934/1"}}},"commit":{"parents":[{"commit":"dbfc0bea227e45ba36756bc5e91aba36f83c2f09","subject":"Merge \"Fix incorrect in-depth search of affected tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfc0bea227e45ba36756bc5e91aba36f83c2f09"}]}],"author":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2020-03-13 12:21:40.000000000","tz":330},"committer":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2020-03-13 12:23:41.000000000","tz":330},"subject":"[DNM] Testing OOO Mistral tests.","message":"[DNM] Testing OOO Mistral tests.\n\nDepends-On: https://review.opendev.org/#/c/712903/\n\nChange-Id: I5399cee2b95ec66574507f08b8baa26057f4bde9\nSigned-off-by: Amol Kahat \u003camolkahat@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b018b6a43b2a3f04e7f64978cb284361334125ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b018b6a43b2a3f04e7f64978cb284361334125ec"}]},"branch":"refs/heads/master"},"59a9375d11afda13da6c43163313ac2d9a975b5c":{"kind":"REWORK","_number":2,"created":"2020-03-13 13:44:01.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/34/712934/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/712934/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/712934/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/712934/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/712934/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/712934/2"}}},"commit":{"parents":[{"commit":"dbfc0bea227e45ba36756bc5e91aba36f83c2f09","subject":"Merge \"Fix incorrect in-depth search of affected tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfc0bea227e45ba36756bc5e91aba36f83c2f09"}]}],"author":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2020-03-13 12:21:40.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-03-13 13:42:40.000000000","tz":330},"subject":"[DNM] Testing OOO Mistral tests.","message":"[DNM] Testing OOO Mistral tests.\n\nDepends-On: https://review.opendev.org/#/c/712903/\n\nChange-Id: I5399cee2b95ec66574507f08b8baa26057f4bde9\nSigned-off-by: Amol Kahat \u003camolkahat@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/59a9375d11afda13da6c43163313ac2d9a975b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/59a9375d11afda13da6c43163313ac2d9a975b5c"}]},"branch":"refs/heads/master"},"876ce0869c562097159c72c47a9d14fbc4b5f583":{"kind":"REWORK","_number":3,"created":"2020-03-16 11:20:46.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/34/712934/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/712934/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/712934/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/712934/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/712934/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/712934/3"}}},"commit":{"parents":[{"commit":"dbfc0bea227e45ba36756bc5e91aba36f83c2f09","subject":"Merge \"Fix incorrect in-depth search of affected tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfc0bea227e45ba36756bc5e91aba36f83c2f09"}]}],"author":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2020-03-13 12:21:40.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-03-16 11:20:15.000000000","tz":330},"subject":"[DNM] Testing OOO Mistral tests.","message":"[DNM] Testing OOO Mistral tests.\n\nDepends-On: https://review.opendev.org/#/c/712903/\n\nChange-Id: I5399cee2b95ec66574507f08b8baa26057f4bde9\nSigned-off-by: Amol Kahat \u003camolkahat@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/876ce0869c562097159c72c47a9d14fbc4b5f583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/876ce0869c562097159c72c47a9d14fbc4b5f583"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
