)]}'
{"id":"openstack%2Fmistral~713873","triplet_id":"openstack%2Fmistral~master~I3dbf26e3fdf2d2bd17d99b678172bd9ac02a7323","project":"openstack/mistral","branch":"master","topic":"fix_missing_map_option","hashtags":[],"change_id":"I3dbf26e3fdf2d2bd17d99b678172bd9ac02a7323","subject":"Keep openstack mapping_path option","status":"MERGED","created":"2020-03-19 13:11:40.000000000","updated":"2020-03-20 05:59:00.000000000","submitted":"2020-03-20 05:57:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713873-1584683855586-7629276d","meta_rev_id":"0cae674a87ed98c2a40cb7a407a36a78aa14bb1e","_number":713873,"virtual_id_number":713873,"owner":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 05:57:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":2,"date":"2020-03-20 05:05:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-19 16:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"value":0,"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"value":1,"date":"2020-03-20 05:05:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 13:46:53.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-03-19 14:17:26.000000000","updated_by":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"reviewer":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"state":"REVIEWER"},{"updated":"2020-03-19 16:56:53.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-03-20 05:05:40.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-03-20 05:57:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f8c360334a2eee17a678b3eb04002c9470fa313","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-03-19 13:11:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f08ceaddbffa8d77f1a67a5822dff480519f2cb9","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-03-19 13:12:50.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"66caa228c4022d635fefc969ecf11a555cb2768f","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-03-19 13:16:19.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b92167ed54f9e23eb9bd8de9c6d6b021c42ea7ca","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-03-19 13:46:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ad05015d81078aa917320f93a22f79d502db4408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 14:16:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e6d06dc91f145bca6ab36408334be3b : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e50b29312354956ad9f7cd6bcf58d5e : SUCCESS in 10m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17e130ea3e034d2aa0978153c8965529 : SUCCESS in 13m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c256f721d5214226aceba2a197b58646 : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09f4f506b37745f58e8af158309254a5 : SUCCESS in 9m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3c4e008e5d64f838c8051df949f6e0b : SUCCESS in 10m 44s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/857ffe56de144868bf27efd6499b7c3e : SUCCESS in 47m 16s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0fb27d4e9e849a893d527767c53b22b : SUCCESS in 56m 06s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d10ead48e9dd4164bd9f6c5d6773b492 : SUCCESS in 47m 50s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/213a7c121c024152978d99abb8680fe3 : SUCCESS in 47m 25s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/1e25285abd3247c5885713695d920369 : SUCCESS in 52m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/05ddb65f53484bf5b82d1cf59ecbf3ad : SUCCESS in 29m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/66fd8c548d804d7caad998589d42845c : SUCCESS in 31m 26s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/3fccd3e7c6b64306b7ccc22bc60bb9f8 : SUCCESS in 56m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b1d96d7cb754e4cad6aae6e1c655e2e : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":3},{"id":"b72d9d4353d6a466f3808197b03957fb7b9782b0","author":{"_account_id":9712,"name":"Dougal Matthews","email":"dougal.matthews@canonical.com","username":"dougal"},"date":"2020-03-19 14:17:26.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"874c4334e128ddca2ca8f1eeed2e243d58fcc101","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-03-19 15:35:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e08414c359363cd5023119606b921a0cccd6f1da","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2020-03-19 15:36:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d2583f68deaae4d966517d93d6c08a37d7e4164a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 16:49:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0308208563349eca90766401ccbef32 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54e716627af84056a03ac025331b7cbb : SUCCESS in 9m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca97c7a98c874a7a970168447534e116 : SUCCESS in 11m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1be9c89b3b5b4fab9bc04108eec3738a : SUCCESS in 12m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fb4aa9cfa6a44a5bab9a0c2a04a23fd : SUCCESS in 7m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d3dd48134c24ca9bb2e6d9643eb1b7d : SUCCESS in 11m 00s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/039ff0c2813f457bbe293b95d249e0d6 : SUCCESS in 47m 11s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58f3e41b1ec54cfc83b9bb6289ac4803 : SUCCESS in 47m 28s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2edc0a96207c45c497c6eebaf015f693 : SUCCESS in 53m 26s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3c176d5725b54e8eb2c7ffa87768f96e : SUCCESS in 42m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/569ee4cce25a4c10b768a687013566a3 : SUCCESS in 46m 48s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1fb5577cac9b43baa690624f58a5bfdb : SUCCESS in 30m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b19f502417404ff4bc7d548af1d7a2c8 : SUCCESS in 33m 46s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/2060891142cf49d1b98aa535f566abd6 : SUCCESS in 1h 07m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc02d490c9af44bc8fc946f9b1e58616 : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":4},{"id":"4c8e0186a64f88abba14c361b82e18c4ad947efa","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-03-19 16:56:53.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0e70ec529cb747eba5eac51a701214e5dd66a5cb","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-03-20 05:05:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLooks OK to me although we\u0027re going to change installation procedures significantly. Currently it\u0027s all hardcoded and inflexible whereas we want to make it more customizable (i.e. installing only part of the OpenStack actions).","accounts_in_message":[],"_revision_number":4},{"id":"04685eaebd1a0a7014e9d7ebfc2bb45db60f578b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 05:05:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0f5b1c92bd1ea3646563205e8fb364d223eeff38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 05:57:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3bc12ada1af4b30a1e75bea42fe6d06 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d9430e8f5634c08b738153261a6c51b : SUCCESS in 7m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3184aa7e6f049d99efd8a5082294ddb : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d86edc397e64c84965cf9d6682c23f6 : SUCCESS in 6m 11s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/0516e0d3ab50455da3ab0b25ab2e9cd7 : SUCCESS in 44m 54s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdf9f32d7b48489db66eae07a616237c : SUCCESS in 46m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99026c97ff1f4ef1b275a1a0aa0252c8 : SUCCESS in 45m 45s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0214b9303cb54057a865338ca645806b : SUCCESS in 42m 31s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/f4e6b76f161049108c2174807e80750e : SUCCESS in 28m 06s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7988e536c57746ad854a8e7ea62160f3 : SUCCESS in 26m 37s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/87966c39cc7249bb963757434100ecee : SUCCESS in 43m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/375bfd366f0a4ae48ef3d834d17a3bb2 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":4},{"id":"fe41c296859e5226a81e3cf2dfd0217650baf627","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 05:57:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ca67b31b33c08c3f44ece10a43e57fc22ddeb7c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 05:59:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65f4b4397c241d3bb89632c25b2f7cf : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a594a2523e0b19100f3699f502cf617405698fc7","revisions":{"5921b064d8a3a7b88862e64d2632945df7aaf5b2":{"kind":"REWORK","_number":1,"created":"2020-03-19 13:11:40.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/73/713873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/73/713873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/713873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/713873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/713873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/73/713873/1"}}},"commit":{"parents":[{"commit":"dbfc0bea227e45ba36756bc5e91aba36f83c2f09","subject":"Merge \"Fix incorrect in-depth search of affected tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfc0bea227e45ba36756bc5e91aba36f83c2f09"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2020-03-19 13:01:29.000000000","tz":60},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2020-03-19 13:01:29.000000000","tz":60},"subject":"Keep openstack_actions_mapping_path option to keep backwards compatibility","message":"Keep openstack_actions_mapping_path option to keep backwards compatibility\n\nWhen we moved the openstack actions to mistral-extra this option was removed\nThis adds it back if mistral-extra is installed otherwise it keeps the same\n\nChange-Id: I3dbf26e3fdf2d2bd17d99b678172bd9ac02a7323\nCloses-bug: #1866621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5921b064d8a3a7b88862e64d2632945df7aaf5b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/5921b064d8a3a7b88862e64d2632945df7aaf5b2"}]},"branch":"refs/heads/master"},"20c6633050deacee05517d6243f2841c8cd1aa6e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-19 13:12:50.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/73/713873/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/73/713873/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/713873/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/713873/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/713873/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/73/713873/2"}}},"commit":{"parents":[{"commit":"dbfc0bea227e45ba36756bc5e91aba36f83c2f09","subject":"Merge \"Fix incorrect in-depth search of affected tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfc0bea227e45ba36756bc5e91aba36f83c2f09"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2020-03-19 13:01:29.000000000","tz":60},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2020-03-19 13:12:27.000000000","tz":60},"subject":"Keep openstack mapping_path option to keep backwards compatibility","message":"Keep openstack mapping_path option to keep backwards compatibility\n\nWhen we moved the openstack actions to mistral-extra this\noption was removed. This adds it back if mistral-extra is\ninstalled otherwise it keeps the same\n\nChange-Id: I3dbf26e3fdf2d2bd17d99b678172bd9ac02a7323\nCloses-bug: #1866621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/20c6633050deacee05517d6243f2841c8cd1aa6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/20c6633050deacee05517d6243f2841c8cd1aa6e"}]},"branch":"refs/heads/master"},"f22461af361934d1eb12568549d6d17d83c09423":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-03-19 13:16:19.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/73/713873/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/73/713873/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/713873/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/713873/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/713873/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/73/713873/3"}}},"commit":{"parents":[{"commit":"dbfc0bea227e45ba36756bc5e91aba36f83c2f09","subject":"Merge \"Fix incorrect in-depth search of affected tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfc0bea227e45ba36756bc5e91aba36f83c2f09"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2020-03-19 13:01:29.000000000","tz":60},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2020-03-19 13:13:05.000000000","tz":60},"subject":"Keep openstack mapping_path option","message":"Keep openstack mapping_path option\n\nRemoval of this option breaks backwards compatibility\n\nWhen we moved the openstack actions to mistral-extra this\noption was removed. This adds it back if mistral-extra is\ninstalled otherwise it keeps the same\n\nChange-Id: I3dbf26e3fdf2d2bd17d99b678172bd9ac02a7323\nCloses-bug: #1866621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f22461af361934d1eb12568549d6d17d83c09423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/f22461af361934d1eb12568549d6d17d83c09423"}]},"branch":"refs/heads/master"},"a594a2523e0b19100f3699f502cf617405698fc7":{"kind":"REWORK","_number":4,"created":"2020-03-19 15:35:53.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/73/713873/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/73/713873/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/713873/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/713873/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/73/713873/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/73/713873/4"}}},"commit":{"parents":[{"commit":"dbfc0bea227e45ba36756bc5e91aba36f83c2f09","subject":"Merge \"Fix incorrect in-depth search of affected tasks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/dbfc0bea227e45ba36756bc5e91aba36f83c2f09"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2020-03-19 13:01:29.000000000","tz":60},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2020-03-19 15:35:38.000000000","tz":60},"subject":"Keep openstack mapping_path option","message":"Keep openstack mapping_path option\n\nRemoval of this option breaks backwards compatibility\n\nWhen we moved the openstack actions to mistral-extra this\noption was removed. This adds it back if mistral-extra is\ninstalled otherwise it keeps the same\n\nChange-Id: I3dbf26e3fdf2d2bd17d99b678172bd9ac02a7323\nCloses-bug: #1866621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a594a2523e0b19100f3699f502cf617405698fc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a594a2523e0b19100f3699f502cf617405698fc7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
