)]}'
{"id":"openstack%2Fmistral~716485","triplet_id":"openstack%2Fmistral~stable%2Ftrain~I9e73ea7cbba215c3e1d174b5189be27c640c4d42","project":"openstack/mistral","branch":"stable/train","topic":"bug/1869168","hashtags":[],"change_id":"I9e73ea7cbba215c3e1d174b5189be27c640c4d42","subject":"Fix serialization of structures that might contain YAQL types","status":"MERGED","created":"2020-04-01 05:34:08.000000000","updated":"2020-04-02 22:54:15.000000000","submitted":"2020-04-02 22:50:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716485-1585867830592-9820b1aa","meta_rev_id":"7e6ffe28c511a7ddc11212ceb5f6c1eff68ae80d","_number":716485,"virtual_id_number":716485,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 22:50:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-04-02 07:50:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-04-02 07:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-01 09:14:41.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-02 22:50:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06531fa2d124ceb5bf8e13bbc47752737c6bed65","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-01 05:34:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"402888708cda60c2bc5586cf9ebf9f358b581fb0","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-01 05:38:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis is a bit modified version of https://review.opendev.org/#/c/714416/ that doesn\u0027t have handling for generators. This is because we can\u0027t use the newer version of oslo.serialization (2.21.1) that contains the parameter \"fallback\" in the method \"jsonutils.to_primitive\".","accounts_in_message":[],"_revision_number":1},{"id":"1bd5b325adb98f062c1a11ddfee9686df48b9e59","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-01 05:41:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"10db3adb76274fa22f292c55b691b4306780bc4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 09:13:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f131679c7f914fbda36e659a73052aba : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b056efc260c44ee8b866ba41a9b7ed5 : SUCCESS in 8m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea916891016c44a098da862c4f60a832 : SUCCESS in 8m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e0680e122db4fbba2ab4c8ef48e9ad3 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a0c0f5d1f744d2bbd8e68dbc2828a1b : SUCCESS in 4m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fecab55310cc48d784b08766398aa275 : SUCCESS in 10m 07s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/1ee4d323346641b0846766020bf29f37 : SUCCESS in 41m 06s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/e6fa3612113a4bc08afd8a5aff116e04 : SUCCESS in 42m 50s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/522607482d6547899490b60b35d1ec18 : SUCCESS in 48m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/bdab1895e8b44a4cb3249c6857542c9d : SUCCESS in 30m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6598fb295d664d76b48b12cb54e63f1b : SUCCESS in 30m 20s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/030fcc1add10466d8d14aed644169f47 : SUCCESS in 1h 25m 05s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a8ce6489fdf043ab824e3151a9ef2e35 : TIMED_OUT in 3h 30m 43s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/a48cf0f45c134f7f99ca0e1fdaf40255 : FAILURE in 18m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fdf8dc3ee9c427eaea1f1c955dbab23 : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":2},{"id":"9aba77bab53c447eca27105412a826fdf394cd69","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-01 09:14:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3621082908c98f1ebf7a6ca507db21a1e7b5be48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 12:10:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b3cb42ee3564f9a8a7b3fcec79002d1 : SUCCESS in 6m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b87806b98d91417c8a96e63e6ffebab3 : SUCCESS in 12m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0afff4d59eb435f8feee655193913a4 : SUCCESS in 10m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5a24fac82244e4296d943df274721f0 : SUCCESS in 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66a99e5d68884f868f7e580aea4fb2fd : SUCCESS in 5m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb2f60f1da3b4eb792aae6b953215b55 : SUCCESS in 12m 30s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/fd8477ecb12341a7afad9c5ab2d2a26e : SUCCESS in 45m 39s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/07ac41a5903e462ea79749381145f1fb : SUCCESS in 46m 49s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/2f2b990c6f484d8b8e799a9c2505d619 : SUCCESS in 49m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/e6a429b9e70642b694013ea33285d1e1 : SUCCESS in 31m 38s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/38ab39c758054ec0ba8f100f60a36144 : SUCCESS in 29m 39s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/0c2069bba95c4f78bc062db0b10377ba : SUCCESS in 1h 26m 36s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/2270a60ebcf34ae4bbfe67c55186e8e8 : SUCCESS in 2h 50m 33s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9dc81e5758c242329cee5823ea87f372 : FAILURE in 18m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2cab30f520843fab18e92dfa5032001 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5575b186d22cdb37701203ca14b81e1dd02eb36e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-01 13:49:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"716b893ac203bbfd5c232a7e0176e71188519273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 13:49:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0b0cb352fab47b0f61f970e989ba06a9fe0b8f87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 16:23:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06011deed364dbea67096bb52ac98dc : SUCCESS in 8m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb2f6fa4ced848bbb4f45fbfe419431b : SUCCESS in 11m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/286fde3054b44ca5b425f6403472e7bd : SUCCESS in 10m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15fc210c6e6a416cbdda327793ddbede : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d50209cfb841229d6f2b002349dccf : SUCCESS in 7m 08s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/c897fab059d341c595c28460a589c63c : SUCCESS in 49m 53s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/45091616dd904f09a477286fe3be934e : SUCCESS in 48m 02s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/005b52fc400e4eeb81ce2055ae23d025 : SUCCESS in 30m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/e6c803ef6795474392db6f2473d2c401 : SUCCESS in 29m 51s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e390e03831c34f83b2734b26d0e79aee : SUCCESS in 1h 24m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/14e368f7472449e3b44263c5aa02ad13 : SUCCESS in 49m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ac2ea2a0a033408c8cd3cf8050405707 : SUCCESS in 11m 23s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2b90383cea2c4c8e97c39a72b6ec213a : FAILURE in 2h 31m 22s","accounts_in_message":[],"_revision_number":2},{"id":"0f6870b5711b137f3eeb0b33a438ddbd824e4cfd","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-02 06:22:37.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1cb9aaa8580179dc9a45ed389c3e4935005e3822","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-02 07:50:46.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"310a123f032388e3e151feaa713f94eb1b5186cc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-02 07:50:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"39c7a2624ee7d8e81f06ceef4f01b6857ec1678d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 11:00:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3433f6953765435ebc5233d02cb347f3 : SUCCESS in 7m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/180368e4e05548458877a3a546c21241 : SUCCESS in 13m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc3965e4ef474a9396f8e9b07de47620 : SUCCESS in 10m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1dd0e81b6e5a4bc789df52a2ca785e50 : SUCCESS in 12m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/388a0945aa744e0790548f4e5f49a00c : SUCCESS in 10m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e7a48b7d27d4a9fbc493da7f813e1c1 : SUCCESS in 12m 44s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/73216abe5a3546198d0fa5de958e8f5a : FAILURE in 40m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/1d6c4ea31dfe41f9a55a2e2463e0c9c7 : FAILURE in 38m 30s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/5fbda2cca66546c69ce0fef0f1a2aefb : FAILURE in 38m 49s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/89428d1fabd04658a0ae2645d705022e : SUCCESS in 30m 53s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8d46a330132141da9fef750951fb59df : SUCCESS in 30m 11s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/8997abc948aa4e08908b62f8459d9538 : SUCCESS in 1h 16m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ef5eb781ef7c43c9a88df46790f1c209 : SUCCESS in 3h 03m 31s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/50faa1a5a79542f181dbbb5131b4f9da : FAILURE in 21m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fba6c6ca9114d79b3d29c3d5100e634 : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":3},{"id":"8f094d0149f4ddeb8ce1c46c1fc8183f98f4b6aa","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-02 16:28:52.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cbd05ac397ed05a77e66dacaa6f917dee60afe23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 21:01:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34cedc9d69e4488c9aab724ace0e0a8a : SUCCESS in 6m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4969f7b9578a48aea8e2abf3fd26a5bd : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca6c71e44c654930b0d1e376e56642c0 : SUCCESS in 10m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/641b03efa65f4db19750eae624379a17 : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62423931401a4a54af79d6186611b73d : SUCCESS in 8m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bee1e4983344c5faa314068a2cef2a0 : SUCCESS in 10m 39s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2c8a9026c8834886bb95f5f996b82816 : SUCCESS in 47m 20s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/0b9e6146ef1a4ada8b95708a8eac6501 : SUCCESS in 50m 38s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/dab6c4771e524e37a874b2b0eeba5a37 : SUCCESS in 54m 18s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3f1c89da0adb41e1a5898ca345099ce1 : SUCCESS in 30m 40s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0f326d658c40446891f6291f27a450b0 : SUCCESS in 28m 21s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/c8b290c812ee4140a599f84e5d274875 : SUCCESS in 1h 22m 10s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a661266998b240429740bf7d8748b578 : SUCCESS in 2h 54m 29s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/65a9f595bd4a4cd6801fada75eef5b3e : FAILURE in 21m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dd0c89b5d1846e597d87316dceaa5a5 : SUCCESS in 12m 46s","accounts_in_message":[],"_revision_number":3},{"id":"f4085c3fdc1bd633eb7d0b82ad91e221c91a0af4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 21:06:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"135db199c6dfedaf031a0214349067e7949e247d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:50:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ece65264dd74a1ba5b2cdff60cbe8ec : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d32dae04ed247c6822aefdcdd7b8883 : SUCCESS in 11m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5939f8670f3c4ac78cdbc9b6ef38b050 : SUCCESS in 12m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d5c65d375364a3295752f3797ee7601 : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3d4eec57cb44f14b909081d9996836d : SUCCESS in 7m 49s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/f62b4578861c4fb895ce119059b83974 : SUCCESS in 49m 38s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/484d8708a2d044e0a451007d70408105 : SUCCESS in 49m 04s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c9e52d8b03e349e58cec586105769e98 : SUCCESS in 32m 57s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/d59073ee2d1144d8b2edad9c8d6595f3 : SUCCESS in 30m 08s\n- tripleo-ci-centos-7-scenario003-standalone https://zuul.opendev.org/t/openstack/build/b59e1fcbf2744eba95de4d2895ca2fea : SUCCESS in 1h 31m 55s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/de7ccb3ea57d4506a20b1f0ca231dca7 : SUCCESS in 44m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6987e4a5f10f4e329cf79e6567aecb5c : SUCCESS in 13m 48s\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/1f05bcfdcf4f48c488166bb151504ee1 : SUCCESS in 1h 36m 16s","accounts_in_message":[],"_revision_number":3},{"id":"256a0bffb4db41c307bdea175aaf3886ae6bdc3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:50:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7e0d3f2dbb751303d802239e965e57eccd4a0a15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:54:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f25186ad62a44ed8c0cdb2f2ea455e5 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1e0013835c10cbbdfb747625d29cbbe9d6852948","revisions":{"15ca377cdc81d29d797076e18100d670051da58b":{"kind":"REWORK","_number":1,"created":"2020-04-01 05:34:08.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/85/716485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/716485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/716485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/716485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/716485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/716485/1"}}},"commit":{"parents":[{"commit":"d6037f1d5f0d5024966b46c80a517d6111e76809","subject":"Merge \"Fix ContextView JSON serialization\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6037f1d5f0d5024966b46c80a517d6111e76809"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-23 10:42:28.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-01 05:31:14.000000000","tz":420},"subject":"Fix serialization of structures that might contain YAQL types","message":"Fix serialization of structures that might contain YAQL types\n\n* When YAQL output data conversion is disabled there\u0027s still\n  an issue caused by presence of not JSON-compatible types within\n  a YAQL result. The internal Mistral code is already able to deal\n  with that (due to the previous changes) by checking that and\n  converting them to what\u0027s needed. However, JSON serialization\n  may still not work if it\u0027s done via the standard \"json\" library.\n  The library simply doesn\u0027t handle those non-standard types and\n  raises an exception. We have a sanitizing function that all YAQL\n  results go through, however, it doesn\u0027t make sense to do recursive\n  sanitizing for performance reasons. It does make sense to convert\n  data as late as possible to avoid redundant data manipulations. So\n  the sanitizing function handles only the root object in the object\n  graph. The solution for this problem is to use our own utility\n  function based on the \"oslo_serialization.jsonutils\" that is able\n  to deal with at least part of the mentioned types, specifically\n  FrozenDict and iterators.\n* Used the new function \"utils.to_json_str()\" everywhere for JSON\n  serialization, including the action \"std.http\".\n* Added necessary unit tests.\n\nCloses-Bug: #1869168\nDepends-On: I1081a44a6f305eb1dfe68a5bad30110385130725\nChange-Id: I9e73ea7cbba215c3e1d174b5189be27c640c4d42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/15ca377cdc81d29d797076e18100d670051da58b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/15ca377cdc81d29d797076e18100d670051da58b"}]},"branch":"refs/heads/stable/train"},"ebb2d260beef51ec3e4b48a46a21b4c00642d631":{"kind":"REWORK","_number":2,"created":"2020-04-01 05:41:02.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/85/716485/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/716485/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/716485/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/716485/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/716485/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/716485/2"}}},"commit":{"parents":[{"commit":"d6037f1d5f0d5024966b46c80a517d6111e76809","subject":"Merge \"Fix ContextView JSON serialization\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6037f1d5f0d5024966b46c80a517d6111e76809"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-23 10:42:28.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-01 05:40:47.000000000","tz":420},"subject":"Fix serialization of structures that might contain YAQL types","message":"Fix serialization of structures that might contain YAQL types\n\n* When YAQL output data conversion is disabled there\u0027s still\n  an issue caused by presence of not JSON-compatible types within\n  a YAQL result. The internal Mistral code is already able to deal\n  with that (due to the previous changes) by checking that and\n  converting them to what\u0027s needed. However, JSON serialization\n  may still not work if it\u0027s done via the standard \"json\" library.\n  The library simply doesn\u0027t handle those non-standard types and\n  raises an exception. We have a sanitizing function that all YAQL\n  results go through, however, it doesn\u0027t make sense to do recursive\n  sanitizing for performance reasons. It does make sense to convert\n  data as late as possible to avoid redundant data manipulations. So\n  the sanitizing function handles only the root object in the object\n  graph. The solution for this problem is to use our own utility\n  function based on the \"oslo_serialization.jsonutils\" that is able\n  to deal with at least part of the mentioned types, specifically\n  FrozenDict and iterators.\n* Used the new function \"utils.to_json_str()\" everywhere for JSON\n  serialization, including the action \"std.http\".\n* Added necessary unit tests.\n\nCloses-Bug: #1869168\nDepends-On: I1081a44a6f305eb1dfe68a5bad30110385130725\nChange-Id: I9e73ea7cbba215c3e1d174b5189be27c640c4d42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ebb2d260beef51ec3e4b48a46a21b4c00642d631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/ebb2d260beef51ec3e4b48a46a21b4c00642d631"}]},"branch":"refs/heads/stable/train"},"1e0013835c10cbbdfb747625d29cbbe9d6852948":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-02 07:50:46.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/85/716485/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/85/716485/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/716485/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/716485/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/85/716485/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/85/716485/3"}}},"commit":{"parents":[{"commit":"d6037f1d5f0d5024966b46c80a517d6111e76809","subject":"Merge \"Fix ContextView JSON serialization\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d6037f1d5f0d5024966b46c80a517d6111e76809"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-03-23 10:42:28.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-02 07:50:41.000000000","tz":0},"subject":"Fix serialization of structures that might contain YAQL types","message":"Fix serialization of structures that might contain YAQL types\n\n* When YAQL output data conversion is disabled there\u0027s still\n  an issue caused by presence of not JSON-compatible types within\n  a YAQL result. The internal Mistral code is already able to deal\n  with that (due to the previous changes) by checking that and\n  converting them to what\u0027s needed. However, JSON serialization\n  may still not work if it\u0027s done via the standard \"json\" library.\n  The library simply doesn\u0027t handle those non-standard types and\n  raises an exception. We have a sanitizing function that all YAQL\n  results go through, however, it doesn\u0027t make sense to do recursive\n  sanitizing for performance reasons. It does make sense to convert\n  data as late as possible to avoid redundant data manipulations. So\n  the sanitizing function handles only the root object in the object\n  graph. The solution for this problem is to use our own utility\n  function based on the \"oslo_serialization.jsonutils\" that is able\n  to deal with at least part of the mentioned types, specifically\n  FrozenDict and iterators.\n* Used the new function \"utils.to_json_str()\" everywhere for JSON\n  serialization, including the action \"std.http\".\n* Added necessary unit tests.\n\n\nCloses-Bug: #1869168\nDepends-On: I1081a44a6f305eb1dfe68a5bad30110385130725\nChange-Id: I9e73ea7cbba215c3e1d174b5189be27c640c4d42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1e0013835c10cbbdfb747625d29cbbe9d6852948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/1e0013835c10cbbdfb747625d29cbbe9d6852948"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
