)]}'
{"id":"openstack%2Fmistral~718316","triplet_id":"openstack%2Fmistral~master~I47abe0904b49d72e33eb10080c71fb81980d44da","project":"openstack/mistral","branch":"master","topic":"bug/1871567","hashtags":[],"change_id":"I47abe0904b49d72e33eb10080c71fb81980d44da","subject":"Add YAQL sanitizing for iterators","status":"MERGED","created":"2020-04-08 06:26:31.000000000","updated":"2020-04-08 11:14:17.000000000","submitted":"2020-04-08 11:12:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718316-1586344332648-fdafd74b","meta_rev_id":"13c8ca0b2f18a861238341cc4e59637d357e585f","_number":718316,"virtual_id_number":718316,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 11:12:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"all":[{"value":2,"date":"2020-04-08 09:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-08 09:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":0,"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"value":0,"date":"2020-04-08 09:48:55.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},{"value":1,"date":"2020-04-08 10:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-08 09:17:57.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-08 09:48:55.000000000","updated_by":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"reviewer":{"_account_id":29124,"name":"Oleg Ovcharuk","email":"vgvoleg@gmail.com","username":"vgvoleg"},"state":"REVIEWER"},{"updated":"2020-04-08 09:52:55.000000000","updated_by":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"reviewer":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"state":"REVIEWER"},{"updated":"2020-04-08 11:12:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"28e7d6e73518a61ce50c5d200ec20b63c670925c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-08 06:26:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1990d958221e1e5937fe710cd21ba232961efd52","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-08 07:11:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ba67672653bcad866e5cebaf0068fcea4746d6e3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-08 08:07:57.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c99145dda4e69bff73b02ebdeb6c57621584175a","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-08 08:30:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a82af49ad475e2234c95fe32d712dae0af9f4fff","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-08 08:44:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6d10d0f5e79250c7ac59d8bfc651a45d21077dde","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-08 09:17:57.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"058db5f19fd5842dac3235f0f99d099913f64126","author":{"_account_id":21970,"name":"Andras Kovi","email":"akovi@nokia.com","username":"andras.kovi"},"date":"2020-04-08 09:52:55.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d3199a0881a500edd29d69906c02f08df102c883","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-08 10:04:13.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"be1b9f145b95cfb137eb2cb8c1a8bf2d1fdd1b13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 10:08:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1211b549e0b74e0da3f63872e3258679 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7650d42673d45ea80290f89f3c32ff7 : SUCCESS in 10m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4f3e3e449f264d2f84e8adaa3b1f0598 : SUCCESS in 11m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/189f7ffcfcc04a049d5a84dda1c0333c : SUCCESS in 11m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2fa7f3aff54b56af57ab785ba85b63 : SUCCESS in 10m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/97576765df3e4c73b50a7a9af49e856b : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/790f66716bc74d96b9daa8603c605acc : SUCCESS in 6m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb9e228f7b1547c481ac357df8cb02e7 : SUCCESS in 10m 22s (non-voting)\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/5d4173060b0a4cbaa239de0410eca0a7 : SUCCESS in 44m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c6356d75d5547daad8433ab5841ada6 : SUCCESS in 50m 10s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ea05dfeca7447c19119d735b4d8b143 : SUCCESS in 55m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3414bbb3512f4cadb0d44ac6890e3f18 : SUCCESS in 55m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8684b7d4cc474d7ab28d9af06e2e5130 : SUCCESS in 54m 16s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/0f3b7c8b50c04df0ac18232c6dd32385 : SUCCESS in 31m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0da8e8fdc0c14612a7bc09e403b77e28 : SUCCESS in 31m 11s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/d9abd00ee41745c6807839032afbdae6 : SUCCESS in 1h 04m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6b2b65c18a74656955019e879c2f24e : SUCCESS in 10m 42s\n\nWarning:\n  Comments left for invalid file mistral/tests/unit/base.py\n  Comments left for invalid file mistral/context.py\n  Comments left for invalid file mistral/rpc/kombu/base.py\n  Comments left for invalid file mistral/tests/unit/engine/base.py\n  Comments left for invalid file mistral/db/sqlalchemy/base.py\n  Comments left for invalid file mistral/tests/unit/services/test_event_engine.py\n  Comments left for invalid file mistral/tests/unit/api/v2/test_workflows.py\n  Comments left for invalid file mistral/expressions/std_functions.py\n  Comments left for invalid file mistral/tests/unit/rpc/kombu/test_kombu_hosts.py","accounts_in_message":[],"_revision_number":5},{"id":"cb8e21c0c7178563af26288057c319fe1fa528bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 10:09:00.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"035acd36c8c91ac69fc18ec940b4c816d20d0bf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:12:12.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae4b089bed39428785ec3136a6548516 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88a179b3648f4a1dbfaa2299db106e55 : SUCCESS in 9m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d27310e6f49d43d2b5b79214df9d778b : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85b7b2f96d354a25b99471673d17b636 : SUCCESS in 9m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b632e5046a32434095012e6e6725eb4d : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0325e8e18344b97ba796e25b981a2e7 : SUCCESS in 4m 52s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6a2af8bfc31249d190f5b35e0f038209 : SUCCESS in 58m 05s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fbd8cc879ba469daf3d7322e3901513 : SUCCESS in 52m 22s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c517291869947558d3b1953ecea3bce : SUCCESS in 59m 17s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/8ccc1226a06748c0a8886c25534e2388 : SUCCESS in 1h 00m 33s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/a509789b6f3f4817a7eefd78ce731921 : SUCCESS in 31m 13s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/1157b4f5030b41d996839dab95fbe5d9 : SUCCESS in 29m 44s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/64f6ffbacade4f9593a605189d8ce4c6 : SUCCESS in 47m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbfeef0567214afda7b66b514bf2161a : SUCCESS in 11m 01s\n\nWarning:\n  Comments left for invalid file mistral/context.py\n  Comments left for invalid file mistral/expressions/std_functions.py\n  Comments left for invalid file mistral/tests/unit/engine/base.py\n  Comments left for invalid file mistral/db/sqlalchemy/base.py\n  Comments left for invalid file mistral/tests/unit/services/test_event_engine.py\n  Comments left for invalid file mistral/tests/unit/api/v2/test_workflows.py\n  Comments left for invalid file mistral/rpc/kombu/base.py\n  Comments left for invalid file mistral/tests/unit/rpc/kombu/test_kombu_hosts.py","accounts_in_message":[],"_revision_number":5},{"id":"5d748aa886c6940941c4f6311947f48c5d8b535c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:12:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"32d42ad9a430fe7933f4c4a197c454c7e7763441","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 11:14:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f58e179226c04f1fb396518e9573ecd3 : SUCCESS in 1m 34s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab9135b1227b4e26a05b1df4d5eb7bf1 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6fb17d8266ab6cc6d6389a5fec1282d1094658c1","revisions":{"a552181961f451f73b414e0e8814dce07e32c7f8":{"kind":"REWORK","_number":1,"created":"2020-04-08 06:26:31.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/16/718316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/16/718316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/16/718316/1"}}},"commit":{"parents":[{"commit":"28202566c0515d5ccb61436a5a87639d3fbb2b05","subject":"Merge \"Added two new Rally scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28202566c0515d5ccb61436a5a87639d3fbb2b05"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-08 06:22:49.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-08 06:26:14.000000000","tz":420},"subject":"WIP: Add YAQL sanitizing for iterators","message":"WIP: Add YAQL sanitizing for iterators\n\nChange-Id: I47abe0904b49d72e33eb10080c71fb81980d44da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a552181961f451f73b414e0e8814dce07e32c7f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/a552181961f451f73b414e0e8814dce07e32c7f8"}]},"branch":"refs/heads/master"},"92570af5254d55fc998144725c390cd7b8497909":{"kind":"REWORK","_number":2,"created":"2020-04-08 07:11:45.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/16/718316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/16/718316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/16/718316/2"}}},"commit":{"parents":[{"commit":"28202566c0515d5ccb61436a5a87639d3fbb2b05","subject":"Merge \"Added two new Rally scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28202566c0515d5ccb61436a5a87639d3fbb2b05"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-08 06:22:49.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-08 07:11:28.000000000","tz":420},"subject":"WIP: Add YAQL sanitizing for iterators","message":"WIP: Add YAQL sanitizing for iterators\n\nChange-Id: I47abe0904b49d72e33eb10080c71fb81980d44da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/92570af5254d55fc998144725c390cd7b8497909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/92570af5254d55fc998144725c390cd7b8497909"}]},"branch":"refs/heads/master"},"0b55e38c5264e30616693301f7ba2330c66dedff":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-08 08:07:57.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/16/718316/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/16/718316/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/16/718316/3"}}},"commit":{"parents":[{"commit":"28202566c0515d5ccb61436a5a87639d3fbb2b05","subject":"Merge \"Added two new Rally scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28202566c0515d5ccb61436a5a87639d3fbb2b05"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-08 06:22:49.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-08 07:59:34.000000000","tz":420},"subject":"Add YAQL sanitizing for iterators","message":"Add YAQL sanitizing for iterators\n\n* This patch also adds handling for iterators in the sanitizing\n  function for YAQL results. JSON serialization for iterators now\n  works correctly but the issue is that they can be used only once.\n  So, for example, if we got a result from YAQL and performed JSON\n  serialization for it, then the iterator is already empty and\n  the data structure is just corrupted. So we can\u0027t further pass\n  to executors and other subsystems. The solution is just to sanitize\n  iterators after they\u0027re just returned from YAQL.\n* Added a test to make sure that action input reaches an executor\n  not corrupted, although it gets saved into DB and hence serialized\n  into JSON.\n\nCloses-Bug: #1871567\nChange-Id: I47abe0904b49d72e33eb10080c71fb81980d44da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b55e38c5264e30616693301f7ba2330c66dedff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0b55e38c5264e30616693301f7ba2330c66dedff"}]},"branch":"refs/heads/master"},"468c3e3a7236ec10d21b560c9795f42fb7e6db29":{"kind":"REWORK","_number":4,"created":"2020-04-08 08:30:36.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/16/718316/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/16/718316/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/16/718316/4"}}},"commit":{"parents":[{"commit":"28202566c0515d5ccb61436a5a87639d3fbb2b05","subject":"Merge \"Added two new Rally scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28202566c0515d5ccb61436a5a87639d3fbb2b05"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-08 06:22:49.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-08 08:28:07.000000000","tz":420},"subject":"Add YAQL sanitizing for iterators","message":"Add YAQL sanitizing for iterators\n\n* This patch also adds handling for iterators in the sanitizing\n  function for YAQL results. JSON serialization for iterators now\n  works correctly but the issue is that they can be used only once.\n  So, for example, if we got a result from YAQL and performed JSON\n  serialization for it, then the iterator is already empty and\n  the data structure is just corrupted. So we can\u0027t further pass\n  to executors and other subsystems. The solution is just to sanitize\n  iterators after they\u0027re just returned from YAQL.\n* Added a test to make sure that action input reaches an executor\n  not corrupted, although it gets saved into DB and hence serialized\n  into JSON.\n* Changed the Sphinx entry in requirements.txt so that the version\n  3.0.0 is now excluded (it breaks the build).\n\nCloses-Bug: #1871567\nChange-Id: I47abe0904b49d72e33eb10080c71fb81980d44da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/468c3e3a7236ec10d21b560c9795f42fb7e6db29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/468c3e3a7236ec10d21b560c9795f42fb7e6db29"}]},"branch":"refs/heads/master"},"6fb17d8266ab6cc6d6389a5fec1282d1094658c1":{"kind":"REWORK","_number":5,"created":"2020-04-08 08:44:32.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/16/718316/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/16/718316/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/16/718316/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/16/718316/5"}}},"commit":{"parents":[{"commit":"28202566c0515d5ccb61436a5a87639d3fbb2b05","subject":"Merge \"Added two new Rally scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/28202566c0515d5ccb61436a5a87639d3fbb2b05"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-08 06:22:49.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2020-04-08 08:44:13.000000000","tz":420},"subject":"Add YAQL sanitizing for iterators","message":"Add YAQL sanitizing for iterators\n\n* This patch also adds handling for iterators in the sanitizing\n  function for YAQL results. JSON serialization for iterators now\n  works correctly but the issue is that they can be used only once.\n  So, for example, if we got a result from YAQL and performed JSON\n  serialization for it, then the iterator is already empty and\n  the data structure is just corrupted. So we can\u0027t further pass\n  to executors and other subsystems. The solution is just to sanitize\n  iterators after they\u0027re just returned from YAQL.\n* Added a test to make sure that action input reaches an executor\n  not corrupted, although it gets saved into DB and hence serialized\n  into JSON.\n* Changed the Sphinx entry in requirements.txt so that the version\n  3.0.0 is now excluded (it breaks the build).\n\nCloses-Bug: #1871567\nChange-Id: I47abe0904b49d72e33eb10080c71fb81980d44da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6fb17d8266ab6cc6d6389a5fec1282d1094658c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/6fb17d8266ab6cc6d6389a5fec1282d1094658c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
