)]}'
{"id":"openstack%2Fmistral~722225","triplet_id":"openstack%2Fmistral~master~I48f36ad4ed2decc872e7a5ab2e086283d504da89","project":"openstack/mistral","branch":"master","topic":"mistral-guide","hashtags":[],"change_id":"I48f36ad4ed2decc872e7a5ab2e086283d504da89","subject":"typo of guide \u0027How to write a Custom Action in mistral\u0027","status":"MERGED","created":"2020-04-23 08:07:21.000000000","updated":"2020-04-24 06:54:16.000000000","submitted":"2020-04-24 06:49:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722225-1587710992877-cd7ad812","meta_rev_id":"c988609a797bfe7c56e62cd7edaae5085f181132","_number":722225,"virtual_id_number":722225,"owner":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 06:49:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-04-24 05:54:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-23 08:32:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-04-24 05:54:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 08:32:03.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-24 05:54:18.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-04-24 06:49:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a992d455568ec05d2ca9fa6183bb8dac68d05bec","author":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"date":"2020-04-23 08:07:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21e9b601c2d0cba38c15afd35b8f85076deb5b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 08:31:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc6edcfd05046d090e7117425887c2b : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ad90c1f91d342319263b3d7d1f7e369 : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":1},{"id":"8363e77058f925ca6430d0978adac67f3be4128e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-23 08:32:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dce0222e30a3f74e6b3a5103d67f142e39f2e830","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-24 05:54:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8858d05e49938b4e8c3f784bbb8961494f0e90f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 05:54:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74458e7afca0f4db5a2bc72c1557e2a453513ad0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:49:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"961ef8493b7941c91873000b2a4910f59c7b6c67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:49:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1da0a4f10cbe4ae8a6ba0acc973d24a9 : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8edf582eb2c744b58f2e54c77fd5f272 : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c4df7895698c3234403a485d864d30d52d5e74b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:54:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/645c32edd4134ae9b6ed2e5f8caba1a6 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d33313f84e4a77433ee5d3a3ef35f3440f7c34bd","revisions":{"d33313f84e4a77433ee5d3a3ef35f3440f7c34bd":{"kind":"REWORK","_number":1,"created":"2020-04-23 08:07:21.000000000","uploader":{"_account_id":22290,"name":"hyunsik Yang","email":"yangun@dcn.ssu.ac.kr","username":"hyunsikYang"},"ref":"refs/changes/25/722225/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/25/722225/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/722225/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/722225/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/25/722225/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/25/722225/1"}}},"commit":{"parents":[{"commit":"545a34c7511b6e6b5f1b59ca99891684303b4480","subject":"Use unittest.mock instead of third party mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/545a34c7511b6e6b5f1b59ca99891684303b4480"}]}],"author":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2020-04-23 08:06:11.000000000","tz":0},"committer":{"name":"hyunsikYang","email":"yangun@dcn.ssu.ac.kr","date":"2020-04-23 08:06:11.000000000","tz":0},"subject":"typo of guide \u0027How to write a Custom Action in mistral\u0027","message":"typo of guide \u0027How to write a Custom Action in mistral\u0027\n\nChange-Id: I48f36ad4ed2decc872e7a5ab2e086283d504da89\nRelated-Bug: #1874391\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d33313f84e4a77433ee5d3a3ef35f3440f7c34bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d33313f84e4a77433ee5d3a3ef35f3440f7c34bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
