)]}'
{"id":"openstack%2Fmistral~722290","triplet_id":"openstack%2Fmistral~stable%2Fussuri~If568fce338ef28bcbbb5dddde399cf96fdbab39a","project":"openstack/mistral","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"If568fce338ef28bcbbb5dddde399cf96fdbab39a","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-23 12:21:18.000000000","updated":"2020-04-24 06:26:49.000000000","submitted":"2020-04-24 06:24:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722290-1587709450708-653b66dd","meta_rev_id":"f5ca1880ceb32f0099e4e22a45e2d0ede1f423fa","_number":722290,"virtual_id_number":722290,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 06:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-04-24 05:51:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-23 12:25:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-04-24 05:51:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 12:25:33.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-24 05:51:49.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-04-24 06:24:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e60b9fa94d5e79e49b99de73b33486d999882967","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-23 12:21:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e136d3a86d7e1b2b85775fd9ebfa641d523bc9bf","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-23 12:25:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"313e12fa228f8676454b33a35b1ee6df187e60fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 12:55:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ed1228f5a7d44ff9c0c7170141a4cd0 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28b88ee08ebd4078a732066c21c78d98 : SUCCESS in 11m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ada65e914c154cbdb7c78ef47fbbe21c : SUCCESS in 10m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/385c837420314e09a3f17f7843e7b650 : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e5a29abbcf44499854abc12a4bb536 : SUCCESS in 10m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a13928a5e9674e1f8e6b39865635031a : SUCCESS in 11m 23s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/c60e27bfea05416ca85a2599dcea4954 : SUCCESS in 31m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/ed38ed0d2a234135a5bc08e618185c91 : SUCCESS in 30m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/915a6bfbe4824664b89e13ced821f05f : SUCCESS in 9m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c0875a5a75d9628e08e2f221a72754b204c6285e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-24 05:51:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f67ea248b41f89bfd4ab5fd078bd63a1bb86045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 05:52:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb919c8ca11169d0e53b9cb5849d25e12bbc8fe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:24:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dc4ce493eab4f7da337194f989f12c9 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f197d87d15e742599bc0d3f73f8c76ab : SUCCESS in 9m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62bd1d4a12c346f385a3a70748131440 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/460a59205080495c9bec717f51c0a56a : SUCCESS in 7m 36s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1e9d2d2cba0e4cf9a8be92cebea04804 : SUCCESS in 31m 11s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/55cce0eb7d2541058a124ea7a4de5903 : SUCCESS in 29m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3abf19280a14dff83482ebe188911bb : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"af81a72e841a3a538093e253027cee9d7a6eeaa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:24:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a9972308781d13ca408c0bafba458e48d707d2fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:26:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b1c1fafa1b4458b9931c8136078eff : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"940d777ae14dac2b77558b5e1ce3f701ac2f8ce0","revisions":{"940d777ae14dac2b77558b5e1ce3f701ac2f8ce0":{"kind":"REWORK","_number":1,"created":"2020-04-23 12:21:18.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/90/722290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/90/722290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/90/722290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/90/722290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/90/722290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/90/722290/1"}}},"commit":{"parents":[{"commit":"545a34c7511b6e6b5f1b59ca99891684303b4480","subject":"Use unittest.mock instead of third party mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/545a34c7511b6e6b5f1b59ca99891684303b4480"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 12:21:16.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 12:21:16.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: If568fce338ef28bcbbb5dddde399cf96fdbab39a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/940d777ae14dac2b77558b5e1ce3f701ac2f8ce0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/940d777ae14dac2b77558b5e1ce3f701ac2f8ce0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
