)]}'
{"id":"openstack%2Fmistral~722291","triplet_id":"openstack%2Fmistral~stable%2Fussuri~I92e491f07c7abbb0ac8b0d3e499fcd114b74f337","project":"openstack/mistral","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I92e491f07c7abbb0ac8b0d3e499fcd114b74f337","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-23 12:21:21.000000000","updated":"2020-04-24 06:30:29.000000000","submitted":"2020-04-24 06:24:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722291-1587709488098-653b66dd","meta_rev_id":"7a342e4a5d94284ed5bd2c8ccfee10a6c77b81ac","_number":722291,"virtual_id_number":722291,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 06:24:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-04-24 05:52:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-23 12:25:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-04-24 05:52:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 12:25:12.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-24 05:52:00.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-04-24 06:24:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd8d1c3431cce752ed2702fdb9a765f1784a75f0","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-23 12:21:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00e7b80c0e639af5fb88842e6b02a1b0c870da19","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-23 12:25:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e3944659d2ae1177353c597d882a293583c95e5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 12:58:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c0807c4feaf472ab6723dc5ed8da9db : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ac6b4e5e4f043a285f9a5cff57b9f95 : SUCCESS in 10m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6e906c6ce9e476db8499bbe6e7d3a27 : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdff9344e8094b1b80d0635b54cd648a : SUCCESS in 10m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59258688a94e41ea91a705978134f16e : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7aaab12cb87a40bbabb61f443a787ee9 : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62999c14c989475aaae539a7ce62a372 : SUCCESS in 11m 00s (non-voting)\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/041897d326d946fd87515b8c65363113 : SUCCESS in 31m 26s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8a23e0e9d58d4575b42df8cef251f8ad : SUCCESS in 28m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dec97840da5481c9d2ec9737923a68c : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3702d825b418ab1194675efd16774548d6e6ef50","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-24 05:52:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e86a02f81095f5f123c8427ce7af87911a658be6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 05:52:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba788748e639da8e2462449f92daf4912ec60075","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:24:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12177287005d4f9db2c4a775150cf942 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0497a3ea2d714f6cbc1e961c4c030735 : SUCCESS in 8m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01d4a83cda8945d0b4e8875e646d1321 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14532386e2464d9e88a96d9f7151b78a : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/561d249e33634814ab2df55ff07322aa : SUCCESS in 4m 39s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/b0485ad9a6124538b1039fbe7601a901 : SUCCESS in 31m 51s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/dff86b91481f4030af6bd029f849165b : SUCCESS in 30m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cec6825338f43c8b4d0930391d19593 : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7e96ef9e33c0dbccb61c4d7fdbffce9efb57bc1f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:24:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9fbbb13358fa06d7947ab333b92a3732e161aa86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:30:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fe5dd53586d4595beb800046d36a6b9 : SUCCESS in 3m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e73148a0582c404a90492f13ae3e3bf8 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c09906cef8fd9338d75d9d91e55969919104d8d","revisions":{"0c09906cef8fd9338d75d9d91e55969919104d8d":{"kind":"REWORK","_number":1,"created":"2020-04-23 12:21:21.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/91/722291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/91/722291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/722291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/722291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/91/722291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/91/722291/1"}}},"commit":{"parents":[{"commit":"940d777ae14dac2b77558b5e1ce3f701ac2f8ce0","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/940d777ae14dac2b77558b5e1ce3f701ac2f8ce0"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 12:21:18.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 12:21:18.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I92e491f07c7abbb0ac8b0d3e499fcd114b74f337\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0c09906cef8fd9338d75d9d91e55969919104d8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/0c09906cef8fd9338d75d9d91e55969919104d8d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
