)]}'
{"id":"openstack%2Fmistral~722293","triplet_id":"openstack%2Fmistral~master~I1882ad126c8e0cd41eaa29b74d13c11602c0a0a3","project":"openstack/mistral","branch":"master","topic":"add-victoria-python-jobtemplates","hashtags":[],"change_id":"I1882ad126c8e0cd41eaa29b74d13c11602c0a0a3","subject":"Add Python3 victoria unit tests","status":"MERGED","created":"2020-04-23 12:21:26.000000000","updated":"2020-04-24 08:52:21.000000000","submitted":"2020-04-24 08:42:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722293-1587717763386-276f548f","meta_rev_id":"81e289d96a4877cd37760fe966eae90325a0bd0d","_number":722293,"virtual_id_number":722293,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 08:42:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-04-24 05:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-23 12:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-04-24 05:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 12:24:39.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-24 05:52:18.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-04-24 08:42:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da0307d2ca0f56fc1b5d0e6d154170fe1c887c91","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-23 12:21:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f22947ab19ae6b9a127a219227885d29f75d685b","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-23 12:24:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1eade05fc5b42a70132ed32b14831dd10ac6d98f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 13:40:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83b6a1abe45044f6b12d6910748ef226 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b179a753cfbb4ebe9b4f6272b80e493b : SUCCESS in 9m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db582579ce6e4c09a054803268f83dbe : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c5226568fb4b46978fe651c2cbb283 : SUCCESS in 9m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4118bd3c91324b69b68e53b66fe3fe6f : SUCCESS in 10m 07s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/9e130ae6ddd842a6aa2b494966359408 : SUCCESS in 50m 46s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c90cabfef5b4bccb6dc78ad7fc76a1f : SUCCESS in 44m 50s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33b9ea8aa2b14aecb9bd3abc9b906619 : SUCCESS in 43m 18s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/917d802cf47d409c9c89f1e3468eaa67 : SUCCESS in 41m 48s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/fcc2059bd6f54e109cb4b4ce22e406ce : SUCCESS in 40m 56s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/3122f45a8c824099854d53df33b6a0e0 : SUCCESS in 29m 48s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/0614e7295d70478d88b9e1d84481fb5a : SUCCESS in 30m 17s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/f099b9ccba7a40088431a8dceaf874b6 : SUCCESS in 1h 09m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/034ba9110c6042d790bf1640cf63b774 : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":1},{"id":"b7b88bdb83d14fa3746168c74281b48a59b2757e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-24 05:52:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f52861f6ea6f51a088dc439941720459498d3d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 05:52:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73482f5e8d0e76954024d645befa2e5b2cf92960","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:40:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be7e6569bb554d94a9a5e3006fc77096 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3d8c9f7ed4c431eadb01392d9b70d50 : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2a89ee8707b4080a6dd8a58eff30b05 : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2077f031085c4b72b9196d1ace9c7772 : SUCCESS in 8m 06s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/7b587a52b4dc4c0a87e1b17a3b4fc332 : SUCCESS in 45m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a42152752054a3998b2caef268d4c1a : SUCCESS in 45m 51s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0039510a06b144e8b077a6738018df8c : SUCCESS in 38m 46s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b786f3971bc0404fa18f9144ef454a5d : SUCCESS in 43m 41s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/938b345b35eb4fbd9d78524a07b5f7e5 : SUCCESS in 32m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/8dfa0246158d4a4db37764893abb6993 : SUCCESS in 29m 12s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/a369420a951f48408961f805ae68d2bd : FAILURE in 44m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93112cbc49d8454c9cd09a74e73d3cb1 : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d5c0356aadaad0ccba04767edc02755a76624a15","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-24 06:42:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"542094ed1c3922beca150648782e83a1e0686788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 07:49:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540bb70e7e4f46eba770a12d4328da19 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c47229f434c348cd8eb403e8d95fb418 : SUCCESS in 11m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5faac5565d454f11a641b48ef48adcbb : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3898b45a64084299917ccdb5c5288e94 : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20d1baaa49644b7a82b04a9db585afe1 : SUCCESS in 10m 49s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/8c1525822380453fa925695cbc0ce17e : SUCCESS in 57m 36s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25aca75802f5443a91fb302bcfa2a6ec : SUCCESS in 40m 56s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f5d2ac567204de6b5919a251b99e3dd : SUCCESS in 55m 54s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/f6bc9c8a597a44ffb4de5c31c5e1b8f5 : SUCCESS in 45m 14s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/8c78e6324e73470c8bde2f1d892baae7 : SUCCESS in 47m 12s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/9c1213c123934da98998defea9f9cd53 : SUCCESS in 30m 56s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b8265557cecf45d3a34df35ddc588f4d : SUCCESS in 30m 25s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/38f365dbf04f411f9e9efabaa2b339db : SUCCESS in 1h 03m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bec7c751c51a463eac643a36f7e9057d : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"07b532f4ad066fb8516c7cca65b541cc9614db56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 07:49:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70374b0a0f6cb1554c9cc19717bc8b7e58491150","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 08:42:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"821a9d289af1da0baa09f73aa41756cf725767a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 08:42:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0750c020841f4ab8a5d11b2604d47bf7 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37a30587fa3341e49a0a88ad13b100ed : SUCCESS in 9m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b1f27e891a8444eb5919541019bc97b : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d90fc8b0f064729ae5a67b6d8820af0 : SUCCESS in 8m 30s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/e2a95a8ecda8483ead3f8a6ccef91c9c : SUCCESS in 41m 28s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa8c9d8de7c643a09f2a27b42759609a : SUCCESS in 40m 53s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/673640aa47b94680a1177e9320becbda : SUCCESS in 39m 10s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/ea43bc385b7a421fa7278695dd9f9334 : SUCCESS in 42m 40s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/81ae72fb713a43d9937c3de7f8a93e2c : SUCCESS in 30m 58s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/7936921f4e224bc89fb1bb3c2fbb9d0a : SUCCESS in 28m 03s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0d129cd530674d49b8f231b715bba6c1 : SUCCESS in 46m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7775664c4eb246fba08d2855b63dbcb6 : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":1},{"id":"49259d7ae25f7cfab702db41e46383b1c2bb5857","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 08:52:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dfdfc7846f1436d98b01d1f443bfeaf : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ddd3c637e2d0b1b88af27580d23fdf4dc2eb219","revisions":{"4ddd3c637e2d0b1b88af27580d23fdf4dc2eb219":{"kind":"REWORK","_number":1,"created":"2020-04-23 12:21:26.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/93/722293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/93/722293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/722293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/722293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/93/722293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/93/722293/1"}}},"commit":{"parents":[{"commit":"b56576eef61ff3ecb7db893d4b8679cebb212ee4","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/b56576eef61ff3ecb7db893d4b8679cebb212ee4"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 12:21:24.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-23 12:21:24.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I1882ad126c8e0cd41eaa29b74d13c11602c0a0a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4ddd3c637e2d0b1b88af27580d23fdf4dc2eb219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/4ddd3c637e2d0b1b88af27580d23fdf4dc2eb219"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
