)]}'
{"id":"openstack%2Fmistral~722702","triplet_id":"openstack%2Fmistral~master~Id140764a69f443d3f7e9f69ef71601f1a694fddb","project":"openstack/mistral","branch":"master","topic":"add-py38-meta","hashtags":[],"change_id":"Id140764a69f443d3f7e9f69ef71601f1a694fddb","subject":"Add py38 package metadata","status":"MERGED","created":"2020-04-24 13:25:01.000000000","updated":"2020-04-24 15:49:59.000000000","submitted":"2020-04-24 15:47:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722702-1587743264637-c713193e","meta_rev_id":"6c062c22b617dd237f2f6ca819af1d973a7932f7","_number":722702,"virtual_id_number":722702,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 15:47:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-04-24 14:15:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-24 13:46:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-04-24 14:15:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 13:46:07.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-24 14:15:11.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-04-24 15:47:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d862e42b26a9cef93ef2668015e2f2ea413b400","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 13:25:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74b847492913e75730a5c1edbf260a355dde9dcd","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-24 13:46:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a37c0dae0ab7a3cb46b52cbca5620b00401d11fc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-24 14:15:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d5711dafec7682d31d32f983118f737e6dfb5520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 14:44:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/65cd8d164c564a5c8b684784c07e6e94 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aa0fee882634892948afaf5a3070e1d : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/761f8aab6aa2425ca4a832371aa5fc33 : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a6fd1401c6249b4b0a42a68e74bf360 : SUCCESS in 11m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbd524db1a4c4160adffc7213533b36a : SUCCESS in 9m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac6ff4be89e7454783860c56a8a6def3 : SUCCESS in 13m 15s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af40a7801dc648b6ba4dea8356b1c9d8 : SUCCESS in 11m 23s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/2f1de3e4661a481cb6de67cc5fabcad8 : SUCCESS in 44m 27s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c34f6c20d4c149a094ba65c37903cba5 : SUCCESS in 52m 02s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad5862091ca940ddb770d86953901217 : SUCCESS in 48m 49s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/aad266a4f10549d3a3caa28c4d72bee3 : SUCCESS in 47m 56s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/0ed72dcd6a0a4536bfdcb1582727edcd : SUCCESS in 46m 03s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/6fbfabceca17491891ee797cf17c74c6 : SUCCESS in 32m 25s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/6c55723adddb454b985a4d26a7578977 : SUCCESS in 30m 07s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/9c544c6282114b60b87ea1416b60c468 : SUCCESS in 1h 09m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49fe4c5a4cef441e906efb8294171a15 : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b8348fd8450ef1e4e005bdefe768cafaa68f0fb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 14:44:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8d472ca8a2249d8f8e8bbfc07a7e13ef09e7698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 15:47:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bcb9208a16c74548b0127b16242bded4 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffae19be1f9440c1a0fce7787dff6e32 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d95af2a6038349b084fc827ad1c3cc16 : SUCCESS in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82dc7e4cb0c64d5d94febe0db2f85735 : SUCCESS in 12m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca9781e45dde4ce79cd36f5882088224 : SUCCESS in 9m 13s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/4f72c9e1e19741448e08ffbd81536331 : SUCCESS in 42m 42s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8edb03a482f14c69abaa95aae006eb6d : SUCCESS in 45m 48s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc32468c148347bf8cc41aca85ab8314 : SUCCESS in 41m 37s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/b4566fc60ad14337a6234723fcb5156f : SUCCESS in 1h 00m 36s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/30c79b472d7340b9bc755faa055a6543 : SUCCESS in 33m 05s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/eda0949ed04f4852970bf45be4e2ce90 : SUCCESS in 28m 10s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/c6f27c356bb14106a1ddb7c7f732fcc1 : SUCCESS in 41m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8de78e94f7d4290a028edfc0f93f256 : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":1},{"id":"77ad5f4c97fa5cfb8bcaa55fcb66e50b944d8355","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 15:47:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6a8ba14a43e320973da1b26e04e3477699bd05df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 15:49:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32128b7972da4726991cbdfddd237242 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3eef103b0ca613d190388706c18f9b4568939258","revisions":{"3eef103b0ca613d190388706c18f9b4568939258":{"kind":"REWORK","_number":1,"created":"2020-04-24 13:25:01.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/02/722702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/02/722702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/722702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/722702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/02/722702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/02/722702/1"}}},"commit":{"parents":[{"commit":"d4834a0c11b4149aa313a48bec7b7032ef67834c","subject":"Merge \"Add Python3 victoria unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d4834a0c11b4149aa313a48bec7b7032ef67834c"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 13:22:36.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 13:22:36.000000000","tz":-300},"subject":"Add py38 package metadata","message":"Add py38 package metadata\n\nNow that we are running the Victoria tests that include a\nvoting py38, we can now add the Python 3.8 metadata to the\npackage information to reflect that support.\n\nChange-Id: Id140764a69f443d3f7e9f69ef71601f1a694fddb\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3eef103b0ca613d190388706c18f9b4568939258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/3eef103b0ca613d190388706c18f9b4568939258"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
