)]}'
{"id":"openstack%2Fmistral~722834","triplet_id":"openstack%2Fmistral~master~I444260c25615c23644c5c60e0934f6cbbde07b15","project":"openstack/mistral","branch":"master","topic":"py38-tox","hashtags":[],"change_id":"I444260c25615c23644c5c60e0934f6cbbde07b15","subject":"Bump default tox env from py37 to py38","status":"MERGED","created":"2020-04-24 15:28:11.000000000","updated":"2020-04-27 05:38:23.000000000","submitted":"2020-04-27 05:37:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722834-1587965820668-354c625b","meta_rev_id":"c4028c90677c5023547f3b57a2b992dff1222c15","_number":722834,"virtual_id_number":722834,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 05:37:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":2,"date":"2020-04-27 04:37:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-24 15:31:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"all":[{"value":1,"date":"2020-04-27 04:37:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 15:31:16.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"},{"updated":"2020-04-27 04:37:37.000000000","updated_by":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"reviewer":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"state":"REVIEWER"},{"updated":"2020-04-27 05:37:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25643a5387293134d734a91edaad4dc606a86f34","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-24 15:28:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8393d9ff82aaa20af5e0ba5585db068942eeee26","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-04-24 15:31:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f65d6be6c06866651bb9112d9edfa95d0a1d0da6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 16:52:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3e9aa2e2ca64506860bc654881e034e : SUCCESS in 8m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea93e2bbaad04814bc4e086b01ff084b : SUCCESS in 10m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c03a72ba0a14e51989d08b1919dbbaa : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27e5426638554548b854a194eea10a28 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae56d40764ec44ea8b87774c3ddea9bd : SUCCESS in 5m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4de4b55e107f40eab84ed459a0aa46f5 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0fb457d4e4f4d349bc614dbf3320b32 : SUCCESS in 8m 59s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/76d810b3d0294fc4a375fc39957a45ce : SUCCESS in 44m 09s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbf577135ba04c64a64a6094dcce6d61 : SUCCESS in 45m 56s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe2a3cece22e43e7bc40b5cc0369ee8a : SUCCESS in 46m 58s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/3d0b01a3964443fa9e05075b4a4fea2b : SUCCESS in 45m 28s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/da2d0d81407f4b4bbde1dd8beceb54f8 : SUCCESS in 45m 23s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/1e48100a3f144d4596bdb6756e384929 : SUCCESS in 31m 10s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/b2954faa12a9468297abc8ae2f0bb685 : SUCCESS in 30m 35s\n- mistral-rally-task https://zuul.opendev.org/t/openstack/build/cb86e639ea614514933c93de618237c1 : SUCCESS in 1h 13m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/549394f8acdc4319a87866d6f5b739fe : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4ec3d7ff2fc17fe31f02b484a73ce04a192d97cc","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2020-04-27 04:37:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d75ac36a289eb681086c434b8c7a6c678c4050e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 04:37:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e323b891c916947f8e689d3acb51ce306481820","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 05:37:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5123e286d8844392ac93cf658088f9ce : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1210cf56615b450c882c15d72d6eea91 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5f8ab3f733d44bba0e474f5d2f88dc2 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f3a6ede92d42c99a0f8d9a5a52d474 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6886af6af4044022afbdaa289499da77 : SUCCESS in 4m 20s\n- mistral-devstack https://zuul.opendev.org/t/openstack/build/6c1db7bd65604757b567d516491de98b : SUCCESS in 51m 59s\n- mistral-devstack-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71c6a9beb5ab4aca809eda72aafb6341 : SUCCESS in 42m 42s\n- mistral-devstack-non-apache-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3749da4acef44a9da74736447974b60d : SUCCESS in 54m 07s\n- mistral-devstack-non-apache https://zuul.opendev.org/t/openstack/build/7b24ad8d80b44b33adf928a42654ccb1 : SUCCESS in 57m 38s\n- mistral-tox-unit-mysql https://zuul.opendev.org/t/openstack/build/02e8194ef31e4e43ada2de0577c907ad : SUCCESS in 30m 24s\n- mistral-tox-unit-postgresql https://zuul.opendev.org/t/openstack/build/67ead9abcb6b4e289475f43973fb1f84 : SUCCESS in 28m 32s\n- mistral-devstack-kombu https://zuul.opendev.org/t/openstack/build/9af1ca6ab3594e9eb22a34aef4e4d28c : SUCCESS in 51m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/974c68537a1d475b82a90562a7f50fd2 : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":1},{"id":"eb2dd35bae42d1473436cf35d00f4563db254edf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 05:37:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"983186dadeac9d47d173a2685ae9161e0cdb7899","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 05:38:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2398d5ba84424eca9b068d064dc10438 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b11a1b638bf4c8da951abdaf03ac146 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99cf8cbaebc674c527cd52b0b80ae0a3475cfdff","revisions":{"99cf8cbaebc674c527cd52b0b80ae0a3475cfdff":{"kind":"REWORK","_number":1,"created":"2020-04-24 15:28:11.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/34/722834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/mistral","ref":"refs/changes/34/722834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/722834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/722834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/mistral refs/changes/34/722834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/mistral refs/changes/34/722834/1"}}},"commit":{"parents":[{"commit":"d4834a0c11b4149aa313a48bec7b7032ef67834c","subject":"Merge \"Add Python3 victoria unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/d4834a0c11b4149aa313a48bec7b7032ef67834c"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:25:43.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-04-24 15:25:43.000000000","tz":-300},"subject":"Bump default tox env from py37 to py38","message":"Bump default tox env from py37 to py38\n\nPython 3.8 is now our highest level supported python runtime.\nThis updates the default tox target environments to swap out\npy37 for py38 to make sure local development testing is\ncovering this version.\n\nThis does not impact zuul jobs in any way, nor prevent local\ntests against py37. It just changes the default if none is\nexplicitly provided.\n\nChange-Id: I444260c25615c23644c5c60e0934f6cbbde07b15\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99cf8cbaebc674c527cd52b0b80ae0a3475cfdff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/mistral/commit/99cf8cbaebc674c527cd52b0b80ae0a3475cfdff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
